)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~660307","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ia3c00c4403e6a354e66e02b16de37247c6bc663b","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"dnm","hashtags":[],"change_id":"Ia3c00c4403e6a354e66e02b16de37247c6bc663b","subject":"[DNM] remove python-httplib2 during venv build requires","status":"ABANDONED","created":"2019-05-21 09:04:30.000000000","updated":"2019-05-23 06:23:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa5ef677916a72b04d26ba4517b254df9d2e55de","_number":660307,"virtual_id_number":660307,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-21 10:39:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 10:39:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"912d29b7355c307934fd500ff738e58d82437a25","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-21 09:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"015170b1c9d45aae6c809082e2186c947b7090ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:39:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/07/660307/1/check/openstack-ansible-linters/609a120/ : SUCCESS in 6m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/07/660307/1/check/openstack-ansible-deploy-aio_metal-debian-stable/f4861a0/ : SUCCESS in 57m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/07/660307/1/check/openstack-ansible-deploy-aio_metal-centos-7/a03e43d/ : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/07/660307/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/cd7ebf6/ : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/54f4976/ : FAILURE in 15m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/07/660307/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0929fca/ : SUCCESS in 54m 58s\n- openstack-tox-docs http://logs.openstack.org/07/660307/1/check/openstack-tox-docs/219a540/html/ : SUCCESS in 2m 49s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovs-ubuntu-bionic/452e766/ : FAILURE in 13m 43s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovn-ubuntu-bionic/b10be58/ : FAILURE in 12m 14s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovn-centos-7/d7e2f4b/ : FAILURE in 29m 34s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-calico-ubuntu-bionic/290c9f8/ : FAILURE in 12m 46s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-opendaylight-ubuntu-bionic/7f91aa7/ : FAILURE in 14m 14s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/d19e5bf/ : FAILURE in 14m 05s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/d1ffc7e/ : FAILURE in 13m 36s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/da0465f/ : FAILURE in 13m 57s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/f8ff2a3/ : FAILURE in 19m 11s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/07/660307/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/479400a/ : FAILURE in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85b2d5ba81611f30ee1019dfd8d76965e2518d68","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-05-23 06:23:29.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7611e858e12d3b7ecf693949c4312026a50eb00b","revisions":{"7611e858e12d3b7ecf693949c4312026a50eb00b":{"kind":"REWORK","_number":1,"created":"2019-05-21 09:04:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/07/660307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/07/660307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/07/660307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/07/660307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/07/660307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/07/660307/1"}}},"commit":{"parents":[{"commit":"9f619e3f059d2bf8db930ee23b8847456775708a","subject":"Remove LBaaS v2 plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9f619e3f059d2bf8db930ee23b8847456775708a"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-21 09:03:37.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-05-21 09:04:27.000000000","tz":330},"subject":"[DNM] remove python-httplib2 during venv build requires","message":"[DNM] remove python-httplib2 during venv build requires\n\nChange-Id: Ia3c00c4403e6a354e66e02b16de37247c6bc663b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7611e858e12d3b7ecf693949c4312026a50eb00b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7611e858e12d3b7ecf693949c4312026a50eb00b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
