)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~660697","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Iae15e90e0e572db806c712a60b8ea3794b9e9a71","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"python-httplib2","hashtags":[],"change_id":"Iae15e90e0e572db806c712a60b8ea3794b9e9a71","subject":"Stop installing python-httplib2","status":"MERGED","created":"2019-05-22 12:19:47.000000000","updated":"2019-05-23 05:08:20.000000000","submitted":"2019-05-23 05:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660697-1558588100491-3e17a491","meta_rev_id":"370a167e561560ea338e6482d7a36d8d3a750c16","_number":660697,"virtual_id_number":660697,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-23 02:31:22.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-23 05:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-05-22 12:38:21.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"date":"2019-05-22 12:21:26.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 17:52:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 12:21:26.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-05-22 12:38:21.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-22 17:52:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-23 05:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c2cd6d86f9fc50fd7e268506a28711546d969b0","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 12:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc215cb834c50d844e7fdc9bb05f8aef39c9c797","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 12:38:21.000000000","message":"Patch Set 1: Code-Review+2\n\nGood explanation.","accounts_in_message":[],"_revision_number":1},{"id":"aea3f4804228a6cc728242e997179eb1444f72d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 14:00:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/660697/1/check/openstack-ansible-linters/cda4b2c/ : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0ac9ad8/ : SUCCESS in 48m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-centos-7/3c1d88b/ : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/0b3a0c6/ : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6e1ecb4/ : SUCCESS in 56m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bd282f0/ : SUCCESS in 54m 21s\n- openstack-tox-docs http://logs.openstack.org/97/660697/1/check/openstack-tox-docs/68b6802/html/ : SUCCESS in 2m 43s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-ubuntu-bionic/a6c3624/ : SUCCESS in 36m 15s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovn-ubuntu-bionic/033a38c/ : SUCCESS in 37m 35s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovn-centos-7/5d2229e/ : FAILURE in 30m 45s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-calico-ubuntu-bionic/128d887/ : FAILURE in 38m 08s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-opendaylight-ubuntu-bionic/3963c27/ : SUCCESS in 1h 00m 28s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/87a6ec6/ : SUCCESS in 55m 53s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/d31da1d/ : SUCCESS in 55m 38s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/2da1810/ : SUCCESS in 37m 58s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/3d3d8e9/ : SUCCESS in 1h 00m 00s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/17fc468/ : SUCCESS in 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f538a22f051b8fc495546dca4e83fb9f5768780d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 16:05:20.000000000","message":"Topic set to python-httplib2","accounts_in_message":[],"_revision_number":1},{"id":"3b5214414cc828c3270113c8aad1de64d9acd7ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-22 17:52:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38bb07d7eb4866cac74e6c041a8e79025a118df7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 17:53:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f55d9b7551c43cd47e0e3ed61ff7a4f41581e2d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:11:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/97/660697/1/gate/openstack-ansible-linters/f913b80/ : SUCCESS in 6m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable finger://ze08.openstack.org/10d5dab4939344a795472668d4e8bfb5 : RETRY_LIMIT in 39m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 finger://ze02.openstack.org/2a0fbe41d6b048ef9fcba3fde3353700 : RETRY_LIMIT in 37m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-150 finger://ze03.openstack.org/50b8e4bc301d4adea081a02ae2c22778 : RETRY_LIMIT in 31m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic finger://ze01.openstack.org/8efdcb2e696b47789d04b02d2443ba43 : RETRY_LIMIT in 42m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 finger://ze04.openstack.org/d8060277d8774b36a6e9bf42c96cad46 : RETRY_LIMIT in 28m 05s\n- openstack-tox-docs http://logs.openstack.org/97/660697/1/gate/openstack-tox-docs/fa6fff9/html/ : SUCCESS in 3m 09s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-ovs-ubuntu-bionic/c306d61/ : SUCCESS in 37m 30s\n- openstack-ansible-ovn-ubuntu-bionic finger://ze06.openstack.org/bf76b4b7992f47ed964e7499aebb8d1e : RETRY_LIMIT in 10m 57s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/3cfb084/ : RETRY_LIMIT in 9m 18s\n- openstack-ansible-ovs-nsh-ubuntu-bionic finger://ze11.openstack.org/1f7ac66d89cb49c5bd354431914c872b : RETRY_LIMIT in 37m 46s\n- openstack-ansible-odl-sfc-ubuntu-bionic finger://ze02.openstack.org/6ff0045a82584ebba6200b616955873e : RETRY_LIMIT in 38m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1d85db36d85f878cf40aa6b8a9749420f1f21ab3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-23 02:31:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6abfaf6288fd5e441cab64d48400d6c2fa976128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 03:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/660697/1/check/openstack-ansible-linters/470d8f7/ : SUCCESS in 4m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-debian-stable/e7edd24/ : SUCCESS in 57m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-centos-7/75e029d/ : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/1c2f5bf/ : SUCCESS in 1h 09m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2e586a9/ : SUCCESS in 55m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/98a9c6d/ : SUCCESS in 53m 34s\n- openstack-tox-docs http://logs.openstack.org/97/660697/1/check/openstack-tox-docs/5dcfe7d/html/ : SUCCESS in 3m 17s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-ubuntu-bionic/eedcb02/ : SUCCESS in 42m 27s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovn-ubuntu-bionic/3af630d/ : SUCCESS in 41m 06s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovn-centos-7/d542fc0/ : FAILURE in 25m 17s (non-voting)\n- openstack-ansible-calico-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-calico-ubuntu-bionic/7adb11c/ : FAILURE in 37m 53s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-opendaylight-ubuntu-bionic/468d048/ : SUCCESS in 50m 23s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/5c97ba1/ : SUCCESS in 45m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/6fef5f2/ : SUCCESS in 48m 24s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/daf75b7/ : SUCCESS in 37m 24s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/3ad57a5/ : SUCCESS in 48m 49s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/97/660697/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/9afb343/ : SUCCESS in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4c26732808b0ea7007527227690e31ed940f0e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 03:49:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ddb256f97bd70cf6a7da287c6e0fd99cd3e07d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 05:08:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"370a167e561560ea338e6482d7a36d8d3a750c16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 05:08:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/660697/1/gate/openstack-ansible-linters/d5a1155/ : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/97/660697/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/c2ef738/ : SUCCESS in 1h 02m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/97/660697/1/gate/openstack-ansible-deploy-aio_metal-centos-7/8ed139d/ : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/97/660697/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/087131d/ : SUCCESS in 1h 07m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/74c1d16/ : SUCCESS in 1h 00m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/97/660697/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/6bb7e8a/ : SUCCESS in 51m 10s\n- openstack-tox-docs http://logs.openstack.org/97/660697/1/gate/openstack-tox-docs/8a67d85/html/ : SUCCESS in 2m 46s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-ovs-ubuntu-bionic/c7a042f/ : SUCCESS in 39m 05s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-ovn-ubuntu-bionic/fcf3fe3/ : SUCCESS in 36m 42s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/c535cd0/ : SUCCESS in 49m 06s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/a22e489/ : SUCCESS in 53m 07s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/97/660697/1/gate/openstack-ansible-odl-sfc-ubuntu-bionic/5f08a89/ : SUCCESS in 48m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe1144a7fdd046dbc5656a2b79f4c30919a7bafb","revisions":{"fe1144a7fdd046dbc5656a2b79f4c30919a7bafb":{"kind":"REWORK","_number":1,"created":"2019-05-22 12:19:47.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/97/660697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/97/660697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/97/660697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/97/660697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/97/660697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/97/660697/1"}}},"commit":{"parents":[{"commit":"423a63c53c1671028c9dac7ad38ba3ac534272d6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/423a63c53c1671028c9dac7ad38ba3ac534272d6"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-22 12:16:31.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-22 12:16:31.000000000","tz":-300},"subject":"Stop installing python-httplib2","message":"Stop installing python-httplib2\n\nIn https://review.opendev.org/#/c/575701/, we installed python-httplib2\nbecause we needed it for python-keystoneclient to work. However,\nwe no longer install python-keystoneclient as of\nhttps://review.opendev.org/#/c/627791/. Except, we never stopped\ninstalling python-httplib2 even though it is no longer needed.\n\nThis currently breaks CentOS because python-httplib2 is not present\nin the base or updates repos that we enable, only in extras.\n\nChange-Id: Iae15e90e0e572db806c712a60b8ea3794b9e9a71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/fe1144a7fdd046dbc5656a2b79f4c30919a7bafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/fe1144a7fdd046dbc5656a2b79f4c30919a7bafb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
