)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~664334","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Frocky~I7741c7b5601b2d12918568f322a7829f433c63f2","project":"openstack/openstack-ansible-os_neutron","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7741c7b5601b2d12918568f322a7829f433c63f2","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:44:51.000000000","updated":"2019-06-10 18:58:24.000000000","submitted":"2019-06-10 18:58:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664334-1560193104338-48a02ede","meta_rev_id":"7f1e0b137fe0a86d6ac3acdfab5425ae924b2e1a","_number":664334,"virtual_id_number":664334,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-06-10 18:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-06-10 17:34:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2019-06-10 17:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:35:17.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-06-10 18:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c41b46d645b4a2edfc271d262624c6142f973664","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ae9c4fb7b1fea20425b68d32d1e5980ba7385df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/664334/1/check/openstack-ansible-linters/9ddd46d/ : SUCCESS in 4m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-centos-7/e335787/ : SUCCESS in 53m 29s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-opensuse-423/f077d9f/ : FAILURE in 11m 00s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-opensuse-150/9554d0a/ : FAILURE in 9m 44s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-ubuntu-bionic/6e5ef9e/ : SUCCESS in 44m 21s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-ubuntu-xenial/100d3ff/ : SUCCESS in 47m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/34/664334/1/check/build-openstack-sphinx-docs/db29421/html/ : SUCCESS in 3m 24s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-ovs-ubuntu-xenial/920fdd6/ : SUCCESS in 35m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/0178b59/ : SUCCESS in 57m 59s\n- openstack-ansible-dragonflow-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-dragonflow-ubuntu-xenial/cfe6c86/ : FAILURE in 27m 57s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-opendaylight-ubuntu-xenial/3e80965/ : SUCCESS in 49m 05s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/1786ce0/ : SUCCESS in 1h 07m 23s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/14b799c/ : SUCCESS in 1h 20m 27s\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/34/664334/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/86f3501/ : SUCCESS in 55m 42s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/935cec6/ : SUCCESS in 43m 14s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-distro_install-centos-7/3b22a94/ : SUCCESS in 40m 11s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/34/664334/1/check/openstack-ansible-functional-distro_install-opensuse-423/706f6e4/ : FAILURE in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe899023de64d25c7019a525c28f8694f3c1ea25","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:34:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7356233fe027cf8d9b6548fbb938cfce8421112","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-10 17:35:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f15c172478ba6ef526fc01160fdda62a62f2d90e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf332916c86dbff8890a8a7fcac93bd5cdeba8f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:58:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/664334/1/gate/openstack-ansible-linters/2d165ae/ : SUCCESS in 7m 12s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-centos-7/1b5702e/ : SUCCESS in 48m 04s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-opensuse-423/82f3eda/ : FAILURE in 9m 23s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-ubuntu-bionic/11c5b46/ : SUCCESS in 36m 47s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-ubuntu-xenial/7846866/ : SUCCESS in 38m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/34/664334/1/gate/build-openstack-sphinx-docs/30f6f6a/html/ : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-deploy-aio_metal_calico-ubuntu-xenial/a3473cc/ : SUCCESS in 50m 58s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-ovs-ubuntu-xenial/862d643/ : SUCCESS in 40m 33s\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-opendaylight-ubuntu-xenial/5ace496/ : SUCCESS in 48m 32s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-ovs-nsh-ubuntu-xenial/f11fab1/ : SUCCESS in 59m 53s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/34/664334/1/gate/openstack-ansible-odl-sfc-ubuntu-xenial/997fef7/ : SUCCESS in 1h 02m 01s\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/a641623/ : SUCCESS in 40m 46s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/34/664334/1/gate/openstack-ansible-functional-distro_install-opensuse-423/d8e8ba4/ : FAILURE in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d139b4acd9eca4f6af9b006c3317e922602bd17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:58:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27ea89b839faa60376a0973906ae8b77edec8af9","revisions":{"27ea89b839faa60376a0973906ae8b77edec8af9":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/664334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/664334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/664334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/664334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/664334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/664334/1"}}},"commit":{"parents":[{"commit":"f135d659b6661546d41e90420b6116e265c53b7c","subject":"Make venv download configurable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f135d659b6661546d41e90420b6116e265c53b7c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7741c7b5601b2d12918568f322a7829f433c63f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/27ea89b839faa60376a0973906ae8b77edec8af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/27ea89b839faa60376a0973906ae8b77edec8af9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
