)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~665191","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I1154d9a027daaf9eaaae3670f6cf6823f3fe03e4","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"systemd-journald","hashtags":[],"change_id":"I1154d9a027daaf9eaaae3670f6cf6823f3fe03e4","subject":"Log to systemd-journald","status":"MERGED","created":"2019-06-13 17:52:02.000000000","updated":"2019-06-18 19:10:23.000000000","submitted":"2019-06-18 19:10:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665191-1560885023225-ad3aa802","meta_rev_id":"7814819af7ae13c35dcf10f1bf91f0df5405ca83","_number":665191,"virtual_id_number":665191,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-17 12:24:07.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-18 19:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-06-18 17:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-18 17:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-06-18 17:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 17:08:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-18 17:15:01.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-18 19:10:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09c92b01da65f5b6e773abe565eedd7c6245f98a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-13 17:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"193ffd678a95fc265cec0dbc8e5b3b9875cbf1e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 19:17:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/91/665191/1/check/openstack-ansible-linters/a68c205/ : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-debian-stable/9f236a9/ : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-centos-7/bb2fc0f/ : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/ce56da7/ : SUCCESS in 1h 06m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/be52c3c/ : SUCCESS in 1h 00m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/93eaf30/ : SUCCESS in 56m 13s\n- openstack-tox-docs http://logs.openstack.org/91/665191/1/check/openstack-tox-docs/0345390/html/ : SUCCESS in 3m 04s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-ubuntu-bionic/babe4fa/ : FAILURE in 36m 29s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-ubuntu-bionic/3cdf82d/ : FAILURE in 7m 39s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-centos-7/c619ce7/ : FAILURE in 30m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/b8e3c1d/ : FAILURE in 16m 34s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-ubuntu-bionic/ded277a/ : FAILURE in 8m 11s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/1c4ed28/ : FAILURE in 8m 39s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/9785219/ : FAILURE in 1h 02m 48s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/2804d2a/ : FAILURE in 42m 06s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d78f413/ : FAILURE in 56m 34s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/94ce1ae/ : FAILURE in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61d39633c82b39105af98174c02d6d912cd51175","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-13 19:38:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"233cb751cb008561c01204db5d22d15e2c54d790","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 21:47:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/91/665191/1/check/openstack-ansible-linters/123e41d/ : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0639ec8/ : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-centos-7/1e2fe6c/ : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/03c2548/ : SUCCESS in 1h 27m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8652cb0/ : SUCCESS in 1h 03m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/50ed2a2/ : SUCCESS in 1h 03m 17s\n- openstack-tox-docs http://logs.openstack.org/91/665191/1/check/openstack-tox-docs/19b8fa3/html/ : SUCCESS in 2m 42s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-ubuntu-bionic/2810e9b/ : FAILURE in 40m 54s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-ubuntu-bionic/ef20289/ : FAILURE in 39m 26s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-centos-7/eff8bab/ : FAILURE in 29m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/ddadd49/ : SUCCESS in 1h 04m 06s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-ubuntu-bionic/fbb62f0/ : FAILURE in 54m 40s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/24e2fb4/ : FAILURE in 52m 08s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/8034273/ : FAILURE in 7m 44s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/2d3733b/ : FAILURE in 7m 48s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d17ec69/ : FAILURE in 7m 52s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/e915dbe/ : FAILURE in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b6b2327ddf7dfc8cbf49e44024ee4ae80a83373","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-14 13:55:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5adb110bcfeadfdff3ebc459cda7c7711136754f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 17:40:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/91/665191/1/check/openstack-ansible-linters/f4edba0/ : TIMED_OUT in 16m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-debian-stable/92a58a8/ : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-centos-7/6ead7d5/ : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/06fae16/ : SUCCESS in 59m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/872b278/ : SUCCESS in 57m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0e589c3/ : SUCCESS in 55m 33s\n- openstack-tox-docs http://logs.openstack.org/91/665191/1/check/openstack-tox-docs/4b6e170/html/ : SUCCESS in 2m 46s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-ubuntu-bionic/c7506de/ : TIMED_OUT in 1h 31m 14s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-ubuntu-bionic/594360d/ : TIMED_OUT in 1h 30m 57s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-centos-7/5dd0317/ : FAILURE in 43m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/fbdc727/ : SUCCESS in 1h 08m 33s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-ubuntu-bionic/a18f9a0/ : TIMED_OUT in 1h 30m 51s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/9506134/ : TIMED_OUT in 1h 31m 08s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/9ab2db3/ : TIMED_OUT in 1h 31m 14s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/309962a/ : TIMED_OUT in 1h 31m 04s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/e0f4154/ : TIMED_OUT in 1h 31m 07s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/2ee1c27/ : TIMED_OUT in 1h 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0799fb6f8f14fe5be73dce8417b7721fa9e855ba","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-17 12:24:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"63324f21d6526e9f6afb9bcaa609460b32910512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 13:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/665191/1/check/openstack-ansible-linters/a253e38/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-debian-stable/84550e8/ : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-centos-7/dfbbd07/ : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/ef17bfd/ : SUCCESS in 1h 09m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/70d263d/ : SUCCESS in 57m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/43b22a2/ : SUCCESS in 54m 53s\n- openstack-tox-docs http://logs.openstack.org/91/665191/1/check/openstack-tox-docs/d6b8cdc/html/ : SUCCESS in 2m 43s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-ubuntu-bionic/a0e4042/ : SUCCESS in 43m 22s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-ubuntu-bionic/a45dd56/ : SUCCESS in 38m 13s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovn-centos-7/3ff5a67/ : FAILURE in 31m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/e94140d/ : SUCCESS in 58m 05s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-ubuntu-bionic/fa37411/ : SUCCESS in 43m 27s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-nsh-ubuntu-bionic/aba4ac8/ : SUCCESS in 41m 58s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-odl-sfc-ubuntu-bionic/e42ea87/ : SUCCESS in 45m 44s\n- openstack-ansible-ovs-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-sfc-ubuntu-bionic/8d44a6a/ : SUCCESS in 39m 02s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic/d7748f2/ : SUCCESS in 55m 14s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/91/665191/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/f3a315d/ : SUCCESS in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50a7bc5970a6811fbcd27e531a31c7d019a3b7bc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-18 17:08:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"776cec3f88e0a705b275f99027bbbdcca95c1241","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-18 17:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1083c749ec8c3d6056f0bd4bff7cd120b5a6ba7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 17:15:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df93d14eb4d2ff5ecb7f4debadb373dd52796c9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:10:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/665191/1/gate/openstack-ansible-linters/62e7924/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/6e94dfb/ : SUCCESS in 1h 54m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_metal-centos-7/a5c0d2e/ : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/5334717/ : SUCCESS in 1h 09m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/da07c89/ : SUCCESS in 57m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5313094/ : SUCCESS in 57m 28s\n- openstack-tox-docs http://logs.openstack.org/91/665191/1/gate/openstack-tox-docs/51ea123/html/ : SUCCESS in 2m 49s\n- openstack-ansible-ovs-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-ovs-ubuntu-bionic/d9fd180/ : SUCCESS in 39m 50s\n- openstack-ansible-ovn-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-ovn-ubuntu-bionic/c20f83f/ : SUCCESS in 37m 51s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/28edf85/ : SUCCESS in 59m 42s\n- openstack-ansible-opendaylight-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-opendaylight-ubuntu-bionic/450b454/ : SUCCESS in 40m 52s\n- openstack-ansible-ovs-nsh-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-ovs-nsh-ubuntu-bionic/8cce6c2/ : SUCCESS in 40m 41s\n- openstack-ansible-odl-sfc-ubuntu-bionic http://logs.openstack.org/91/665191/1/gate/openstack-ansible-odl-sfc-ubuntu-bionic/580c99e/ : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8cb89e4000e5292991d4fffea439e62792eb9514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:10:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7425016564dd8baaf4769b609ac338508ac22752","revisions":{"7425016564dd8baaf4769b609ac338508ac22752":{"kind":"REWORK","_number":1,"created":"2019-06-13 17:52:02.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/91/665191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/91/665191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/91/665191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/91/665191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/91/665191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/91/665191/1"}}},"commit":{"parents":[{"commit":"61dc3828b42cab1b0b8ee49d994da107b269bd3d","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61dc3828b42cab1b0b8ee49d994da107b269bd3d"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-06-13 17:44:59.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-06-13 17:44:59.000000000","tz":-240},"subject":"Log to systemd-journald","message":"Log to systemd-journald\n\nThis patch add the neutron logs to systemd-journald.\n\nChange-Id: I1154d9a027daaf9eaaae3670f6cf6823f3fe03e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7425016564dd8baaf4769b609ac338508ac22752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7425016564dd8baaf4769b609ac338508ac22752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
