)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~666356","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I5db81be8fa6026ff2ca689a80267fd6738d0fdf1","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I5db81be8fa6026ff2ca689a80267fd6738d0fdf1","subject":"Sync Sphinx requirement","status":"ABANDONED","created":"2019-06-19 16:00:34.000000000","updated":"2019-07-18 13:30:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14866bea3f39fd8e89eb67c5cc8bcf4bd0ca5c5a","_number":666356,"virtual_id_number":666356,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-20 14:19:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":1,"date":"2019-06-27 21:35:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 14:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 21:35:27.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"e492f853398d6625203b61b4f0f94fab236112fe","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-19 16:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8cdcf72daeef95841ef119c3a117b577b2bfec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:17:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/56/666356/1/check/requirements-check/41fc894/ : SUCCESS in 2m 31s\n- openstack-ansible-linters http://logs.openstack.org/56/666356/1/check/openstack-ansible-linters/70fa461/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/56/666356/1/check/openstack-tox-docs/fc87a51/html/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/56/666356/1/check/build-openstack-releasenotes/1315b42/html/ : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d5824e0bcf6852c18274f7bbd6bcd66367ea578b","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-20 14:07:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e75dd0e69e8cf06e0e5998a28b479b06999041cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 14:19:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/56/666356/2/check/requirements-check/672e3bc/ : SUCCESS in 2m 28s\n- openstack-ansible-linters http://logs.openstack.org/56/666356/2/check/openstack-ansible-linters/db41999/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/56/666356/2/check/openstack-tox-docs/d212fbf/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/56/666356/2/check/build-openstack-releasenotes/6ddc746/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b94bc4fad5b8ab2bdffa82104583071bee7a3826","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 21:35:27.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"14866bea3f39fd8e89eb67c5cc8bcf4bd0ca5c5a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 13:30:14.000000000","message":"Abandoned\n\nalready merged in other patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bfbb5513b561672a7984438bd00ee36d0d5e0be","revisions":{"d97821743ce1a8179ee027fca26ae171387e7dd0":{"kind":"REWORK","_number":1,"created":"2019-06-19 16:00:34.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/56/666356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/666356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/1"}}},"commit":{"parents":[{"commit":"9188111488c051a9f466b5c28c8e96c2cc3f0481","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9188111488c051a9f466b5c28c8e96c2cc3f0481"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:18.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:31.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nChange-Id: I5db81be8fa6026ff2ca689a80267fd6738d0fdf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d97821743ce1a8179ee027fca26ae171387e7dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d97821743ce1a8179ee027fca26ae171387e7dd0"}]},"branch":"refs/heads/master"},"8bfbb5513b561672a7984438bd00ee36d0d5e0be":{"kind":"REWORK","_number":2,"created":"2019-06-20 14:07:52.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/56/666356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/666356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/666356/2"}}},"commit":{"parents":[{"commit":"9188111488c051a9f466b5c28c8e96c2cc3f0481","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9188111488c051a9f466b5c28c8e96c2cc3f0481"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-19 16:00:18.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-06-20 14:07:34.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Remove unncessary \"\u003d\u003d\"\n\nChange-Id: I5db81be8fa6026ff2ca689a80267fd6738d0fdf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8bfbb5513b561672a7984438bd00ee36d0d5e0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8bfbb5513b561672a7984438bd00ee36d0d5e0be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
