)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~666713","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fstein~Ibf0223d964dd7c31eae65c96878f2dcbcfe5d60e","project":"openstack/openstack-ansible-os_neutron","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ibf0223d964dd7c31eae65c96878f2dcbcfe5d60e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:06:00.000000000","updated":"2019-06-21 00:11:23.000000000","submitted":"2019-06-21 00:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666713-1561075883877-d1910bce","meta_rev_id":"c0e9fdfd1f697f8970e661d4ae6a9465fcd70f52","_number":666713,"virtual_id_number":666713,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-21 00:11:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 22:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 22:27:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:09:05.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-20 22:27:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-21 00:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dbc11cf4ef6ab9e7917924d5e9b5d86b90aecc4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22572fdfa2a5fce2491f3d3fc6b62b3193bf4c6e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:09:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3aeb64090c603e3544a32ef32f8fe53e30445dd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/666713/1/check/openstack-ansible-linters/7cfe543/ : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/666713/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a869493/ : SUCCESS in 55m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/666713/1/check/openstack-ansible-deploy-aio_metal-centos-7/2a5074d/ : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/666713/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0a6665f/ : SUCCESS in 54m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/13/666713/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3c55ef7/ : FAILURE in 39m 55s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/13/666713/1/check/openstack-tox-docs/018e1ac/html/ : SUCCESS in 2m 50s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovs-ubuntu-xenial/de1eb35/ : SUCCESS in 1h 15m 45s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovn-ubuntu-xenial/ccc20c0/ : SUCCESS in 1h 31m 38s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovn-centos-7/15783f0/ : SUCCESS in 44m 07s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/13/666713/1/check/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/beb6969/ : SUCCESS in 56m 43s\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-opendaylight-ubuntu-xenial/40c5011/ : SUCCESS in 42m 55s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovs-nsh-ubuntu-xenial/52ff64d/ : SUCCESS in 38m 04s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-odl-sfc-ubuntu-xenial/1e8d45e/ : SUCCESS in 47m 02s\n- openstack-ansible-ovs-sfc-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovs-sfc-ubuntu-xenial/f61427f/ : SUCCESS in 36m 28s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial http://logs.openstack.org/13/666713/1/check/openstack-ansible-opendaylight-bgpvpn-ubuntu-xenial/910dbfd/ : SUCCESS in 44m 45s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic http://logs.openstack.org/13/666713/1/check/openstack-ansible-ovs-fwaasv2-ubuntu-bionic/99928db/ : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16c32b94c9696b200abdfe601bcd791b5bba07cd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 22:27:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34392d16954299be101d110656b84d7452aecddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8026fe0dc81de43222cd75d36b0770b66fc76d84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 00:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"210447289c0f8fd52620d3acbb62bc35662a35ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 00:11:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/13/666713/1/gate/openstack-ansible-linters/ad3b01d/ : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/13/666713/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/30761f5/ : SUCCESS in 58m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/13/666713/1/gate/openstack-ansible-deploy-aio_metal-centos-7/f65e41b/ : SUCCESS in 1h 40m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/13/666713/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/504f44f/ : SUCCESS in 53m 01s\n- openstack-tox-docs http://logs.openstack.org/13/666713/1/gate/openstack-tox-docs/4e92786/html/ : SUCCESS in 6m 28s\n- openstack-ansible-ovs-ubuntu-xenial http://logs.openstack.org/13/666713/1/gate/openstack-ansible-ovs-ubuntu-xenial/05c1259/ : SUCCESS in 36m 32s\n- openstack-ansible-ovn-ubuntu-xenial http://logs.openstack.org/13/666713/1/gate/openstack-ansible-ovn-ubuntu-xenial/a261ac0/ : SUCCESS in 1h 00m 16s\n- openstack-ansible-ovn-centos-7 http://logs.openstack.org/13/666713/1/gate/openstack-ansible-ovn-centos-7/9a765d9/ : SUCCESS in 43m 22s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic http://logs.openstack.org/13/666713/1/gate/openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic/56c94cb/ : SUCCESS in 53m 30s\n- openstack-ansible-opendaylight-ubuntu-xenial http://logs.openstack.org/13/666713/1/gate/openstack-ansible-opendaylight-ubuntu-xenial/926cd14/ : SUCCESS in 41m 43s\n- openstack-ansible-ovs-nsh-ubuntu-xenial http://logs.openstack.org/13/666713/1/gate/openstack-ansible-ovs-nsh-ubuntu-xenial/d70e935/ : SUCCESS in 41m 45s\n- openstack-ansible-odl-sfc-ubuntu-xenial http://logs.openstack.org/13/666713/1/gate/openstack-ansible-odl-sfc-ubuntu-xenial/b76e0d5/ : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d303d2e4d766965c63290bd3a8dc49da93a05f75","revisions":{"d303d2e4d766965c63290bd3a8dc49da93a05f75":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:06:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/666713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/13/666713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/666713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/666713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/666713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/666713/1"}}},"commit":{"parents":[{"commit":"cb63d273ca51d4b7351076b834dc309e1c4f3ddc","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/cb63d273ca51d4b7351076b834dc309e1c4f3ddc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:05:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:05:58.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ibf0223d964dd7c31eae65c96878f2dcbcfe5d60e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d303d2e4d766965c63290bd3a8dc49da93a05f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d303d2e4d766965c63290bd3a8dc49da93a05f75"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
