)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~690366","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Ftrain~I8b4eebf43379e8c028b029a6105185bf9fd7f49e","project":"openstack/openstack-ansible-os_neutron","branch":"stable/train","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8b4eebf43379e8c028b029a6105185bf9fd7f49e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-10-22 19:20:22.000000000","updated":"2020-06-27 09:44:19.000000000","submitted":"2020-06-27 09:43:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690366-1593250982545-88eb4db9","meta_rev_id":"c3645075309537b3b64648ff737b050361995f8f","_number":690366,"virtual_id_number":690366,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-27 09:43:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-27 07:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-27 07:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-27 07:47:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-27 09:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d644c3b4e666b737d0ed571ca810351c97843c0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-22 19:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fef1dca38a2bb8b5787f2ac5412bbb0ce27d4fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:36:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac73e1b6c6af4ca2ae02e2e0ee0e9820 : SUCCESS in 3m 16s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/39f644f36cc247e3bab777a45e73bebc : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86320142bb7f4ff2b1784c16155dcd2c : FAILURE in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db8012d381e458aac92888673755e4a : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ccf7ed77352ef63c06ba8f540c971494c08466bf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-24 18:35:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b1679ac40950fd32972b42ee2690ed07cd0091c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 21:08:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ba1d6510ce40479ab6a012f725d2c292 : SUCCESS in 10m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d206445a19384c03b5e1af784dfa37d6 : SUCCESS in 1h 02m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/43e6902c1bae49fcb15d40df30714c72 : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/407f7e31b06d4e5388d2497663213956 : RETRY_LIMIT in 11m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd7aba022e0847448b005987e6d0fdf6 : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d63aaec10e1b466b950aada799c1a7f5 : SUCCESS in 1h 14m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/89ce1f787dd4426194915b09f21120f1 : SUCCESS in 2h 09m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfdf76d75d5543108bacb5bef0c7491a : SUCCESS in 1h 40m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b482b3bb5b1c4d4ea19fbbb577b331a9 : RETRY_LIMIT in 13m 53s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/926f657b35b64d26a30c940204d3b3c6 : FAILURE in 34m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a9eec5993d74985b7216b0956c4e250 : SUCCESS in 55m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b220ade9dd704285aa05baaa2da8a223 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54cbd0b28762440eb726ba59a5212104 : SUCCESS in 5m 56s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/709c6f0909c945dab161a19113d53b27 : SUCCESS in 45m 16s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6331310b208b46eb814df1bb87864459 : FAILURE in 28m 40s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/68a60b02bba54aa8ba2521d5f75ec37e : FAILURE in 30m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d74af9a58b1c44c2ace0d40fdb3557f1 : FAILURE in 1h 06m 15s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6e0856bfa40428ea07db8c30121f6ff : FAILURE in 36m 24s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3682abe65b234223a077b2be773f97de : FAILURE in 30m 01s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9948a44859514e499dbfb2de56f38b35 : FAILURE in 31m 00s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/091821c777514c0f8b78625ac07cb4c3 : SUCCESS in 48m 47s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0923545d4115480c83254d788c3e5048 : FAILURE in 32m 52s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15ec5159f0ac42eeace3b191a8c90f2c : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9916622f11112d19cea5a1ebe68c2ce811045a9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 20:31:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"59e0897c602e5a4de1a0d8808a4cd713594f2be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:04:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/915bfda16e5d4d22ada6e71396d4a55b : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fa83c9abcb2349d1b4330440e7800ca9 : SUCCESS in 1h 04m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bd5f3176680c407ca272562037078a82 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/413a037170c34cf294322c363d9c9291 : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/80a027c5e0684f419b8d0f1ad702612e : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a6ba8b2f0de4da1910d098443856a23 : SUCCESS in 1h 11m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f67508c2ab17429a95c9b78e3135d96a : SUCCESS in 2h 41m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a239fda9ae9449b2a808fb6ce28554c1 : SUCCESS in 1h 50m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/380466fc90aa4799bb7f8e5425e152e7 : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2b8d4f9055574d7dae925e1f36919518 : FAILURE in 34m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e243282c78f4d14ab2ac450c840bcb1 : SUCCESS in 58m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9293aad5d91c4770b27768f9a1f8e372 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/260eabe4c8b94b92a9abb44c761d8499 : SUCCESS in 5m 00s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14f231ed265645cda714319541867618 : FAILURE in 39m 48s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2118594dce454a878d8b85222f72d162 : FAILURE in 33m 09s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/c2643cd41b7b446983fd3cd2d3060dec : FAILURE in 34m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ae83e4b999b4edca3141eeca94f777c : FAILURE in 1h 05m 11s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b91552d8416445f4b6252fad4b2c9d22 : FAILURE in 29m 21s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acaf31108d594bf6a53126fe06d8b177 : FAILURE in 31m 45s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab87da52c7c34a1dbfc48fded4ba3cf6 : FAILURE in 32m 25s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/175c524cf5ff4f1ba11a610df2429929 : FAILURE in 38m 42s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8062e2ff3b234f8b90a5e77c4c0e0b57 : FAILURE in 33m 34s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6130563565d74b609ed67d7e432c58d3 : FAILURE in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9496432386cdf1b9a1252de8834070388d0af5de","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-06 18:13:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5dff4cb277345fc18a828ee84df9d13a48e7de3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:59:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ae271b881b440e0afcb29df117b86a3 : SUCCESS in 9m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3b5445c793314fe8ba6be1bafd8cef4d : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cdbdae8b7759479ca80b7951463ce865 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dd7258913cc34d0398c00d0fc9fe88e5 : SUCCESS in 1h 41m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2217beff040749bc924fe6ce2158c1dd : FAILURE in 32m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/703fdc8596fd48b58d4b97b72134f5eb : SUCCESS in 1h 18m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d9241708b9a94e468b17d4e497a40f8b : SUCCESS in 2h 21m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9edc7ddb98f74160ba58acc51c6ff9a2 : SUCCESS in 1h 55m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d547a322f23a4643816c530ccae50c71 : SUCCESS in 1h 14m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/425bb1900619433095611b6cf51a6f6f : FAILURE in 34m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3722e8ca52b24398b703ec82952da0a2 : POST_FAILURE in 54m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cc1a0ac2b5453da61780f58d75083b : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3c8d96287e04836b154ced58b9a9f17 : SUCCESS in 6m 30s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b449d04018b4ec1a537dc396cc1ed3f : FAILURE in 58m 52s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d723799639424e6a8b8637cf05e66ee5 : FAILURE in 29m 59s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/26f817c6606e41fb9871323d754c24af : FAILURE in 29m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c4e4fe4a9b3480388f8af0771dd9ab8 : FAILURE in 1h 10m 52s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1ca1eb8a7cc42459ab5dfd4763d96d3 : FAILURE in 54m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fc99645aa064380a0cf3c17d35b6157 : FAILURE in 47m 04s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1f6e3b9eba74688a0bff4324c586f83 : FAILURE in 34m 40s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/843b870faf584043a206c5d262e0b833 : FAILURE in 1h 15m 10s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38ce4a4e372949c6bba248005c6447da : FAILURE in 38m 00s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/698136f45aee436f9fd2a8648f52243f : FAILURE in 39m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eeaa270067e9feda0e9cc388f13cbc52612daa6f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-21 13:20:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c5bf28702f8c4ba8dccc44194e8c8947d32b36e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:03:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6ae836e682a14f4897afebca98ff88eb : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/126f455d0da14fda83358c922cded477 : FAILURE in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b3e58a44abf74e74938de759a4501196 : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bec236f7033f48398f71d70bc455fc42 : SUCCESS in 1h 37m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5da019f6c4b14d3e9df3b52d27334cdc : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5da48b1d450b4814a7089a4985555abd : SUCCESS in 1h 14m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/77b11e7b587e4eefb6fe4355178c134e : SUCCESS in 2h 37m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36881d28e02e4ac2a1a1a1f5d4d1f5d2 : SUCCESS in 2h 00m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/83e9d80b25d743c0b8d13ba79ccf710f : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b2fa46bb2bd34b08bfdf6b178c25f1b7 : FAILURE in 40m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0de04a1725044059449d339b3ddf0d1 : SUCCESS in 1h 03m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf9da289fc24e908851ccd80bafa984 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30333c5d8e2f45289fd13e7606887655 : SUCCESS in 4m 26s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bd18679025340438fb832dcca45719c : FAILURE in 46m 17s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5e6001646124c78bcb672ae588790ba : FAILURE in 45m 44s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/11c1fd9372894e2399eb7fb096ddb536 : FAILURE in 36m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/361e1fefed6c4756b9718b4ff6c7a6ba : FAILURE in 1h 13m 26s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6287fd0a380643e3acb1aa9a688707db : FAILURE in 50m 33s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31d96b3da8474d4094d1c15740195bef : FAILURE in 51m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87ff756b225e4382b596c6d96d9c7c2b : FAILURE in 1h 07m 40s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f6f5a19f31d46469542cf0f0eba307e : FAILURE in 1h 07m 44s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fedd793ce052403fb66d000d8d261846 : FAILURE in 50m 25s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12eeddfedd5a4760af5a9a85b6dfcdc7 : FAILURE in 55m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e50f6e1dc67bc8399449ebdfa78d4a233cfb1d1f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-26 18:37:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7cb987e561cabfa68652124fbab4a3c8ae93b240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:15:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/36dd6a2fc82e4788a4d19c94ebdf919c : SUCCESS in 7m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8d953b5a8783444b9d160630b9c029d9 : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bfa7c4f2fc0c4f3e93ed95eb10d6917d : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/21d42b60671046c7855f653bf009ec9e : SUCCESS in 1h 39m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bb6f130d2fb345b1adfa7575ec7edee9 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c9c80b0d23844c38db41cdf1e323f6e : SUCCESS in 1h 18m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/44f7b67e3d91495ca36a3192208eff23 : SUCCESS in 2h 46m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce12d8af650e47c0b1e58a2156c85914 : SUCCESS in 2h 03m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa65d960ab284e1eabb1575cb90b9f8e : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1c4c4afb39e2426eb21811f6c2538b95 : FAILURE in 34m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7f984bbe9cc44d580318f757d6c567d : SUCCESS in 1h 00m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155e98f72c814b97b036cd94b5183b07 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77bde530cc84078b1b012b07f399d1a : SUCCESS in 5m 58s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bae9c5a2bae4ce18401ed50fdfd8da4 : SUCCESS in 39m 32s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5c58514a29e4317bcc3e0268f82d3ae : POST_FAILURE in 38m 50s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/6a7c520787724108925512b457c64935 : FAILURE in 29m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/345904c958ba405289de5e1eeb499b4d : FAILURE in 1h 03m 32s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70897689d0554ab9984b1ff6c7a4eaba : SUCCESS in 48m 37s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f277fcaad6c4b8d9c578523b3c8a3be : SUCCESS in 48m 44s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74d8433aca784e46a44be76060b05422 : SUCCESS in 49m 25s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd83187c95264cccb0ae43ac1c49f02a : SUCCESS in 43m 29s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11336e052cd24f3190a4dd95e822da0a : SUCCESS in 46m 25s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25f6e8da31de46efb5177375e066000d : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9038a04035cadf6d1ea566f19e0979182399f36c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 06:01:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9429e58a1a7c2b5d16e0f9745a5c2c86f5342a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:51:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f40c4564cae7440cbe681da315836099 : SUCCESS in 9m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8578dbf947e74edea4c13520e0d5a8f1 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/438afddc0af3463c90aeed456bb92e0c : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/feed999f00c54c7ba29a6f036dd87fe3 : SUCCESS in 1h 43m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0e619dbd37a8425982194b1e394996ea : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86097fb081db4007aba5997f6f5b4158 : SUCCESS in 1h 13m 14s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/66a41f13f701420a85020c7a187f2672 : SUCCESS in 2h 28m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cad21af1eae47e1a360dc508398ae2b : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9995d958b954412e883aa25fae241ccb : SUCCESS in 1h 11m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1a4da33a47d6449aab0f1f341f5b7ecd : FAILURE in 47m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b26222547fb54a9fb6ede9a7dc031408 : SUCCESS in 46m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc0a7bce97748b59aee75f84f161384 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b96df4d662d4bac9820f7cabd4d44d3 : SUCCESS in 5m 33s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b84db2355d74296867b91eb66638cef : FAILURE in 11m 02s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3e65de388b74a78850cf333b6f6aada : FAILURE in 10m 04s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/1400f6f3c61049dc9495bad609540652 : FAILURE in 33m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cc2e63cf802456b82826ed17d2cf169 : FAILURE in 50m 58s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/956c3a48bc9042fba3128b4204b83772 : FAILURE in 11m 32s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89c1641dfa1f4f5a88f2b6b5b012ae89 : FAILURE in 15m 48s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c7162967b6845819dc3e98ed4336883 : FAILURE in 15m 39s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c9205b5c33e49fe86f370c2388db3d9 : FAILURE in 11m 57s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfa354bea8a945f6b39d2675e5c9ae86 : FAILURE in 12m 24s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce2dba274b1f4152bb661778d7eccd90 : FAILURE in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6fbdb17be05e037bec68ee51dc6b72d26ebf3906","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:16:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa99ae6ff4881f5438d2aaa8ba6253751c6517ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:19:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/813a883bae154579b8de29ce6282a897 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8f2c138f72604bb997e51aa140774d40 : RETRY_LIMIT in 14m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7ef76300f4434581836e21a35f3b14cc : RETRY_LIMIT in 10m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/68f135c61bc945ff9979ea9b6b4a2a2c : RETRY_LIMIT in 11m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40399acb8bd74cec90252f46f8514685 : RETRY_LIMIT in 18m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a39e5286b79f4bd68376e5359fac3f6e : RETRY_LIMIT in 18m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35a01a47c36848bd91ac0c30cde3e702 : RETRY_LIMIT in 11m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d75cf15393242ee82c9f083e09f517e : SUCCESS in 1h 53m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2599ae0f867f49f1a848cce9850e405b : RETRY_LIMIT in 11m 37s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5e3bfb7cdf81440585651cde6d2c29b2 : RETRY_LIMIT in 17m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e68f0315258470f9939391a91a4c8ff : RETRY_LIMIT in 24m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fac6469bffd46e08236557f1895608e : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fe00ed89242483b98354e85842f7189 : SUCCESS in 4m 04s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60f43b491b6843fc8f404f8f607ab943 : FAILURE in 8m 50s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82b7f741e86448bd8669eb374b476e2e : FAILURE in 8m 53s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/ea8a2f91942c44cc9e025beaf3130e02 : FAILURE in 21m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e738361f23d743efa5e6de82eef0769d : RETRY_LIMIT in 13m 40s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e46a51e29f6e4936b8464dbff6b5f8c2 : FAILURE in 10m 32s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dac62811d2df416eab8b1c6ff83ef20c : FAILURE in 10m 05s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d7ec9f4954f4997939c318ca5abcb38 : FAILURE in 9m 43s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c995d2fe713e49829f2cee17f5661775 : FAILURE in 12m 47s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f5d20bfd3b4f508c1032b6068b908a : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adb506c8c79e4ea99d234d9b2880c879 : FAILURE in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab940c03a8cff58692c9eaa8cfdf72d9fa0a3ab0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:12:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c5de4fed3d6c56a982bfc72d81e0030f462a8315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:03:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42f3214a178846449f0bcfa1d5b0c014 : SUCCESS in 9m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/276dfb8678174165b5678571c3321977 : SUCCESS in 2h 27m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e1bc309e6e5442e4ae266ab107443e4d : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ec965e6096f94e778761e47d9a56c42d : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b6bd90ea29541069458f8cc8c270afe : FAILURE in 1h 03m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3cdceb230db41738bbaef7d75a87cdb : SUCCESS in 58m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c8f7f4d520da4d3ca984b49dd5cb91ef : SUCCESS in 2h 31m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d9564cabd4a4003b80ab8132b95ebde : SUCCESS in 1h 51m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/82bd358cdb7240c393c6c503a3e33bb7 : SUCCESS in 1h 06m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7008440900bb4346974f1f2fc74fc7b0 : FAILURE in 31m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef6f92a919534739a98312fb7eb6cbd1 : SUCCESS in 1h 09m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8990139067440d862250faa213d1fb : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/665fd44c9b3f49c9a9b8d0dd7c67b221 : SUCCESS in 4m 49s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea59fe0eb03e4cd6b24aa68b516b64a4 : SUCCESS in 52m 20s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af6b1103d1ba44579c11115761c5da49 : SUCCESS in 1h 12m 56s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/faa2baa18a79471692c681e8b3fc4930 : FAILURE in 34m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82eed3ddb7c246b8be5e3a11704d66b2 : FAILURE in 1h 10m 26s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a706dbda27bb4ccfbe85ad87d24d3d3c : SUCCESS in 52m 26s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/225924b8965b45a191a7adc7414e46d4 : SUCCESS in 1h 10m 04s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/069f105abb2c427cb420553b61ad7245 : SUCCESS in 48m 17s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93f6c834b91449c7b6a7e8898020e52a : SUCCESS in 40m 12s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8423d4976b64294a2c2c2e75ff98247 : SUCCESS in 51m 22s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d4bfb91fceb4bfc8dedd69db0e6143c : FAILURE in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f38e643e3414228c5ed10026d7248bc61865f7d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-27 07:47:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8cad7d92e89fbd5c8738f09ea1b1f29b362b1fc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 07:47:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6dbdda81eb0c77689747382be97ffb23cbfe9963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 09:43:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/553f1dcfa8994188bac4e4b2d1f8e8a6 : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/01b9a5383c714dfdb422c6d3a800af23 : SUCCESS in 1h 00m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/07ce6c0afafd409baa7c9c27370f8345 : SUCCESS in 1h 50m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae49677a19094e15bfe07f79db0ba849 : SUCCESS in 1h 41m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a8dd08f4cac8406cbedfe23e4ed86529 : FAILURE in 58m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f92015c8af5f4681a4da292c44b89595 : SUCCESS in 1h 06m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ceca2a7b65042ebbac27af1b11de164 : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3565d42df87e456298ed615a9f339449 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c55cb88fd53431ab9471e6d29906bea : SUCCESS in 1h 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abec21031b6145df8e0b8b92be32d0ae : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/855c4b63843a4d6b8dd4feb0768831ea : SUCCESS in 3m 37s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efc49ac7583d4ffcb942515b5aef9fa1 : SUCCESS in 37m 33s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efee91bcf33a4b13a4991e5af837086a : SUCCESS in 46m 40s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcd5968057cd42118b374c76394bc0d8 : FAILURE in 1h 08m 42s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac412ee34ddd4a2a8d4cf08dcb23cc23 : SUCCESS in 41m 49s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/860b9ceb62a747bca1c0ff48f788a8b9 : SUCCESS in 33m 30s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/148ceed74b5446fea4e01d7d73ad4c64 : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b392260a00ec828e7b75eeeaf2a042b1ca51e441","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 09:43:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bc9ab2b1b5508daee5ac440fd472ac276a8d0926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-27 09:44:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c15b1e008247d1b054de2575380d82 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec190ed0a93c46f6bad3ef43f77292cf : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dddf5ad1ee7ff450e99a76a5b597179e384c9ba8","revisions":{"1e3c5a0f28d8de60c3062d98d5cba69a3564b5ef":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:20:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/690366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/66/690366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/1"}}},"commit":{"parents":[{"commit":"2f1b789e06b3e91201ca312c81c9d2a17fc322f8","subject":"PDF Documentation Build tox target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2f1b789e06b3e91201ca312c81c9d2a17fc322f8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:20:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-22 19:20:18.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8b4eebf43379e8c028b029a6105185bf9fd7f49e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1e3c5a0f28d8de60c3062d98d5cba69a3564b5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1e3c5a0f28d8de60c3062d98d5cba69a3564b5ef"}]},"branch":"refs/heads/stable/train"},"dddf5ad1ee7ff450e99a76a5b597179e384c9ba8":{"kind":"REWORK","_number":2,"created":"2020-03-24 18:35:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/690366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/66/690366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/66/690366/2"}}},"commit":{"parents":[{"commit":"b35e1994ddd01820d25343d4b29df787ffc99f0e","subject":"Add extra rootwrap filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b35e1994ddd01820d25343d4b29df787ffc99f0e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:35:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-24 18:35:37.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8b4eebf43379e8c028b029a6105185bf9fd7f49e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/dddf5ad1ee7ff450e99a76a5b597179e384c9ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/dddf5ad1ee7ff450e99a76a5b597179e384c9ba8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
