)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~702313","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fstein~Iadcb4c7451cd51526dfd96b305a9d0b1948ce8da","project":"openstack/openstack-ansible-os_neutron","branch":"stable/stein","hashtags":[],"change_id":"Iadcb4c7451cd51526dfd96b305a9d0b1948ce8da","subject":"Update vpnaas rootwrap filters","status":"MERGED","created":"2020-01-14 00:38:08.000000000","updated":"2020-01-27 12:26:34.000000000","submitted":"2020-01-27 12:25:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702313-1580127903366-c94174cc","meta_rev_id":"983437535f58caec0fc26411753c49cba0ca099a","_number":702313,"virtual_id_number":702313,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-14 13:05:50.000000000","_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 12:25:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-01-27 11:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-26 08:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2020-01-27 11:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 13:05:50.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-01-26 08:42:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-01-27 11:07:43.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-01-27 12:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9c8a22f4b0fbe6c205d476d1abaa0b848bc2ea1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-01-14 00:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f236e2817965d01932f551a5cd378cc8f94a9b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:09:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d95a6ae0736a463489313afa5b45647e : SUCCESS in 9m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/36595ce65d5a463eb19d4895a5673d97 : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b488f3b68334d05a0068872a73e8bb8 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a2235dfc45f4b6580bcc4366ac39235 : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b6b90b64258b40c1b99a8094306c960b : SUCCESS in 58m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c43de1d53044c9f9b5828f057dad360 : SUCCESS in 41m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053575e25d644553979cc7850cfa210b : SUCCESS in 2m 50s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07dbfb39e02e417381208feab15be796 : SUCCESS in 36m 05s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f99cc37a7e17480b9df736b801ad2a6c : SUCCESS in 42m 14s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/43f802aadc5242a2a2c77f9bce878cad : SUCCESS in 47m 22s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7845ab41a6f4f45b55abb5538d53816 : FAILURE in 50m 20s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb61446a31a9410294d9e377d2407255 : SUCCESS in 47m 50s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ce32b83bc1c4333aafd4ad1917bb51e : SUCCESS in 38m 40s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a6ce4297de7449d9cab210c0e0d67dc : SUCCESS in 41m 59s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a932a65bc07e44ad9e89e760f2cff498 : SUCCESS in 36m 33s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3539fd41183e4ee2a5f8ac080a660d79 : FAILURE in 37m 25s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf8ac1d3296448a0b9349347eb0f763b : SUCCESS in 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7dcd68ba633a7822f5a994f418605922b60dd6c4","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-01-14 13:05:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4c46e6bcb5f32b816b46e651521c0b15686a80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:45:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e955a2aaeb774e4d83b04ec42547c7e6 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5d5853e078364c35a772489c1f736716 : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e09231ca3fd7437a9b9d341f3434d4de : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/222c221378db4b23919f80861134862d : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/381cdf7924f049f78a53704546376878 : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f79b43b692f840baaafd04c4ff4e0da0 : SUCCESS in 49m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c0a7b7254348bba02027318be6542b : SUCCESS in 3m 52s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de86fa8e40a94587a2c64935c5c7b23d : SUCCESS in 58m 41s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6d25a8f19334e8aa34045756192d7ae : SUCCESS in 41m 15s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/9b36e7e3f055455d89c16e5c0eb5e91f : SUCCESS in 41m 17s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f9ff8b43a094518a2094471f7fadb61 : FAILURE in 51m 42s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de4d2197bc234c2288418b2bf15d1cfc : SUCCESS in 52m 41s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63a34fd7a9a34f349244c6eee7f007b0 : SUCCESS in 51m 43s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/926b863d893c4db5b917079e1dc415ef : SUCCESS in 45m 53s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be5469a0e81b4276826b8b373152dd28 : SUCCESS in 38m 42s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00d253a4a84445d496cb0c3e95f1c441 : FAILURE in 40m 17s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f28df3405c84f778844f8add9b85930 : SUCCESS in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b985d66fa8ff179b8a860db3b85298894545c166","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-24 13:42:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e095ce65448f30a8ad0a79b48014b46f4256f5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:31:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b9ff70d0c6384e8190547cad45ff0bdb : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/21a1665f65fe45a3a29a8928dd4fb2e7 : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d15ae14d7e046deb4486f77970a3dc0 : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bae5f8b184934f168f689ad21ff7e886 : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/188014da3b664079bc0ec89688ce88b7 : SUCCESS in 1h 06m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/978c2e917c394457b87764e430dfcbc1 : SUCCESS in 48m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604357ef5a83486eaaedaa350394d71f : SUCCESS in 5m 07s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/296dc540a9ab4647b77ec696cf155907 : SUCCESS in 55m 48s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/037b963c36364ac8accc57e26157810e : SUCCESS in 41m 09s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/bcf1796ff6944630b4761e5285908d9a : SUCCESS in 47m 01s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38d5df464384482d8b5ecd8690413a7d : SUCCESS in 1h 03m 20s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cbe2f02dda443829c63bcb32099b165 : SUCCESS in 42m 37s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67834129a9124303b701222cf3b84bea : SUCCESS in 53m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18ede509f09347b49b02e90a681129bc : TIMED_OUT in 1h 34m 26s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff90ca68060e4b97bdf4cbe4068311a0 : SUCCESS in 45m 09s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c875869f70404287848f0134bbc2a20b : SUCCESS in 49m 02s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c92d5751089e4200bbf444a0c895d25c : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23180b19bea6048b7ab2aae88f3c152aed60b873","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-25 20:08:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22a7ec78d8861774602f03bc9ea22fee2930d891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 21:25:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2596b819e1f34dbbaeb8fdf2f393a38f : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/404738582a83475faa5d99ab6a1a9390 : SUCCESS in 1h 03m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cdbd9ae5e1a64db9b3a8fe71d560f75e : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b66a45a1ccd4f2482b512691907767a : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a86617ef394b4596b66399727a971413 : SUCCESS in 55m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f48634696cc46aeb85ff9690bfbb715 : SUCCESS in 43m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb97b913e714c4a94b7392a70cc476f : SUCCESS in 3m 39s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b513e02641b4170b371141145d995c1 : SUCCESS in 40m 15s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34e96eee5e124e3db1a5155c08cb2fa9 : SUCCESS in 36m 16s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/524282b0c7cc469aa28e3a5234626591 : SUCCESS in 43m 47s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80f1f9982fef4e85bddc797888d764c5 : SUCCESS in 58m 00s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ca6c8cdec3425aa42cc27cf70b78a6 : SUCCESS in 43m 02s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e2d8c91359f4b378dc9d04316369424 : SUCCESS in 40m 59s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b00d2b526b14406a157aaaa6d9db41e : SUCCESS in 40m 50s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81e8402d07ba4d1585ab9eb3b16498a2 : SUCCESS in 38m 07s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bca3c4a05cf46e9817e37beb53a6506 : SUCCESS in 44m 58s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b02167d5af5047ee883395fcce0db94a : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf92bb72f39f65c5b546c48472fe5560ce8b8ec1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-26 08:42:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef623adbcb307514f84e4d19f6920d272125f0e1","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-01-27 11:07:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af02971d62f0834c4df49b5cf889fa32e683276b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 11:07:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26aab2f3bea003f885aa7a83de0f8017d1d7bfe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 12:25:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba554cd304f985450570ebb81d3906c4232b69ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 12:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d3c19c6b30934f84a6a261693af2e888 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/29b7854ca9ce4c5f8026c1ba54a2e0a6 : SUCCESS in 58m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/faaef989d6284301978a021a9bd55d5e : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c2ef56975524feeb716a46d489fb8e9 : SUCCESS in 1h 00m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ef6b3fd694e94a43a7ffb87d414cdf3d : SUCCESS in 58m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d80ebef534e44bffa5aae138cd2c8bf5 : SUCCESS in 45m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ee9d3744a44511921378adf630fce9 : SUCCESS in 3m 23s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/138285d8b3264b71bbf342618e148954 : SUCCESS in 40m 13s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/336464cbedcb43e38d0ef4305c08208f : SUCCESS in 40m 05s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/2822f6f9d01a458faf4e3795b4339868 : SUCCESS in 42m 13s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5cf19fdaecb420b89267c5110ca4faf : SUCCESS in 1h 01m 39s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/551ec61a4f87411d83969f2fb66575f9 : SUCCESS in 42m 40s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c6d9c4f1dec437882f57b1880ac1aad : SUCCESS in 43m 58s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e05d0ed83d044676a48a257ad279e320 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"983437535f58caec0fc26411753c49cba0ca099a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 12:26:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0907ab9901224d75ac4ac8dec1e8c8b0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa4b4b095761c1ffdbb0d068a135eec9dad3fdc3","revisions":{"aa4b4b095761c1ffdbb0d068a135eec9dad3fdc3":{"kind":"REWORK","_number":1,"created":"2020-01-14 00:38:08.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/13/702313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/13/702313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/702313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/702313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/702313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/702313/1"}}},"commit":{"parents":[{"commit":"5b0f6574e3f59e49843cfeb5a582f26eeaf40da1","subject":"l3 agent on network_hosts do dvr_snat, anywhere else dvr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5b0f6574e3f59e49843cfeb5a582f26eeaf40da1"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-14 00:35:33.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-14 00:38:02.000000000","tz":-300},"subject":"Update vpnaas rootwrap filters","message":"Update vpnaas rootwrap filters\n\nThe vpnaas rootwrap filters are out of date and therefore not\nfunctional on the latest release of OpenStack Ansible.\n\nThis updates and adds all the missing ones so that it becomes\nfunctional again.\n\nChange-Id: Iadcb4c7451cd51526dfd96b305a9d0b1948ce8da\n(cherry picked from commit 6ab3f196883b83bec284363b87fe5d1d1d831bb5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/aa4b4b095761c1ffdbb0d068a135eec9dad3fdc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/aa4b4b095761c1ffdbb0d068a135eec9dad3fdc3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
