)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~709500","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ia08ac22121a5188aebb79f4976957594d0b89382","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia08ac22121a5188aebb79f4976957594d0b89382","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:12:57.000000000","updated":"2020-02-29 12:40:01.000000000","submitted":"2020-02-29 12:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709500-1582979922379-cb471cbf","meta_rev_id":"97f43e46058ba4dd900aef148985f9cffe1e8064","_number":709500,"virtual_id_number":709500,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-29 12:38:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 07:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-26 07:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 07:01:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-29 12:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"438fd2783dcf55fb7d0073a2aaa1039c41114e6b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3d7825ffba6130c46f356c3f48210e04b2a830e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:05:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4a9f9fa26f824cd69ef8f3707ff80d93 : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cbf483bb8f5b43d0b9f213eec65669fc : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cf7bdc4ddb5249dd94a2127097e015b5 : SUCCESS in 1h 37m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f09143c174074462bfbfed572ebace16 : SUCCESS in 1h 27m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ce8d208495d4a27a21fd2722adde3b5 : SUCCESS in 1h 12m 16s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c879cfbae35e418a8253681ec0bb0468 : FAILURE in 1h 28m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4e9cbb3b34440979465d96b177591c0 : FAILURE in 1h 00m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40ebdc5e5ccc401791442864e303684c : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ff6c8cc048de4a548878d31804ed9d76 : FAILURE in 28m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/052f0b608b2c4203bb043c685d2daf6f : SUCCESS in 52m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf8ff26fb33480aa185948fa97c83b7 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089b7c993f88452db890746e2932d1e3 : SUCCESS in 4m 50s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/753e487bc1dd4736b534a46319785c5f : SUCCESS in 41m 55s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d98209ad6fb143ce99fcd677f5e48838 : FAILURE in 27m 44s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/97db7d83613641869ca41a911e31ab98 : FAILURE in 34m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0c74f37e2234e8aacdc585793f8bd85 : FAILURE in 1h 01m 37s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93550ba0080b4509856017b8b3493e15 : FAILURE in 30m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/896a9d1139834ad28f3bf5da7c84ba73 : FAILURE in 31m 39s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2574280f01c149b3981f9f20c040ce35 : FAILURE in 34m 55s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7cb9eb628acc4b75aaeef228616571d4 : SUCCESS in 42m 45s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b1dd2665aff40feacade6e5b69ac868 : FAILURE in 29m 11s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fad0acb85bfc4eada19fc004951a7d54 : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"646aa783f1fcc0547a0c8f87c6f06ac1514de161","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:51:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d628fd24a13a40918d26a644d10a1048101b53a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:57:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5c7ad8e00ea1445cb44036ed8955633e : SUCCESS in 9m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3b34c5c3393145449ca9abb011243140 : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ec5d53612de740769b1f585345262655 : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e0a20a2ba7a348fc9a3d58bdeac7c53f : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d82a9bed09e4d68967b399ca9f1bcfe : SUCCESS in 1h 30m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a2ea44f2046e41a4ac4be42fe2268933 : FAILURE in 1h 29m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a4536b64ea84e81a71b204c9db35469 : FAILURE in 1h 15m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fa23e6aa925d4a588bb6dc682fcfbfa1 : SUCCESS in 1h 14m 04s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/12c56cc0575b4a3bae03dbee5a003c74 : FAILURE in 28m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcbaf6fc6e5e4b949785ca308c8f3298 : SUCCESS in 57m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4774c82407541c48367114d2c8361bf : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3642c5855e94787a1518af8fa1b58c2 : SUCCESS in 5m 52s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/638cbd2e4486482bb77992d0c794d57c : SUCCESS in 44m 27s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/154bd4a0778245a4ac1e09b9cc5bcde8 : FAILURE in 34m 46s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/72ba10dd17cb4bd0bf50a9fba4c7261c : FAILURE in 32m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd9bc7fff096419388fbcc44572d022a : FAILURE in 1h 04m 01s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b57e2b2eb7c4085b64fe7df3abaa1e5 : FAILURE in 30m 44s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47b4d6aeeb02413daa17db72a17cd433 : FAILURE in 33m 01s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07e3c331093f48d4bd197a198ae53e04 : FAILURE in 30m 33s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dddd917e5264711b9850598a728312b : SUCCESS in 43m 29s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8d3aaa5adb9468787af3a13f7c4f525 : FAILURE in 29m 40s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90cf4c0e6a604eaba6a0a1945fa891a3 : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f6c6f55c9684a51c7afddb93bf7b4e9f4fcdbc0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 21:11:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b9eaef32a720e0a333cd8094ede2f27a57711b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:50:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de29e8d6f9014c20b4205441682fc7b2 : SUCCESS in 9m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0b1c3e7eab154200868eb427d4f96925 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dd423a51b5924590972cd6d63a1534a2 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/426c4c5930c14d9c953977b9ef249fcb : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49bc41d588704b0eb57b7a612916b894 : SUCCESS in 1h 25m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f90172aac0a438b9a8664724c58eeb6 : FAILURE in 1h 18m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/256525bd45e5428ebce07abbee470dd3 : FAILURE in 1h 03m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d81118a9f25f4172909964cce8bd4a18 : SUCCESS in 1h 08m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1fa10193fc11482fadb5e26e29a9dd22 : FAILURE in 33m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b51e83aef9249239b595da539719a00 : SUCCESS in 55m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9cd7e828cc44408b6ec0ec16e779506 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e76b0df5b16e45ddab84c365d886ca17 : SUCCESS in 5m 10s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13ac0c561be24a69b8832927e0d80fc0 : SUCCESS in 44m 17s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/006b5342d63944efa57248119449fa40 : SUCCESS in 42m 45s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/5603281f2d294616a1c891680c80f700 : FAILURE in 33m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/117d5c55a778475fadbcdb9d10b79572 : FAILURE in 1h 08m 18s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54dd51fb8e954ecc82eaa7a86ca97a3f : SUCCESS in 50m 08s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40290878133741d1a7a0f10a22d93e53 : SUCCESS in 45m 38s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32c44647cf5840e7a33e795baa825810 : SUCCESS in 42m 32s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f96b8696cc5472dad35d0c32a118ca5 : SUCCESS in 47m 13s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a487e556229e4e69ad0c0cd6b8d1c185 : SUCCESS in 45m 16s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c6505b97d7e4bad8f18837963ff4128 : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b705884aeec4c3e1374f651d872e8c8c0895ae7b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-26 07:01:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ae8095081a27d66aa56fe2f6eac3b334ad571c6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-29 11:13:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"48015595dde432127300d4810489dbd280c3c105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 11:13:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ee47ef3c19306f3af9ff6517f146354abf266d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 12:38:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f2cddf0d389e490c9e241e95c97525e7 : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d50ebcba90894285afe8313e61db29fe : SUCCESS in 1h 05m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/47bd705c982540aa8ada42439edde92d : SUCCESS in 1h 24m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8d71411b6aac49d69e7d0d1043e3337f : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc01fcbf68cd499881b84ac6e302eb33 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8d8f4dbf36e49928fefec62e8a9f500 : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62ffc1409ae04152917ffc869e46ab15 : SUCCESS in 55m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38eb0c7506e242dc9008d28f5f9948f3 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19bbc8db78a84e0f97af70afdbfd4b40 : SUCCESS in 4m 47s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b7e427eae1641519a1a75fd6ae48876 : SUCCESS in 43m 51s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15eb6d73d2244f15bfc045da3aeb3ecc : SUCCESS in 41m 01s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/505704321f654deaaab34292b251ac49 : FAILURE in 1h 05m 08s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/567626ae5bd645bbbf9f6a98d016c3fa : SUCCESS in 45m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70fa591fc0b94f88bda5e78f664ff479 : SUCCESS in 44m 15s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0955e07de40e4f3089d83c1299111047 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b550e36d7add92a55eab3f7d6cf47ea66e532b28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 12:38:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8123949694f367bf6217fcdf58d52bea708e40fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 12:40:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4164e5f400401a8939e4f637053ab1 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62315b374d6e45629a47a370afdd458d : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0de8517a6eeba93d12abbab73f99ce1260aa2fa4","revisions":{"900ad2a1587e4d161e00b64485471a4b4169a6d1":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:12:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/709500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/00/709500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/1"}}},"commit":{"parents":[{"commit":"db41d9f740ce66fcc0c878712d6cb5309410ed69","subject":"Merge \"Use appropriate vpnaas drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/db41d9f740ce66fcc0c878712d6cb5309410ed69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:12:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:12:53.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia08ac22121a5188aebb79f4976957594d0b89382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/900ad2a1587e4d161e00b64485471a4b4169a6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/900ad2a1587e4d161e00b64485471a4b4169a6d1"}]},"branch":"refs/heads/master"},"0de8517a6eeba93d12abbab73f99ce1260aa2fa4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-25 21:11:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/00/709500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/00/709500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/00/709500/2"}}},"commit":{"parents":[{"commit":"db41d9f740ce66fcc0c878712d6cb5309410ed69","subject":"Merge \"Use appropriate vpnaas drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/db41d9f740ce66fcc0c878712d6cb5309410ed69"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:12:53.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-25 21:11:18.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: https://review.opendev.org/709720/\nChange-Id: Ia08ac22121a5188aebb79f4976957594d0b89382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0de8517a6eeba93d12abbab73f99ce1260aa2fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0de8517a6eeba93d12abbab73f99ce1260aa2fa4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
