)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~714403","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ieaad930cf022e2d5ef4a6b580db16ad6454569e5","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"Ieaad930cf022e2d5ef4a6b580db16ad6454569e5","subject":"Add systemd-devel to neutron_devel_distro_packages","status":"MERGED","created":"2020-03-23 09:30:41.000000000","updated":"2020-03-26 10:48:04.000000000","submitted":"2020-03-25 16:56:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714403-1585155377977-28344ecc","meta_rev_id":"067f4049528277916357cc2fc798b34ca95b9f33","_number":714403,"virtual_id_number":714403,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 16:56:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2020-03-25 07:48:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2020-03-25 14:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-25 15:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-25 15:19:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 14:46:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-25 15:19:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-25 16:56:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc75b43dfae4a684b52c689e5fb2581ea9c8c020","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-23 09:30:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9e6a726a093cc2f969c6d00f43ffd17bc614dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 11:25:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8cf5eb1441d4a1286199eac3b6b54a4 : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/01c9fc9210e241f49cdf9bdfe363c238 : SUCCESS in 1h 03m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/353e8641482b4253ba556fc16c1cab41 : SUCCESS in 1h 38m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c8157f24bb08413f9099cbe0eb4e699d : SUCCESS in 1h 38m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65c892aa76614f49b060a725c380f2aa : SUCCESS in 1h 22m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d2b2173af56142a69b8fe01d2bb2ed88 : FAILURE in 12m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/822eee5f52d2413c834696da38ff9a89 : SUCCESS in 1h 53m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35c6335b243543c89fbb839d93e15a66 : SUCCESS in 1h 30m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4a493896726d4b8d9bced55224b80ae3 : FAILURE in 33m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/791ab7af37cb4dfa8e46801e6416118b : SUCCESS in 1h 04m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd2798713f74feaa24d695c4b8ec3b4 : SUCCESS in 5m 24s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb604738d244d4795e7af4c9f010d3d : SUCCESS in 48m 45s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d76d6e129f347d187395b6d05a1d121 : SUCCESS in 53m 19s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/75f53e484c6e4e97913a5e3be178b485 : FAILURE in 34m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb6707eafa8b41879b4c4dee4a728f70 : FAILURE in 1h 10m 33s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81264caadb2a4428aa517a585e899010 : SUCCESS in 51m 47s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c9ef769496646638127310b390d7590 : SUCCESS in 55m 30s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45dac893c2b246d496306eab0c589d03 : SUCCESS in 57m 04s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2db52b35abb949809de5170e2684ace0 : SUCCESS in 48m 00s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc61cad8b55429cbf8927e470233e51 : SUCCESS in 58m 02s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26efe897a9e74d8dba66a547a63423a5 : SUCCESS in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"096187b4a1c758798c5d852d579fa8517bdf38aa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-24 19:12:23.000000000","message":"Patch Set 1:\n\nShouldn\u0027t it be added for other OS as well then?","accounts_in_message":[],"_revision_number":1},{"id":"3756b35396e042ed15c3e2fd7f56088dc210c59c","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-25 07:39:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b2812d6dfb9a8f1367cabea676cc3c9048b25da","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-25 07:48:49.000000000","message":"Patch Set 2: Code-Review+1\n\nAdded for all OS, I can\u0027t see it was actually removed here. I had a look at os_nova to see that it was removed and re-added there.","accounts_in_message":[],"_revision_number":2},{"id":"c17a744339495e41a7f676aad1e234286d52b621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 09:58:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9570998450104256ad0787e538e1c4f0 : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/af19dc8394f54ae48a03530b91c966b5 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0187afd487be4b04a2985545bbf746cf : SUCCESS in 1h 55m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/657ebf054d3844969d26861cccd8092e : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cdde7e0af814e1ea06d5398d432fad1 : SUCCESS in 1h 28m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5dfbf8d99fb947a9a940ca17397d59b6 : FAILURE in 10m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cf0e868900f44fe9404eeb2a5f33994 : SUCCESS in 2h 05m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ee1641932a2419995879b9b2fb6c0a0 : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a469386d469f449f9ea26d2a84109676 : FAILURE in 34m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0ec32d0b44e4fa6ada2db79c107bfc8 : SUCCESS in 1h 16m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4bf5ba7eae743118e776672f1aa73ff : SUCCESS in 5m 34s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59e60fd632b64aedbfb48e38402a3206 : SUCCESS in 43m 45s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bebebcaf33e4ce78f04a45418f520e6 : SUCCESS in 54m 31s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/a017b75583114af89820431d7a2ef0ec : FAILURE in 33m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a51c74519bb444d79d3e542553ccc619 : FAILURE in 1h 23m 31s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02ca4352c47347b0b4de859e9073612f : SUCCESS in 59m 09s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/371344e753754f8da0c8bebf4957e9ba : SUCCESS in 44m 51s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4eb8fcb6be1b4862b12c2f9e3eb912db : SUCCESS in 59m 49s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2cb8dc5b036483d96892ff34bbadded : SUCCESS in 56m 11s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e51cdf26ad0e46199e16f7cc68230bfc : SUCCESS in 1h 04m 17s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfd38768e2754afe95fc30c7b45c936d : SUCCESS in 1h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef839c485e5d2089c8452a8b7be4920e06967307","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-25 14:46:23.000000000","message":"Patch Set 2: Code-Review+2\n\nI also didn\u0027t catch problems with missing systemd-dev for neutron, but can believe you\u0027ve faced with such issue","accounts_in_message":[],"_revision_number":2},{"id":"a860d593f11dbb2100eacaca7bb68c90ba270595","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-25 15:19:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"055fcc6f88a18ef0921604879f5a8c83118620ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:19:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b23fd9a18b6fee7aa3428d077cac6c12064bef5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:56:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4ca70acedd735f57a69574f11d3762dc8ed48d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:56:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2fe0829bd4b74a9ead0b7bf07fc3f03a : SUCCESS in 8m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bdbcd7b754974280a62c0f363feaf2a1 : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f8bc0ef8bd654bdab2328c8bd151f219 : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4b3f560b79a04320b3cd5cccb1908f3f : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abfe12159c7f4606a94281467a0703e3 : SUCCESS in 1h 14m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a98412772ae4404a50efbdd7c1f63a6 : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/503b8de6212040a68223d252bbf30d81 : SUCCESS in 1h 06m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e97f2e37f041f7a86f79371aeea1aa : SUCCESS in 6m 06s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e9bcd3de0be4c82b5a507946f5d49bc : SUCCESS in 42m 55s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f3019c3bdaa4e6eaa4945950f2ff7a9 : SUCCESS in 44m 48s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/827a5cb02fc3402eb639a4a307e6186a : FAILURE in 1h 02m 17s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58d8bfd4a46c4780bee8e82ff1150723 : SUCCESS in 45m 08s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29f380f01d7d4cc0b7153e8bce05ff7b : SUCCESS in 45m 43s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/671e7270720a47ae99ea0481518110f7 : SUCCESS in 42m 13s","accounts_in_message":[],"_revision_number":2},{"id":"41c691bc2568f77b7874e4542aec7440cbed52ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:59:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb0279fef864b60a2baab99115c7cb3 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"067f4049528277916357cc2fc798b34ca95b9f33","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-26 10:48:04.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c4071b95b0d61f59fdee84d19f928aacf164efba","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcc7c496be7cbe4aca15450c90a1312f96c3b731","revisions":{"f28f81879294b294412a0d49a7c7093dce4e1ce4":{"kind":"REWORK","_number":1,"created":"2020-03-23 09:30:41.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/03/714403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/714403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/1"}}},"commit":{"parents":[{"commit":"c474abfc599c0f454b8b51952413b72fa0381d4b","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c474abfc599c0f454b8b51952413b72fa0381d4b"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-03-23 09:24:49.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-03-23 09:30:38.000000000","tz":60},"subject":"Re-add systemd-devel to neutron_devel_distro_packages for redhat","message":"Re-add systemd-devel to neutron_devel_distro_packages for redhat\n\nFrom upper-contraints we pull in systemd-python, which in turn\ncomplains about missing libsystemd or libsystemd-journal\n\nChange-Id: Ieaad930cf022e2d5ef4a6b580db16ad6454569e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f28f81879294b294412a0d49a7c7093dce4e1ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f28f81879294b294412a0d49a7c7093dce4e1ce4"}]},"branch":"refs/heads/master"},"bcc7c496be7cbe4aca15450c90a1312f96c3b731":{"kind":"REWORK","_number":2,"created":"2020-03-25 07:39:37.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/03/714403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/714403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/714403/2"}}},"commit":{"parents":[{"commit":"c474abfc599c0f454b8b51952413b72fa0381d4b","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c474abfc599c0f454b8b51952413b72fa0381d4b"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-03-23 09:24:49.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-03-25 07:36:34.000000000","tz":60},"subject":"Add systemd-devel to neutron_devel_distro_packages","message":"Add systemd-devel to neutron_devel_distro_packages\n\nFrom upper-contraints we pull in systemd-python, which in turn\ncomplains about missing libsystemd or libsystemd-journal\n\nChange-Id: Ieaad930cf022e2d5ef4a6b580db16ad6454569e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bcc7c496be7cbe4aca15450c90a1312f96c3b731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bcc7c496be7cbe4aca15450c90a1312f96c3b731"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
