)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~730233","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I024f71c04c5cc1c6d3d747dd1843b6180c5cb994","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I024f71c04c5cc1c6d3d747dd1843b6180c5cb994","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-22 07:07:28.000000000","updated":"2020-05-22 15:51:35.000000000","submitted":"2020-05-22 15:48:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730233-1590162501916-a134e064","meta_rev_id":"1fb514e11b8e56b770fa18373149e57dcfccc360","_number":730233,"virtual_id_number":730233,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 15:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-22 13:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-22 13:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 13:53:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-22 15:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42838896caa972bd87676940adcb60b0a6c690a5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-22 07:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"442c31523e8e88819c0b72020d4a42304ae5aa5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:55:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f9d5772381f440f8af791db2ab97e92b : SUCCESS in 11m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/57ff5dafa5524a8db560435e1bb9d3cf : SUCCESS in 1h 14m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b419a41e4084c20af5164948134114b : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e80c347a0ea469eafce859a6412c792 : SUCCESS in 1h 18m 30s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/32e6c8e07d024813a04cfeb24485890c : SUCCESS in 2h 34m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/300e834a501840638a686ead04086669 : SUCCESS in 1h 49m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ee54c528c134943a6ce4d4627dc210d : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b401c6cd372740b0b1e10868e3d71964 : SUCCESS in 1h 04m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f9e057dd1a49969a47eed69208b0ad : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/418e4755807b4c5898c54c411d4abf14 : SUCCESS in 7m 33s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/512b32eeca504d27a2fa6352596a3ce1 : SUCCESS in 51m 47s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbfe9b85590946ab8c397bb9accb4210 : SUCCESS in 44m 58s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/aec766301412491894b386feb717dfef : FAILURE in 36m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd5ae51f4978453b883809ee10ffdf93 : FAILURE in 1h 08m 09s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c17da1948064c5a9836ec4d3b52df8d : SUCCESS in 47m 18s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f7e522aa7b0404497b370c1eb3ff289 : SUCCESS in 50m 52s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08375963a1e7408391ba028daedd3ed6 : SUCCESS in 1h 04m 03s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5829e8ab8874e8388b1309e6dcb5747 : SUCCESS in 44m 04s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1481f8b770574c538c2d402cae033a9e : SUCCESS in 54m 01s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4c8901c5c7f43d2876976bf6af8bf11 : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b9d5780dd09a15e827fca554c768948af2fcf79","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-22 13:53:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7739712cd194770745fd101a1e2073cf9d9cf6fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:54:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d006dee60ab9e51b89a62cc00f4a9a01914be648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:48:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b5f957cf97e2484e8fc795ec06681717 : SUCCESS in 9m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/998a7feffea14f01acd5abab2495a9a6 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/84b3045156914416b84cc1113762929e : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d077732a0b15428f92896e15ad0a8055 : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fae133f58de04cf7b300d7789c869517 : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92ffbd5c37ae4bd6b7e7457ab7f294e8 : SUCCESS in 1h 05m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c65807bd0604aa0ab5a971bfd2576c0 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/611c5533b6f44761a234185bdf2c6679 : SUCCESS in 4m 35s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcb4949b97104864952e4d48fc3eb522 : SUCCESS in 50m 15s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a1be4f2f1fc45539d0bb3199cb3125d : SUCCESS in 46m 49s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8437dada03e6421aafae762b6ceb857f : FAILURE in 1h 17m 48s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c428ce9d314c4f12b7c1b5a4df86b2a8 : SUCCESS in 58m 23s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe056063ca7449a68f0bf4dd61adb800 : SUCCESS in 46m 23s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88507608891c478ca684355baf544da8 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b6c319225945bbe79b6632b5871c7ee5a27e8c0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:48:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5005d834f92ba4075c6cc09c600383856cf8744b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d6c72d639d4fdf996df9932416aae4 : SUCCESS in 2m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2162fa8607234bbda70546ba20ae13a9 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4933a37696b907b4a18100c9591064988f2b9762","revisions":{"4933a37696b907b4a18100c9591064988f2b9762":{"kind":"REWORK","_number":1,"created":"2020-05-22 07:07:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/730233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/33/730233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/33/730233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/33/730233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/33/730233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/33/730233/1"}}},"commit":{"parents":[{"commit":"2a42d4533cd2dd265529e668d70df0c796669ca5","subject":"Merge \"config: add region_name to nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2a42d4533cd2dd265529e668d70df0c796669ca5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-22 07:07:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-22 07:07:24.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I024f71c04c5cc1c6d3d747dd1843b6180c5cb994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4933a37696b907b4a18100c9591064988f2b9762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4933a37696b907b4a18100c9591064988f2b9762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
