)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~730384","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6","subject":"Remove neutron from tempest_plugins","status":"MERGED","created":"2020-05-22 18:00:19.000000000","updated":"2020-05-26 08:13:46.000000000","submitted":"2020-05-26 08:12:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730384-1590480729131-7d9b3224","meta_rev_id":"3a93b70db8e081c42239a3d5861cc9abb5a8b107","_number":730384,"virtual_id_number":730384,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 08:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-05-26 06:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-22 19:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-05-26 06:35:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 19:16:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-26 06:35:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-26 08:12:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0218c60d01d7b4155b4a910a347d08e2b9835798","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-22 18:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9c539a4f9c8e8fa693c196ee5479fac7b8974c7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-22 19:16:39.000000000","message":"Patch Set 1: Code-Review+2\n\nfunctional tests look to be passing so this is a good simplification","accounts_in_message":[],"_revision_number":1},{"id":"594f4fc4d855e6c4af35cd3159f0560ef98e40d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:25:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5f7590d5474941dfbf2f1d0d872d42d1 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ba51eb178794e768b667719011b5112 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/662589ad4b9f4a5fa9703290c3110260 : SUCCESS in 1h 27m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/baba59ac8be843a1824421deae321e01 : SUCCESS in 1h 41m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/58f5e6d5d3ff43c89aa899dd5888d104 : SUCCESS in 2h 23m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92b01d5ef76a4669ac80ed76eeb97fd0 : SUCCESS in 1h 51m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ae4423b568740068aff50bf7a832196 : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0c0b8e6cd9f41ef9acffef4fd43f558 : SUCCESS in 55m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0be516baff4f24b0c1d6e172481421 : SUCCESS in 6m 58s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c586ce0b8a1b4d6da5e7a286d4c80d8d : SUCCESS in 39m 49s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec0d3c4874a84f5dbd52982aa43b4ed3 : SUCCESS in 43m 06s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/e939f80ed64049d180bb4e6a39f0536d : FAILURE in 29m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d05c4141d0b4b6ebf35d1e31af239db : FAILURE in 1h 23m 13s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d770384708284f798fcce00f8a920792 : SUCCESS in 46m 04s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f165c1af285546f79c102d37d3609c13 : SUCCESS in 53m 35s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afed4f25096846aa8b95bd3a61f108f7 : SUCCESS in 53m 43s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0a57c583ee54481a248fd31055f7049 : SUCCESS in 45m 23s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfcb20ee48eb4bf7b956c46ffac5e957 : SUCCESS in 48m 14s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27dc8b547c7946d79ac2eb42826f5889 : SUCCESS in 42m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e3517e0b2929c35a5dc0b4793d483fd85c1e79d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 06:35:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3186aef9aee459d7963af125ff3cdb25e3d6e4ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 06:35:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e9bac781839d452dc55d573a7f5569362baab56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:12:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db0196152f524e8da3ba8310d5468e86 : SUCCESS in 11m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a93c73ec78184d23aed8b75f06ae833b : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9758af4b375d4c308a18611712f50301 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb11cfd3845c4815ad89a65fe23d76be : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea7dde8a474b4b8fb74d9564af4b42fa : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/242d2c6e7d5d4209a1e36c7ae07aef0e : SUCCESS in 1h 01m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433fced2d0674109a5ce7a79273b778f : SUCCESS in 8m 50s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3726a8bef9df46a6b0d5a8477b1702f7 : SUCCESS in 45m 26s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e66d041a215b49b4be6ceb1c8a9397ea : SUCCESS in 38m 27s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb9ee985bf441d89926a3e1468f50a2 : FAILURE in 1h 17m 22s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc2132c3fa1743ff82b01ef59d297c0e : SUCCESS in 1h 13m 58s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bffd46b37a704386b6895a24c9449f9f : SUCCESS in 59m 23s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34e2ddd968f24a39a23de031fd0da7c4 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1ded1f375262d961daa2b12492205d0fb7dc577a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:12:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a93b70db8e081c42239a3d5861cc9abb5a8b107","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:13:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2e5e9301bf44dab59797e2c9bdb866 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08071049f0b123f0c9a967b1169b23fd5762a042","revisions":{"08071049f0b123f0c9a967b1169b23fd5762a042":{"kind":"REWORK","_number":1,"created":"2020-05-22 18:00:19.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/84/730384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/84/730384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/730384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/730384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/730384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/730384/1"}}},"commit":{"parents":[{"commit":"5a66c004f5db32b761dfe8c8a50f11e5692e93f6","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5a66c004f5db32b761dfe8c8a50f11e5692e93f6"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-05-22 17:58:06.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-05-22 18:00:19.000000000","tz":0},"subject":"Remove neutron from tempest_plugins","message":"Remove neutron from tempest_plugins\n\nPulling the neutron repo into tempest is unnecessary since tests\nare run using the neutron-tempest-plugin or repo-specific plugins.\nThis patch removes neutron from tempest_plugins.\n\nChange-Id: I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/08071049f0b123f0c9a967b1169b23fd5762a042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/08071049f0b123f0c9a967b1169b23fd5762a042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
