)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~730730","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Ftrain~I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6","project":"openstack/openstack-ansible-os_neutron","branch":"stable/train","hashtags":[],"change_id":"I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6","subject":"Remove neutron from tempest_plugins","status":"MERGED","created":"2020-05-26 08:35:01.000000000","updated":"2020-05-26 18:37:03.000000000","submitted":"2020-05-26 18:35:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730730-1590518127136-48d602ed","meta_rev_id":"a82b99c895a8d70b2537550bec618900f22c5014","_number":730730,"virtual_id_number":730730,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 18:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-05-26 16:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-05-26 10:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-05-26 16:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 10:50:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-26 16:38:37.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-05-26 18:35:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7e6d7ed825acc9cf56b0dd2844c24b952ff4706","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-26 08:35:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3f8cf5e38d70b774cdb4486a2b8b11ce50da762","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 10:50:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"805545eca6cb1b1b5597f1dae1a8ead97c9ebda3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 11:15:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0366adf912b4a9c8ef89557ae1c8388 : SUCCESS in 6m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9e865168bbe74005aadb5f1f7e444189 : SUCCESS in 1h 02m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/28a387beefd344eb8eb0c6198a839d3a : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/228b6feaf4ec4d72b6d02e8f4d7a0eef : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/abe33e6aabbf4d2f9b2a3e2c35576786 : SUCCESS in 1h 11m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17ab472c8dbc4d768757d5b7782095f9 : SUCCESS in 1h 08m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ca50b2c0ed741f9823001744cf8d74b : SUCCESS in 2h 34m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21c1667e01644718826ad0e66b68cb70 : SUCCESS in 1h 52m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b26e05cab624163ac8843510b291a04 : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f2385ee90fa0427fa4f5a8f61500686c : FAILURE in 35m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0140fa342b954cf79b9fccc6b119ccad : SUCCESS in 58m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8ef28dafc64b41b24a0aeb0f64184c : SUCCESS in 7m 04s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56fbf41b53c54f8a9e5f490ee0fd044f : SUCCESS in 46m 53s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a21f301b20aa4648a701d93e79b3d936 : SUCCESS in 39m 08s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/c66aa3271d3b404993282b8f2bd0d132 : FAILURE in 32m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/978b0ea166c74292971a808107bac151 : FAILURE in 55m 47s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/618d2c5bb3cd4c28aa052eb55ac8de06 : SUCCESS in 45m 21s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84beee204cd543788fa7a2c8ba3f2018 : SUCCESS in 44m 42s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/149d1c0130664ce08c0145883050a056 : SUCCESS in 46m 14s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/571ee45e5500424ca825088c87cfb022 : SUCCESS in 39m 35s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b50b1f2a9c544cbab4a928e277100bc5 : SUCCESS in 45m 37s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69a13d12255f49ec8c1e3dac2f6b5aef : SUCCESS in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b60787c89420b772ed6444614f1eaecdfc1980a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-05-26 16:38:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b03c00324355d6c2d529cf25b20d6dd75cc492b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c2f1330389cefc46ee382cb8c80f0384e8d1feb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:35:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f7dbea6e1bf04b13a41d532de3d5f8cd : SUCCESS in 9m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f0eec3059868433db34fb1aeed247304 : SUCCESS in 1h 07m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c8aa716ba0604644a1e57e79a78ea33c : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/feb2de7ba0fe44cba0bb35b936ed353b : SUCCESS in 1h 39m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4053b3a54f4843a1b06776260068feb7 : SUCCESS in 1h 32m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f53daa5f534fd2bc52e9e7667983b2 : SUCCESS in 1h 14m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e30ca8f3c1f4e50a5f80a207d8eab8f : SUCCESS in 1h 48m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a510274141b44ca38245851d23d33166 : SUCCESS in 1h 12m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fbf1fb78f324c539fe5a3dd0b8af58f : SUCCESS in 59m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a405e4219c450d99cb79c5f88f0898 : SUCCESS in 11m 08s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d8104852d794a1f8dd1f005790e068d : SUCCESS in 42m 20s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61fa88bd59e54f27b132b63e1a34777c : SUCCESS in 44m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/461f2058386d4342b1e19a3a3ef5d606 : FAILURE in 1h 05m 43s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be4b5d90fff9446a8822878db52b8adc : SUCCESS in 52m 34s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4b11535ecbd4980a7b6774a3ff1d7be : SUCCESS in 50m 27s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bd3f2a0192a4fbf9858570e766d9d5d : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"066e6684dde49ea9e15949ac876994003d11909f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:35:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a82b99c895a8d70b2537550bec618900f22c5014","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:37:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278a6569978f4de2bf38b4af884c33c2 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86008b67b975799a694df0235cdbd576716a783c","revisions":{"86008b67b975799a694df0235cdbd576716a783c":{"kind":"REWORK","_number":1,"created":"2020-05-26 08:35:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/30/730730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/30/730730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/30/730730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/30/730730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/30/730730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/30/730730/1"}}},"commit":{"parents":[{"commit":"b35e1994ddd01820d25343d4b29df787ffc99f0e","subject":"Add extra rootwrap filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b35e1994ddd01820d25343d4b29df787ffc99f0e"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2020-05-22 17:58:06.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-26 08:34:39.000000000","tz":60},"subject":"Remove neutron from tempest_plugins","message":"Remove neutron from tempest_plugins\n\nPulling the neutron repo into tempest is unnecessary since tests\nare run using the neutron-tempest-plugin or repo-specific plugins.\nThis patch removes neutron from tempest_plugins.\n\nChange-Id: I6176c17b7261fecbca82452b7d4ad2a2c5cdeeb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/86008b67b975799a694df0235cdbd576716a783c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/86008b67b975799a694df0235cdbd576716a783c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
