)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~733849","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fussuri~I69a4e5deed4b21e70afd7dac747375b81e98cb22","project":"openstack/openstack-ansible-os_neutron","branch":"stable/ussuri","hashtags":[],"change_id":"I69a4e5deed4b21e70afd7dac747375b81e98cb22","subject":"Update OVN metadata agent binary name","status":"MERGED","created":"2020-06-05 14:05:13.000000000","updated":"2020-06-09 22:42:23.000000000","submitted":"2020-06-09 22:40:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733849-1591742455355-d691af0f","meta_rev_id":"f3e524fc369e7a75ab1cce5ccb69c5e111df7ce9","_number":733849,"virtual_id_number":733849,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-09 18:27:52.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 22:40:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-09 10:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 18:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-06-09 10:24:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 18:53:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-09 10:24:12.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-06-09 22:40:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d29eeefccec15c4e816677a99ef697573348a8a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-05 14:05:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f46c5f6e255b83c66ec7110436e43cbe075f66c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:50:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/580d31d38d9d49998017bf60bfbc26f2 : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/895a8bbaf2f74e7b8d81589c6e531795 : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a96f928e82047848d2cd2d78d328c9f : SUCCESS in 1h 37m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de945507b46242d7b5b5fdb0008c3c8d : SUCCESS in 1h 14m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5c00a06a317644dd83bbe0198158f20f : SUCCESS in 2h 37m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36c8be3d84f04c70b0c203a5a7154d2d : SUCCESS in 1h 54m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/88aa213501ea4fbeb2052ec8945428c0 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14014a80fdff4ebfbf9535d98dad800a : SUCCESS in 1h 03m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4995791f66774ef1831288e54f34a81a : SUCCESS in 6m 09s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a401881bb9ec4e558d3215643e6691e4 : FAILURE in 28m 15s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e76205aae0e542beafc1be62e0e5fff9 : FAILURE in 25m 36s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/d17225454a604fba983e5b5d3f8f6684 : FAILURE in 32m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/522b8307c48a40539fb1372e969e3010 : FAILURE in 1h 17m 02s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e2e2979873f4799a59ed5147167074a : FAILURE in 25m 16s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8a12c8fed0640db89b1c184d37ae404 : FAILURE in 21m 36s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c799979f2224949ab4660cd383be4aa : FAILURE in 22m 03s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/801d094459634e2fb76666929977a1bc : FAILURE in 25m 03s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95e2490d1a1f413cba23c0b1d7483ee5 : FAILURE in 24m 46s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/216144c574f748c58d591aa64d583a7b : FAILURE in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f55ab0389f6a443ed690bc39333c10fbef14d67","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-06-08 13:51:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ecff1bc9d20340ffa2ffee0262dc9e5fc732298","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 15:30:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11dcca4e6a24808502caa62c584499bcdfe6a329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:23:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/82fce4f4949e4e2a937d977ba0ca1fba : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5d8a26f4ccc84450a023bc278789ea3b : SUCCESS in 1h 06m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a8f175d5cbe49ad82adf6029bc70c28 : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e365d8786084c5893c145320e5b3d5b : SUCCESS in 1h 11m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a3885f29e474eb59d0137863c4d0942 : SUCCESS in 1h 41m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/313306e025b8401eaa9f77a775332a31 : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d31a12bb5e014573939a9c3832485d3e : SUCCESS in 1h 06m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/504a52b865d34726b7760aeda62cc0f6 : SUCCESS in 57m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85a6b33f49a43368473abe1bb3b384d : SUCCESS in 8m 26s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52bf6253407e41b9bf620b934a29eef8 : SUCCESS in 55m 14s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17b282c29c984db3a23d63a898d01a8f : SUCCESS in 59m 03s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/a40298bca6c744d49fc7209ae460049a : FAILURE in 39m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef2fc9a7f38845ca88f2b2db628e8139 : FAILURE in 1h 10m 16s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cd62df7ceb94706a1c2f3678a21a10c : SUCCESS in 54m 20s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5574e1ab97b4189bf08b8a9f6cba15e : SUCCESS in 1h 04m 38s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22dabdd024ed430f80f1ba94fdf6b938 : SUCCESS in 47m 59s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b93d662041a4870af66c6698daf85bd : SUCCESS in 53m 10s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6f99e65138a461c9ce084311db7bf5d : SUCCESS in 59m 27s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f32d15fe448e4c7ea50206f0b31f98a0 : SUCCESS in 1h 00m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f799a8ed3310935d0b15d6e763011469e05f880","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 18:53:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03ab7f2a1b64f7237d34b4b359081b38b3fd68e4","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-06-09 10:24:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c05281d85ca500cd65faa790cee8edb60e91f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:24:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a82c7f41baf6000e9cebf7a188ac2e3d59b4723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 12:02:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eedc0cca762449daa40cd467188019f0 : SUCCESS in 12m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b89ad5fda7ac45ddaba9a020046dfce8 : FAILURE in 55m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d31dad92c8e8449392aa137a863c3a77 : SUCCESS in 1h 33m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e38906f5a8b040758997fbc3dc3d139c : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/50313a462ab840bd9648e03e69d64c2b : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51f6f4dd4a434ec8b3073a88a408599f : SUCCESS in 1h 03m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84bf34ce4ad43fba434eff87693496b : SUCCESS in 7m 41s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e653fe89c0b45f6b8c16dfb9ac96b5d : SUCCESS in 45m 59s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e5c93d556d423a8cc25fa3b0615b02 : SUCCESS in 42m 35s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/974977c24a8f46c7a36ce11c7ca0b6b4 : FAILURE in 1h 04m 05s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b29a503d79c2428e9351b9c99a45af1d : SUCCESS in 48m 55s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ba782125fa24ffc8c2add3ff7d24b7d : SUCCESS in 46m 03s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58f8c15eeb914301ab776c844e8d396f : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":2},{"id":"64741dbd4c8d71e9564438d9fe5edc75e61eb885","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-09 15:02:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bdce85a6264f8ff2087e192c073638f0a93f4e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:02:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1af404081e0e48248f4e8c5bcb22ca8d : SUCCESS in 10m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/10260938209b4d218d8439043b710357 : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5e367bc3282400fb6ae48db87c076cd : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a09404373fe4a519fb7fe2fed5b0ba3 : FAILURE in 30m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a2ac06b42d94ecaafc51bb9f8ff3c4b : SUCCESS in 2h 48m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d06e887eec5c47468801520f9fe02650 : FAILURE in 24m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f64b25dc1c734957b520b7647a10fb60 : SUCCESS in 1h 07m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59896852a05a448da8f20baba7b20ffb : SUCCESS in 1h 09m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8263e46a5654447489eaf5a37efef7bc : SUCCESS in 8m 28s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b57e528f222b496c9d107adba9edecaa : SUCCESS in 59m 27s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/467affd710c14deb94e88b2f54186749 : SUCCESS in 45m 16s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/db51a35e002742bea1b5b752db9607ee : FAILURE in 35m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6409b9d997ee47739ab7bf7e1eb8258f : FAILURE in 56m 15s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e81bb699a1844472a5fddd8e76f23d4f : SUCCESS in 47m 11s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d27918a56d84bfbad935a4dd2cb3083 : SUCCESS in 50m 16s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10aeb18f247142f99270a0c87fe51bce : SUCCESS in 49m 53s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9aea96225c448daaf4fc1143a3924de : SUCCESS in 59m 49s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40b5f7d590d949ab8a5d5eeb693b4e4d : SUCCESS in 1h 05m 24s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fd474b69a804b5a8438a076ddf8d824 : SUCCESS in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe18082e1f79d8a5f0097163ae2d1b2fc017bb6a","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-09 18:27:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"70b30702ea4ed2ec78d8b1196bf849bc60b5e7f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 21:01:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa0cccee27d849ff9275bb4805c2c930 : SUCCESS in 11m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/116d0639ad8c444e9a1ad398a0a5c14d : SUCCESS in 1h 12m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6ab66ca2b194dcc809577fb1c0591ef : SUCCESS in 1h 39m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43648974ff8c4d3d8a3806a441d873c7 : SUCCESS in 1h 13m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/70b5d12eb2324ced835e354e929ccb37 : SUCCESS in 2h 24m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb60c0177eeb43bca6552953c3af6a14 : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d7c6dd2068c439280bf47c5a582d30f : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95e5d84f469a4ab98a2d3368f06b493a : SUCCESS in 1h 03m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22cec9bc38847579312893d49e70aad : SUCCESS in 8m 46s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14f703e90d574cbf9d8e75211f142b40 : SUCCESS in 42m 10s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa23514d64e342d5b0c5dfc5f6263449 : SUCCESS in 43m 27s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/c6884261779c43b9bc9ab1bce13a5ed7 : FAILURE in 29m 31s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06761c2f6645413ea98ed0662cf569e8 : FAILURE in 1h 10m 34s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e1e4635e66f427899a08284e542ec8a : SUCCESS in 47m 21s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/055c257ddace48d1aea0e6d9fa2c03fb : SUCCESS in 47m 35s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9ce3bffd0ea45ca84369f8bf18f225a : SUCCESS in 44m 59s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/feafc479394b4dc3a1f2b203c325b38c : SUCCESS in 42m 21s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2308b75eaa574fdab3a5bbe4276a8924 : SUCCESS in 50m 05s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8ed0ab4c2904bb3b2b3f794d82fb36f : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"490545758aac57c09782ae12e4d94c7cce0d2406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 21:04:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ec7afb099a248366b321371ad1c23edc7780602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 22:40:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a3367dff71377cd0c45bfb17db5034f1ea79525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 22:40:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/600ef915daa04b42824f87723de37b64 : SUCCESS in 7m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/537959597ffa4938bcc1f01a76265087 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/489b6a74071546838910d3c2e4f97b64 : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3cf86416d934558bcf803709743b253 : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01a9ed85d0484ee8b5f9a8f0ffa2f53d : SUCCESS in 1h 12m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/031c07b91df44cdfbca33ec5a94fc3eb : SUCCESS in 57m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1539462951b44518e77434900070ca0 : SUCCESS in 6m 38s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f151d02cf43c495595b8b966b0672123 : SUCCESS in 41m 56s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc03360af3ee49d293ff1535401afd6d : SUCCESS in 40m 48s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c99d4403fcd34a09afed320574cf9f14 : FAILURE in 51m 35s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8eb2fcfd1a574609bd03568087a670d1 : SUCCESS in 44m 01s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a243c9f9c5114ed6ae5a0eb82e077e76 : SUCCESS in 52m 22s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/763968c7524346969526289e3bbdb0a6 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f3e524fc369e7a75ab1cce5ccb69c5e111df7ce9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 22:42:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf6eba4e69b480b89439ec1c0888f65 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"672ae0e31908effc349a0a7d570c6a63024b86e4","revisions":{"5c876ae1f2838df15a9fef8bf20bef9bdd93f027":{"kind":"REWORK","_number":1,"created":"2020-06-05 14:05:13.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/49/733849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/733849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/1"}}},"commit":{"parents":[{"commit":"4f302847f1152b1826ddbe1c2032d8593c5a6bda","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4f302847f1152b1826ddbe1c2032d8593c5a6bda"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-06-03 15:14:10.000000000","tz":-300},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2020-06-05 14:05:13.000000000","tz":0},"subject":"Update OVN metadata agent binary name","message":"Update OVN metadata agent binary name\n\nDuring the transition from networking-ovn to the neutron tree, the name\nof the metadata agent binary changed from networking-ovn-metadata-agent to\nneutron-ovn-metadata-agent. This patch updates the service definition to\nallow the metadata agent to function.\n\nChange-Id: I69a4e5deed4b21e70afd7dac747375b81e98cb22\n(cherry picked from commit fccf1b75ced3a21e2b165a6e5fd626c3a0d6066b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5c876ae1f2838df15a9fef8bf20bef9bdd93f027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5c876ae1f2838df15a9fef8bf20bef9bdd93f027"}]},"branch":"refs/heads/stable/ussuri"},"672ae0e31908effc349a0a7d570c6a63024b86e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-08 15:30:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/49/733849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/733849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/733849/2"}}},"commit":{"parents":[{"commit":"4f302847f1152b1826ddbe1c2032d8593c5a6bda","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4f302847f1152b1826ddbe1c2032d8593c5a6bda"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2020-06-03 15:14:10.000000000","tz":-300},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-08 15:30:16.000000000","tz":0},"subject":"Update OVN metadata agent binary name","message":"Update OVN metadata agent binary name\n\nDuring the transition from networking-ovn to the neutron tree, the name\nof the metadata agent binary changed from networking-ovn-metadata-agent to\nneutron-ovn-metadata-agent. This patch updates the service definition to\nallow the metadata agent to function.\n\nDepends-On: https://review.opendev.org/734129\nChange-Id: I69a4e5deed4b21e70afd7dac747375b81e98cb22\n(cherry picked from commit fccf1b75ced3a21e2b165a6e5fd626c3a0d6066b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/672ae0e31908effc349a0a7d570c6a63024b86e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/672ae0e31908effc349a0a7d570c6a63024b86e4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
