)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~746496","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Iec7f1314a2e4fa4256d9b24d3ea927bd5479e7a1","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"Iec7f1314a2e4fa4256d9b24d3ea927bd5479e7a1","subject":"Make has_offline_migrations output predictable","status":"MERGED","created":"2020-08-17 13:07:00.000000000","updated":"2020-08-18 14:31:19.000000000","submitted":"2020-08-18 14:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746496-1597760978884-ea02c699","meta_rev_id":"4f377bd87d9efededf506b69b40b9ef69f7b2037","_number":746496,"virtual_id_number":746496,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 14:29:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-08-18 12:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-18 12:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-08-18 12:45:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 12:29:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-18 12:45:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-08-18 14:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77ffb2aa257d27ef36f9b7379694c97d0c78d531","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-17 13:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df633fccb10042a8015c332a9f82a9ba73f4bb09","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-17 13:07:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8e1900b4efc940d3508cb03992f671f98c9faebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:25:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/75e9cdee3b6045b7860c93d0a6656a88 : SUCCESS in 7m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/246930cd8130477da342cbf990874372 : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76a9e3d83f64485ea6547ee0b9a603a8 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1db2003b06c14d2db4c4f54afb4fb1b1 : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e21a7382fc14202becf423f5b38535b : SUCCESS in 1h 09m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad76bdc83fcf49e5bc00f7c6e4b75094 : SUCCESS in 57m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fee2aa81e4044460b15ab2a29a90e370 : SUCCESS in 2h 09m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60b448c07cfa4b48adfdf08fcc64e7eb : SUCCESS in 1h 47m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0280a93422b7423b9262aec2c4296ca7 : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a34b372a19f407686b63daee03e8bc7 : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41dca4a8803a4864a2cc2fc4cc371e61 : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff00040857b747b286548573a6c5b309 : SUCCESS in 1h 07m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbbd1f23c57e47cdb9ca74b867997701 : SUCCESS in 7m 08s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0625e3b5280a45f0b5799aac8fb03712 : SUCCESS in 43m 40s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb4fc784e0f4014bc38cb4e5c3295b0 : SUCCESS in 41m 04s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/5cc5873bac0f403199c463e243f873c8 : FAILURE in 26m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1cec102a6584dff83f0eab9f4f9ad02 : FAILURE in 1h 16m 34s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65bd391ad50b46e08f210981f91741c2 : SUCCESS in 42m 12s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c8d1b0525cd44388f780d8dfb63c48f : SUCCESS in 44m 26s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/977bf3581c4346589c53e3f55360ec85 : SUCCESS in 46m 16s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96bdcb286274871b387f5cd2ee91fb5 : SUCCESS in 39m 36s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/474f5633ca074459855d093b3ead79fe : SUCCESS in 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2848e4511d76211a963313d6c429efcb6770e81a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-18 12:29:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0de0bea2b5b20dced6c49123203278fb8d8b6d09","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-08-18 12:45:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"37a44bf34d2640cead5de7a2f8005c34b97ea98c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 12:46:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12d0e4627ac15a0a8b21d7287c870786b37bbcf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:29:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/32b98ce6b88640749ed6e0c94e59e9ea : SUCCESS in 8m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d9dedbc8b788413db52e6ec11da64c63 : SUCCESS in 1h 06m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b583d37a01d649da96a805813029e1da : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/395347c491c340b6b3f8787640de30d8 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe3f8ee553c74277a3cdee6d63519f91 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32de520efbdd4c5790718c0686869151 : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/973cb6eea31d44c0815dcf43226d0886 : SUCCESS in 1h 02m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/93d9aa99386d4d87b6721a51e41ad043 : SUCCESS in 1h 01m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e07146d1cf45a88cccd32803ad55d1 : SUCCESS in 59m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f3ca0d89a2c40da9a8dea7a215d6c92 : SUCCESS in 58m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3aff0106f024afc9b7cd7a156bd7480 : SUCCESS in 5m 47s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7c5304cd1d9418382601f7d534a20b6 : SUCCESS in 41m 46s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb4674623e2249caa54a0dcffcb0e6e3 : SUCCESS in 54m 16s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01b43857efad4d88a6441e714c95251e : FAILURE in 1h 11m 26s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef672fcf6da746268419158ee8fcc719 : SUCCESS in 47m 53s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ad597f2a63541dd96c03aeaf4d2cab9 : SUCCESS in 43m 26s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f2713ac80844802a615b0f8216c95a0 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"cc8a40f2db4c4a563e9c7f1d46d8bebee7838b2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1f3a118748c31f7009e09f4962abe3952313b6ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-18 14:30:54.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 928b204399a9642225d3a935eb2212dfca8dde07","accounts_in_message":[],"_revision_number":2},{"id":"4f377bd87d9efededf506b69b40b9ef69f7b2037","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 14:31:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f599de62df334f45be073fd81b6a5418 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"124dc769190bea27ca35e57d23615ba928a6af20","revisions":{"19c3960c36ed20941e4a47ae8e5aecd2ed685c67":{"kind":"REWORK","_number":1,"created":"2020-08-17 13:07:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/746496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/96/746496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/1"}}},"commit":{"parents":[{"commit":"6390c3b0ef6ad0ac8490d6afea62b7d26af1b38f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6390c3b0ef6ad0ac8490d6afea62b7d26af1b38f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-17 13:05:36.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-17 13:06:45.000000000","tz":180},"subject":"Make has_offline_migrations output predictable","message":"Make has_offline_migrations output predictable\n\nSince we parse neutron-db-manage has_offline_migrations for specific\nphrase in the output, we need to ensure that locale is correct.\nAs in case users locale is passed to the server (with SSH), this will\ncause task failure.\n\nChange-Id: Iec7f1314a2e4fa4256d9b24d3ea927bd5479e7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/19c3960c36ed20941e4a47ae8e5aecd2ed685c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/19c3960c36ed20941e4a47ae8e5aecd2ed685c67"}]},"branch":"refs/heads/master"},"124dc769190bea27ca35e57d23615ba928a6af20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-17 13:07:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/746496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/96/746496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/96/746496/2"}}},"commit":{"parents":[{"commit":"289e844d29db1e16dbd6723ca06a0dc5d80e1dbc","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/289e844d29db1e16dbd6723ca06a0dc5d80e1dbc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-08-17 13:05:36.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-17 13:07:16.000000000","tz":0},"subject":"Make has_offline_migrations output predictable","message":"Make has_offline_migrations output predictable\n\nSince we parse neutron-db-manage has_offline_migrations for specific\nphrase in the output, we need to ensure that locale is correct.\nAs in case users locale is passed to the server (with SSH), this will\ncause task failure.\n\nChange-Id: Iec7f1314a2e4fa4256d9b24d3ea927bd5479e7a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/124dc769190bea27ca35e57d23615ba928a6af20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/124dc769190bea27ca35e57d23615ba928a6af20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
