)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~759914","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I01715b2b711f3ead8f84ece8b56b3e0cea44d51a","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I01715b2b711f3ead8f84ece8b56b3e0cea44d51a","subject":"Switch functional tests to focal","status":"ABANDONED","created":"2020-10-27 17:43:28.000000000","updated":"2021-02-03 18:27:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4441ec034d0f2d5a093439fd526414b63f7275c7","_number":759914,"virtual_id_number":759914,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-17 07:19:54.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-27 21:12:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-27 21:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9df8e1ddbf169c6fab76727d7883ed56d7c8586","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 17:43:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72859028bc89fc59c423dae5a4eea594b14b18b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 21:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/063ecc1342644ead9d2957917dab4779 : SUCCESS in 24m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f3b2fd1966064f6796c601444f83d328 : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12ef2d62abda457980e904fd0c5f9277 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4302a34606fd49bd9b01b0689b86c354 : TIMED_OUT in 3h 08m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c0da9c8e8d2433bb600e2544f5eae73 : SUCCESS in 1h 05m 26s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/150844ec6efa45b9818967bfdd44e833 : SUCCESS in 1h 51m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8de8f1c4230c430191fe708bf9eebd71 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a265acd3a0e944b89727826428fa742e : SUCCESS in 1h 00m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fc874c7e779441a8d9b58fd7092a572 : SUCCESS in 2h 08m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a1a4436f824286a046e93dd19d6263 : SUCCESS in 5m 18s\n- openstack-ansible-ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c6bf9a413c4701ba275f698655d4c1 : FAILURE in 31m 19s\n- openstack-ansible-ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25abaa0049be460688d5eec06e157e86 : TIMED_OUT in 1h 36m 10s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97007430a89f4276979c4de050e2b477 : FAILURE in 2h 16m 42s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29eebca78824f318a615b39f58a1f3c : FAILURE in 34m 00s\n- openstack-ansible-ovs-nsh-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58aa6c582a584b169d9bbc8d98872e4d : FAILURE in 32m 02s\n- openstack-ansible-odl-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac10fcf435348bcb0968fd4e5607f44 : FAILURE in 31m 23s\n- openstack-ansible-ovs-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62c6bbf9938144a09768b96fdee3e533 : FAILURE in 30m 03s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be04d5970c714e45b32f244ceb795add : FAILURE in 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53f14375e2d1dd0c0b20e886d7613224532ef44b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 07:19:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9fdae41294891dad5241f906fc4d7ae6a48c91b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:29:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b92ed3d4ecb48a3be2973e9932dbfb3 : SUCCESS in 9m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/89e1e90b73f141a0a0397a56ddce2aa8 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a4a2f50f4b0640ea9b464a1bdb6b81d1 : SUCCESS in 1h 25m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebc9f1fbcc304737ba295662a7d42400 : SUCCESS in 2h 04m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c71171e14a7740798b3da7a8008a7aaa : SUCCESS in 1h 24m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/05edba5f4509418a99c251922fa32428 : SUCCESS in 1h 53m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/180ffb3b10924690858a8b403c035f3d : SUCCESS in 1h 51m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24f3c7f6d52f43d8849080bde0872498 : SUCCESS in 1h 04m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7a451310d646388aa96dc433897504 : SUCCESS in 57m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9663092e4540fd907ff7d4f6b670cc : SUCCESS in 5m 54s\n- openstack-ansible-ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/441a854a655f425597015ee0190a1a0b : FAILURE in 38m 32s\n- openstack-ansible-ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e84f0c7741c04eb99f5bea27674755c8 : SUCCESS in 51m 11s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ae00940e8dc424e8fe57d9d1b678a12 : FAILURE in 56m 11s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39319653e4434012810ff33874429e09 : FAILURE in 38m 13s\n- openstack-ansible-ovs-nsh-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ec2bfe900e42f8bad6430384cd5a23 : FAILURE in 38m 21s\n- openstack-ansible-odl-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebe4fdfcd4c6478c821b4bd5aee0ad0f : FAILURE in 30m 55s\n- openstack-ansible-ovs-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb28d12e6c7c469cac048d6f2a2b7d82 : FAILURE in 30m 42s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7a80560d9d842e6a40e88c936f261af : FAILURE in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4441ec034d0f2d5a093439fd526414b63f7275c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-03 18:27:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"400c2e48590830a4a292e3921a26436a0ee6faab","revisions":{"400c2e48590830a4a292e3921a26436a0ee6faab":{"kind":"REWORK","_number":1,"created":"2020-10-27 17:43:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/759914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/14/759914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/759914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/759914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/759914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/14/759914/1"}}},"commit":{"parents":[{"commit":"2b655a857ca0d7666feb35a987bb666faa754968","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b655a857ca0d7666feb35a987bb666faa754968"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-27 17:43:15.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-27 17:43:15.000000000","tz":120},"subject":"Switch functional tests to focal","message":"Switch functional tests to focal\n\nChange-Id: I01715b2b711f3ead8f84ece8b56b3e0cea44d51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/400c2e48590830a4a292e3921a26436a0ee6faab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/400c2e48590830a4a292e3921a26436a0ee6faab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
