)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~760149","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I073da66f9e395bbb99b1c21701c808f252c3a6cb","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"I073da66f9e395bbb99b1c21701c808f252c3a6cb","subject":"Rename nova_metadata_* variables","status":"MERGED","created":"2020-10-28 14:27:03.000000000","updated":"2020-12-01 19:56:34.000000000","submitted":"2020-12-01 19:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760149","meta_rev_id":"4af46d39014cae47d6f774be510a942cac877a3a","_number":760149,"virtual_id_number":760149,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 19:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-12-01 19:56:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-01 19:56:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":1,"date":"2020-11-30 17:13:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 19:48:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-10 22:33:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-24 12:11:49.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"926e9d6bf738c825c07e08c79dc66fb8d6ab50ad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-28 14:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98bf8a4dcb510a04546bfb045bb2e769dffcc1de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 16:53:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e4a06f647ff4659a3a44da2bffde9f7 : SUCCESS in 9m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e95c13c3cf6b4ae8abb33ec224086082 : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7b82f09b54d748189fd29b3334072074 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db6fffd52c0d4587a7ee6a2983f18519 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0105ee3fa5734a2586b68cbba1b51c53 : SUCCESS in 1h 07m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9d9ae7e143545478c2c9af3bd733e1c : SUCCESS in 2h 07m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fadc4b9423b4faf90c2afa9ba6ffa42 : SUCCESS in 1h 39m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/798965f15734416cb8b245f59be618c1 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75acc97fb15b4cbdaaba221d91682ae1 : SUCCESS in 1h 01m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21351bb7c29e4fcfa6dad80df2384376 : SUCCESS in 5m 07s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da1533aaf7b445d6a0d5743c49d3ec22 : FAILURE in 12m 52s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6baf6b783313437bbc12d8dde7c8d45a : FAILURE in 14m 19s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bb2cb604de3458ca7f97813c9fe6840 : FAILURE in 1h 11m 44s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f28db9587dcf4fd5ac914faaa02b14f7 : FAILURE in 10m 56s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f96a8e0dceb445a6a4f35d766ea84762 : SUCCESS in 48m 44s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6411c0473024fc68fac01b0885bb1f3 : FAILURE in 12m 13s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad9d14415ede4235b61e2fccce7616f3 : SUCCESS in 47m 36s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e65f5d66c8549a2a41f98b424133c6c : FAILURE in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e031563ba0ff4ccd1e238cafc879899cae42831","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 08:13:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b4bc1282fbedb7ffed09db8697051f91e2f76af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 10:16:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d28a0ebb63be4b84bd4ee4eec4eefad4 : SUCCESS in 9m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bedfcddcdb4c4f8bb11cdfe9f18c89c9 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b0461a89a564cfda4e270ecb92b4c6a : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f057fb5f04a74b41a24eab9fb483ac9e : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74d334b1a0684938aaca0a93feca96ff : SUCCESS in 1h 09m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b00b88cb0fe4e22bb3841ca66ae6b37 : SUCCESS in 1h 55m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904db163b242475ebfcadaf78e56ea84 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7895340229ea4e9482fea7af8883e504 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c946083b6cd4c559ea4c6a7370f48f6 : FAILURE in 46m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52032ef3f5b5418e86de6e6fe0782835 : SUCCESS in 5m 53s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/984011f8288f4ed4b421b27a841fa16e : SUCCESS in 41m 53s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6f1b5593c294de394b570594c9fc846 : SUCCESS in 46m 29s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f69bb86bcf0413baf1142795cbf0ba5 : FAILURE in 1h 04m 39s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/889bb57ca39d4cd98685150bc68a2c46 : TIMED_OUT in 1h 31m 14s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a95ad4b659534e30b801e3cc66cfdda2 : SUCCESS in 50m 17s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6bbc4e0a93f4bc282632b84a2f10085 : SUCCESS in 46m 51s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a28f9a3b96f24901951e35950456b364 : SUCCESS in 50m 03s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b37a6aeb40c4036b840c3c8b67007e9 : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0c0d95489422f99ff1a45091702f8a3ade02652","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 16:54:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bc761f3cb427280055a92253758e2cc0a813bae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 19:48:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9aff5fbf73c14ae6aab8ba8579305fa9 : SUCCESS in 12m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60ceb8bd980a47008186309864895e3b : SUCCESS in 1h 13m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1ca19d8a5774647b29fbde97123486d : SUCCESS in 1h 30m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15420facc7094e72bc6d194fa4012ce8 : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0ffd5b62abb485da9ed86f844eb9952 : SUCCESS in 1h 13m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cbd96b479e174c348651a7d130ec1737 : SUCCESS in 2h 37m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da11ad0fe2743c5b61a460c9e5a2cd4 : SUCCESS in 2h 13m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/570449209f384c7e80d44a8c138b9690 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8941666c56e44f65ae8a4852812be514 : SUCCESS in 56m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecdb3ecf5ac04d5b99a7cdd3d86de740 : SUCCESS in 5m 36s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2419ef44de824d08a053484b0b3e5f61 : SUCCESS in 1h 09m 33s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5098873f36854cb687dad58e12a5bc93 : SUCCESS in 49m 18s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79f36b4e1db94c4a98e98ed02bdead94 : FAILURE in 1h 11m 19s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d5a0af585bd47d59a664a1a10c36f84 : SUCCESS in 49m 44s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11970e782bac493c9a822f9256793e29 : SUCCESS in 51m 06s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/475c3b89b204468eb01a63772f52062c : SUCCESS in 1h 24m 22s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01a770876bfc409d8375c041f23ea224 : SUCCESS in 41m 41s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b7d49dfd7884882b703afc5c7f918f0 : SUCCESS in 51m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"394282e5b0033599c5ad1a51d333df50febe399b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-10 22:33:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56cd7284cd5a4c74ee740d5ab6c0f33235ccca37","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-11-24 12:11:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a78f4aa599418148b29ecdad98e5074e639a012a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:12:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46afafef87286beed1130489498d4f6dd5708410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:24:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/71b948cf212d4466b0602da9bc59d9f9 : SUCCESS in 11m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/843abfc9174447148403dd076055fe39 : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b61d4f81ce54cf3bde4781860482b24 : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6b116563c6b4f2a954d08bc54f20b08 : SUCCESS in 1h 28m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2806b23b6e5a48c5ad9ccbb793de540e : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/76941e309a834b059ad4acf7ea8e2e3f : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/635f1678eb8c42b4bbdc1753811680f5 : SUCCESS in 1h 04m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dda78ce3a2b48a2a840807d34b0fa2b : SUCCESS in 6m 52s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dc44ede1b014bd683d4f90145a3d9fb : SUCCESS in 43m 05s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/97f36145ce6f484d938e032f99627a17 : SUCCESS in 49m 17s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/575eed4914bb4f1b8d6587d419de11ee : SUCCESS in 45m 49s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af0e9d2fa2654edfb9e4aa3dd097444e : FAILURE in 1h 06m 10s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e972cd8997ec436c83173a2550ea4386 : SUCCESS in 52m 45s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea8eab634b464694b67f8b3ae2c5f62f : SUCCESS in 46m 26s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d048f72f7d34aec9d63bad127694410 : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e9522d89315d7fa459960b60cb1119f7e0dfa1e0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-24 15:45:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6d6e757e8d7126fe3409694aa678faf7c4c5809c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:22:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84834976c3c04cc8a28a936ad1540591 : SUCCESS in 11m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bd4eb1f8f3414918947402fe231f2970 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a2136bfea5140ac844ab0a870a029cf : SUCCESS in 1h 30m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f61b453b85c48b2b21323101c53f7f9 : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1db0b5776b204e9b857a3bb618cf885f : SUCCESS in 1h 10m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ec2b5c90aac4263b18a1968da31bcf2 : SUCCESS in 2h 14m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/292261351def4d6d970aa47eafdadae0 : SUCCESS in 1h 45m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fa096722a815464cb4f15c6ce41d0498 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da018a2513a2434492eeb60269f4405f : SUCCESS in 1h 08m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3b7b2124cb42d0808c3b494568704d : SUCCESS in 21m 55s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54f539a7835d4c86b158bbe7c0f18ed4 : SUCCESS in 57m 21s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/3073e7c1d4f948909ae889f52d52399b : TIMED_OUT in 1h 46m 14s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38fe724cd22b4f9b974c9f256368dd6d : SUCCESS in 54m 36s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/db58668ed67b4eaba55ffa24f1e49849 : FAILURE in 17m 56s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b00b9d11c9844f4a1b7c3f977684bd6 : FAILURE in 1h 27m 42s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fbd2341a0cb41feb704501452c4a1ee : SUCCESS in 50m 57s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/235b84dc68e74e18959455e4186570bc : FAILURE in 37m 25s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05a7d5a5db5c43699a6c61ac4f3bc433 : FAILURE in 38m 40s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90a67ae96a91459d84c91eab41b1e217 : SUCCESS in 45m 28s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0cd6ca8f61e4f159c7b3a9b675a9b60 : SUCCESS in 50m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80ba7e889f8b9f91a841d01d1d652be9f9e86e84","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-25 10:28:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4b999712fbcf1748b0cc11e2363a902645a6d0fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:44:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1878bc25626942118bdd74f97c7f88d6 : SUCCESS in 8m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b7bad1e9da34079b16a44cd218107ca : FAILURE in 45m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f7a6d04819d6478aa6dc7a9df744af7c : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b84875e4a80f45bab8efcd4fc867d0e6 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82edc094bef54889b6b5765ac9b2905b : SUCCESS in 1h 03m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ffecefddd1da4d989d6dd0ba7efea09b : SUCCESS in 2h 03m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d0687f26fe2494ea56c97134c5ab7d8 : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bc256358b9f94a0790a1de2b52d31c83 : SUCCESS in 1h 06m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f8473390ff3458f887ba16c252df9df : SUCCESS in 52m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35965b11dd82433e8b241dc61c8da9bf : SUCCESS in 8m 22s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a16175a4b9b4676ae7539b86f7fa51c : SUCCESS in 54m 23s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/ee19ce70daea4e54a574fe5041fbec29 : SUCCESS in 53m 15s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff322a29d93b432f84b567b5dda9bfef : SUCCESS in 43m 30s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/56631e7ef8b146d5b4e85afc9a712bc7 : SUCCESS in 50m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad36bd3df5af4d3a91952443d74badcf : FAILURE in 1h 11m 03s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ec72abbc771468b9b88233d38ddf538 : SUCCESS in 47m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1559af7dfe46492fb8de61834708defb : SUCCESS in 48m 06s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c1e8a12917246f1b53a20d0967cfeb4 : SUCCESS in 49m 11s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e884eb17fad04dd6bc1e75efa52b9413 : SUCCESS in 45m 07s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b39386a22a14bb697a7a4508672a5b2 : SUCCESS in 46m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71f9aeb823c8dfcab3c131c9b36858a9fa4e92fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-25 13:22:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"03a17c1283217964d0554cb48a27f61118378da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:58:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc839c95cefd43a6aa2633b76b519a33 : SUCCESS in 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b497bc04a2b1416f8447ec6d395589f4 : SUCCESS in 1h 23m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/83f1c2559c264cd7b8cd58ffd04274c7 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acaa5d902f374f559f2d34fc8dc1eec7 : SUCCESS in 1h 22m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc51d8026a1046e987bde79fe11973eb : SUCCESS in 1h 15m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f2a37b819cbf416abee79151c7fe6d45 : SUCCESS in 1h 49m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/472f69da8f88483d8dd14cde065543c2 : SUCCESS in 2h 06m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e210e85bf8554132a1fb0268012ad2ac : SUCCESS in 1h 02m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9d5c974527e471da31f9bdf0e084dcd : SUCCESS in 1h 06m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca95dea389954e0aa1999f76c51eb1b2 : SUCCESS in 6m 20s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/701a412d77304af6ab773fd0a0f8363f : SUCCESS in 45m 19s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/dcb198acd24d4f7dacb987e9a82bca57 : SUCCESS in 51m 31s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1680e518052949d69c7aa7c3b0dda2b4 : SUCCESS in 42m 24s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/b5bc1ebddfd3478fba8137f4fa4b9689 : SUCCESS in 44m 47s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e72bb81b6df645e8aae9f0dec705a3a9 : FAILURE in 1h 09m 37s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42272bb0a84644a4966347f49fc10063 : SUCCESS in 48m 33s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44d1d715aca1441aa54e0ab3a510259d : SUCCESS in 47m 03s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1334dd9e58e3488f97068e827a877594 : SUCCESS in 48m 27s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e2691854cb84dd79001399dc6b86ae8 : SUCCESS in 45m 15s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51abaec093ce49bd8895959a69939cd5 : SUCCESS in 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c0ffda1e47d7c2059cf656774a83e68549f3895","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-11-26 16:19:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bcd9c46c12c163b43664c6ef7f58556ed67e5589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 19:21:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84c9668b5d9e401b8a689ee2b3e4c984 : SUCCESS in 12m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9d083efce35442b289b0d90d2b49b633 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/435c5cf0345342c581c6e1a5241f505e : SUCCESS in 1h 31m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8351778089d54453a6296b56d2efc33e : SUCCESS in 1h 20m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f20bbfb75d34af4a8d992cb83b0a39b : SUCCESS in 1h 08m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/50acec0d6c4f49d7ba9e64c9d8e65ace : SUCCESS in 2h 08m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c808c6254c540b0a6c10127900f0534 : SUCCESS in 1h 36m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/636314e4f9e14303ae1c06b640cd5ef8 : SUCCESS in 56m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eab8a4013024e35af39bad3580b5a03 : FAILURE in 2h 54m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d307b39cac54d088db5d3d7b6fa3342 : SUCCESS in 6m 01s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/755ae3655cd44afca7335ae471a27348 : SUCCESS in 49m 36s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/3b9fc177637c4ffb9666d53c243951b4 : SUCCESS in 55m 33s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/445b91505aab4c158832837159c394ff : SUCCESS in 41m 18s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/0e3dbef5e8e540debd8d15c2b5a63c89 : SUCCESS in 48m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b8ae4c65a0748ec83f6f5d6898ead0e : SUCCESS in 1h 16m 00s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36c400e500c445668fb21c2d0e249880 : SUCCESS in 43m 51s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dca9c0d144fd4842829884c5b94d46a5 : SUCCESS in 47m 05s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7bd4023b9fe4c578467c6ba174ab526 : SUCCESS in 50m 44s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/324534297d9342faa790554953588e58 : SUCCESS in 47m 34s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f0f62048f7c4367a03cb9d79a11a552 : SUCCESS in 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3de98412926f6cae9dd73d8721bc0743eb1d0bd6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-30 17:13:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b538e522d677c2fee68f6ab547eb632c661adc34","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-30 17:13:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d049b7b1784a64daf2125c85e608e3cc83452fdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 20:14:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/20c1a7a5e9b948d3b256afe9b77cc0e7 : SUCCESS in 8m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ddac058ad62441c681c7e9291e70b868 : SUCCESS in 1h 06m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8948155afe2e4cf5979cc07f5014e726 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d03cbd33818f45efb96917175b0367f8 : SUCCESS in 2h 54m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a600c8db18fc404992ff0851f297c1c3 : SUCCESS in 1h 06m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/74074fd8332d468d9241934e2867cd25 : SUCCESS in 1h 44m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c5d14c0ed84c039e893f8d633de722 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/495680690f0e4872972eee4b8c6d7eac : SUCCESS in 1h 01m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6755858c081a4c38bc85023a6919a8c6 : SUCCESS in 47m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc7c4950c5542b3bd30a0fbe7dccd1e : SUCCESS in 5m 00s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4644a32705c4af69db43860155e8dbc : FAILURE in 24m 05s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/581cef2f79f2411f8028755968c6691f : FAILURE in 26m 53s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3669dc58d23a4d388cf9d77f3ff89a30 : FAILURE in 26m 56s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/60149d14856146c59522209a13cf427b : FAILURE in 25m 59s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12d9ba27b2e04876be55e373267b6983 : SUCCESS in 1h 13m 06s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87b2c9360b9f4247a7fcf182168eece5 : FAILURE in 23m 20s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fe809ac126041488c29c476d32c66bf : FAILURE in 50m 25s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8135cae51d7e4b338142b353c931e9c8 : FAILURE in 25m 00s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7244083b03c74841883d099adbc9dfe4 : FAILURE in 23m 34s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bdc13c958c4441e98f823e61422ae93 : FAILURE in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0435fc017bb8530eb1a8d9dbd2009e249f44639","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-01 14:56:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1d390a49c4f19e875fbd852a466e5e08e4686fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 17:30:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa2b0d78e6ee4f0e8491f98dacbef35a : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4c211d03928e4d1b9bf554679875cfd3 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a2b3649a0cfa42d1968fa74f71d1740d : SUCCESS in 1h 27m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fedb02d4b77a47168149bd8693501ca1 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c80e7a2db06448e80aba8604de9dc33 : SUCCESS in 1h 11m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f5e5990f6c9646d392c6a8a4b54faa60 : SUCCESS in 2h 07m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6faee2ecaf934579ad48d92b79f8b87d : SUCCESS in 1h 32m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e3a5f3e24bc6465bb7e87e9916fc7bb1 : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e9afe23ed5343eebb21083d9c6835ec : SUCCESS in 56m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e2ac87e12f4f6881557cb7099af2d8 : SUCCESS in 5m 10s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66e615c0aa6d4a34b0d9635a6af80384 : SUCCESS in 45m 02s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/b89c25a9378441039a627b5ef6ba5d94 : SUCCESS in 51m 26s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/552a5d51a3264be5aba9a93f1a96c1bd : SUCCESS in 41m 15s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/1ac3cd2cddc34f9db4bbf4569a9a6e2c : SUCCESS in 46m 26s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4aa5c7a10214c3ba49d2e4eaebe7cb1 : SUCCESS in 1h 21m 37s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8787dd382970455ebefb87f60ccb7f72 : SUCCESS in 46m 16s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56dd50fb096a469b8f9c4d5f88e94f9f : SUCCESS in 49m 23s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f4c377ceda449b7b6289cce4245c940 : SUCCESS in 45m 32s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44372e653fe34eb29dc9b153f9fb029f : SUCCESS in 47m 29s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed4749064e2d4a859642aa18ec883268 : SUCCESS in 1h 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e73037d24bd8d40f4ee726e632da253c4456599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 17:30:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3192b3a4b66a7005bb7c1e10e097b46cf3b7d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:55:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bcdaccb77d7646e4826e5858c3fad4be : SUCCESS in 14m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f0301baee8a445c2aa6850e59b70122f : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/277b67d9a726411c84a23d9ecf43f22f : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c88f45468a5b44269ed01b0ac1b3520c : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c9b9fa1fdf49778d20f435d8890547 : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/69bc91cccd2c4321bd9aff0765212132 : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79e5527b57a84779bb160dc057c66b50 : SUCCESS in 1h 19m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76274e1101545fdb08a127f5ff97872 : SUCCESS in 5m 54s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed18932d34df4c638f42b72dea7db546 : SUCCESS in 43m 47s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/52821dcbdba648699021a188ff1ea782 : SUCCESS in 1h 00m 51s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/094e7e5e75eb4f099be6f0bfe00fea0c : SUCCESS in 47m 36s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6ec75edc72846dcbe76e587fe6b998a : SUCCESS in 1h 24m 30s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/758eca4adf2a4e58914e6c760e4d23ac : SUCCESS in 53m 49s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fff1124a3f2b4cf88b21267a44660e76 : SUCCESS in 51m 44s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d034ecd61c4846ce9c347582213f5e6d : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":3},{"id":"59bf245a03701133cc9d67d054f67fac6ddf4e89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"df6538cb9cc0df8765fae1544280b38bdbe355f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:56:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a02299b4a734313b4591254cffa6e2e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e43313af85dddb8f5f9af454b276a24837b3c10d","revisions":{"81bede17fc289be591b5e01649543533de39ba3d":{"kind":"REWORK","_number":1,"created":"2020-10-28 14:27:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/760149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/760149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/1"}}},"commit":{"parents":[{"commit":"2b655a857ca0d7666feb35a987bb666faa754968","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b655a857ca0d7666feb35a987bb666faa754968"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-28 14:26:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-28 14:26:51.000000000","tz":120},"subject":"Rename nova_metadata_* variables","message":"Rename nova_metadata_* variables\n\nRename nova_metadata_* variables to neutron_nova_metadata_*\nso less confusing names. At the same time these new variables will\nhave their defaults set to nova_metadata_*\n\nChange-Id: I073da66f9e395bbb99b1c21701c808f252c3a6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/81bede17fc289be591b5e01649543533de39ba3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/81bede17fc289be591b5e01649543533de39ba3d"}]},"branch":"refs/heads/master"},"46bf2378ba3c336a077d87930c367410efd3732e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-10 16:54:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/760149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/760149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/2"}}},"commit":{"parents":[{"commit":"27f8e9aa208e383018488f9a8dfb96e65383786c","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/27f8e9aa208e383018488f9a8dfb96e65383786c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-28 14:26:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-10 16:54:51.000000000","tz":0},"subject":"Rename nova_metadata_* variables","message":"Rename nova_metadata_* variables\n\nRename nova_metadata_* variables to neutron_nova_metadata_*\nso less confusing names. At the same time these new variables will\nhave their defaults set to nova_metadata_*\n\nChange-Id: I073da66f9e395bbb99b1c21701c808f252c3a6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/46bf2378ba3c336a077d87930c367410efd3732e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/46bf2378ba3c336a077d87930c367410efd3732e"}]},"branch":"refs/heads/master"},"e43313af85dddb8f5f9af454b276a24837b3c10d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-25 13:22:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/760149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/760149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/760149/3"}}},"commit":{"parents":[{"commit":"6b39e9c85b0375ae4d00a5ddf6ed859b8b112011","subject":"Merge \"Fix ceilometer constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6b39e9c85b0375ae4d00a5ddf6ed859b8b112011"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-28 14:26:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-11-25 13:22:32.000000000","tz":0},"subject":"Rename nova_metadata_* variables","message":"Rename nova_metadata_* variables\n\nRename nova_metadata_* variables to neutron_nova_metadata_*\nso less confusing names. At the same time these new variables will\nhave their defaults set to nova_metadata_*\n\nChange-Id: I073da66f9e395bbb99b1c21701c808f252c3a6cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e43313af85dddb8f5f9af454b276a24837b3c10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e43313af85dddb8f5f9af454b276a24837b3c10d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
