)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~761241","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Ftrain~I92780df25e02dd1125e53a3633715fcf948a47d9","project":"openstack/openstack-ansible-os_neutron","branch":"stable/train","topic":"enable-notifications-designate-stable/ussuri-stable/train","hashtags":[],"change_id":"I92780df25e02dd1125e53a3633715fcf948a47d9","subject":"Enable notifications when Designate is enabled","status":"MERGED","created":"2020-11-03 22:31:50.000000000","updated":"2020-11-05 18:31:08.000000000","submitted":"2020-11-05 18:29:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761241-1604600968579-bdd4afc4","meta_rev_id":"99476e2d202fc236f2641f3011af2523aece0ebe","_number":761241,"virtual_id_number":761241,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 18:29:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-05 16:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-11-05 16:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-11-05 16:21:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 16:17:46.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-05 18:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb760042cf272f68a3ea2e94e3050a0b9a9ffa64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-03 22:31:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"1b2a18704b458a4f6bb878e2eee33d65be7136c6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-03 22:32:23.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d2af6918e97dd27c96fb1c4972e915f8fddfb127","accounts_in_message":[],"_revision_number":1},{"id":"1d6a81f89361db2a45b1930408f2cb0223938933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 00:52:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b6e444a2b0a94d8fbaf3476679ced916 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4baf4d89df4942efbeab2fb4ce3bd00e : SUCCESS in 1h 03m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7cd63dffc8f1490aa350743f78e4430f : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ddb5ae5c3684c209c19108d7dd0c357 : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/24bb9e30431042088561a971b8de5654 : RETRY_LIMIT in 19m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a30fb6de83e8480780723f1672906cbd : SUCCESS in 1h 01m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b92668d5aa1747129dab61cca9ac94d0 : SUCCESS in 2h 08m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22616eb09b9d4b61a87a646c02bfd862 : SUCCESS in 1h 40m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f4ad066c496d4aefa290e348ac4fdf19 : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/23c043be5daa4b9fbfa15f0ff6049a28 : RETRY_LIMIT in 13m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cb762baea744c19adfcda3bda3e40ec : SUCCESS in 55m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a914dca5b4348e3ad4f13a4441f228a : SUCCESS in 5m 25s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe01a887f248418994a6d23eceb80fc2 : SUCCESS in 40m 55s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c42fff2d848f41168f3bf6b7bd5f0b32 : SUCCESS in 38m 57s\n- openstack-ansible-ovn-centos-7 https://zuul.opendev.org/t/openstack/build/239aefe293d64032a3636ff71b767891 : FAILURE in 26m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75dfef3dda0e4137b91cdbb870d6cc53 : FAILURE in 1h 06m 14s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b773909a1c949a89a36c627eb35c9b2 : SUCCESS in 42m 34s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/841ccce97d7c46848b1211b89d7e40a7 : SUCCESS in 43m 25s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e612bdbb06334e46bad13de257a37ce1 : SUCCESS in 41m 16s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41dd405a70a3449282efa74ffc95ca3c : SUCCESS in 32m 50s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/752a7cbb4d8c4f29869a83059e1d5d20 : SUCCESS in 38m 45s (non-voting)\n- openstack-ansible-ovs-fwaasv2-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c75dfe4f20344569c005ce873a92358 : FAILURE in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1741deddc12a6f2accc3400b69faf701d7f6fc2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-05 16:17:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0118339cafae18c84ace50e656518dd221a8a5f6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-05 16:21:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3628744beea93615c7ab131c08668f966b3b0369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 16:21:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bf6905f3a559b465a1d31b4bb34228a7a9701a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:29:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3bf8be371dcc502a4fb8f1c10eae71b6c3a218a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:29:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f45a0971fc4440449234855d2e7227e1 : SUCCESS in 12m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5770883170584633bcad75041e34e146 : SUCCESS in 45m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/392708a6abdc460aa226168c49f67cd9 : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b0449c2c35e472aa591d3772d86d27f : SUCCESS in 1h 30m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0b46de5afbaf45409534a837ecbc9cd9 : RETRY_LIMIT in 16m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d2ee936d3974d139726fe1ab916bf34 : SUCCESS in 1h 03m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4007edc92e86438fb7da82b079ce4a76 : SUCCESS in 1h 55m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4910516108ec45608a3652cc4bb27d99 : SUCCESS in 1h 14m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8f15d9017044c10baa4fb4af1494183 : SUCCESS in 48m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e39c7018360480d8455b7d02970f2e3 : SUCCESS in 5m 41s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4a61f752b5e435285b12148706f1ef7 : SUCCESS in 37m 35s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/369e6c38a00b45df908a1ab56a7a94de : SUCCESS in 38m 59s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d350db3a6cdc4917964648190b3130fe : FAILURE in 48m 12s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dcce07e578348dbb0d3c36e78e7b9d8 : SUCCESS in 41m 57s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/698cfd25d0d2492ea010a10ece409c66 : SUCCESS in 40m 27s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/068496622bcf49a692bb9a85b11eb3c6 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b0fc5e80d3da73c33d2136101a1f4d4945d62f0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:31:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed9e7512cd747919e8fcb19615978eb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c425bdbf66f42204ed5afa79b3e41ffcfd56d69","revisions":{"5c425bdbf66f42204ed5afa79b3e41ffcfd56d69":{"kind":"REWORK","_number":1,"created":"2020-11-03 22:31:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/761241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/41/761241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/761241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/761241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/761241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/41/761241/1"}}},"commit":{"parents":[{"commit":"3d7ffed155d77ee48fb2ce2135158f76dc2bd169","subject":"Merge \"Updated from OpenStack Ansible Tests\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3d7ffed155d77ee48fb2ce2135158f76dc2bd169"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-13 19:08:21.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-03 22:31:50.000000000","tz":0},"subject":"Enable notifications when Designate is enabled","message":"Enable notifications when Designate is enabled\n\nChange-Id: I92780df25e02dd1125e53a3633715fcf948a47d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5c425bdbf66f42204ed5afa79b3e41ffcfd56d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5c425bdbf66f42204ed5afa79b3e41ffcfd56d69"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
