)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~761815","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I4469dc572acf7d53bcf759e28138cc6733101ed4","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"761815-patch2","hashtags":[],"change_id":"I4469dc572acf7d53bcf759e28138cc6733101ed4","subject":"Remove firewall_driver from securitygoup section due to duplication in agents config file","status":"MERGED","created":"2020-11-07 14:45:29.000000000","updated":"2020-11-16 15:13:50.000000000","submitted":"2020-11-16 15:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761815-1605539519668-bced994f","meta_rev_id":"a920d7ad355e2fcd8d7c036b9bbc724e850b3cca","_number":761815,"virtual_id_number":761815,"owner":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-16 10:22:24.000000000","_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 15:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":2,"date":"2020-11-16 07:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-11-10 13:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":1,"date":"2020-11-16 07:34:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 11:37:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-10 13:58:16.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-11-16 07:34:54.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-16 15:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aab048ecf781185bbfbabc66047057763fa3428a","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-07 14:45:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bab71bf7b84c2cd51e85a52b76f56acec3603998","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-07 14:48:48.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7061eedb403e735c4114467fa56a95cc4bfd5378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 16:37:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4536331c2ab544f1b51f7875e289cb0e : FAILURE in 13m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bc06e376b5c64bb49a5c310166af719d : SUCCESS in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7f725acb3bba4d39a92c78386ed67800 : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e0e65b3a3dc4cbe8df1db9396cefeeb : SUCCESS in 1h 15m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac76da55a4ec4230897c484530cd4518 : SUCCESS in 1h 07m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/79511242dc124cc8928024f9082399c2 : SUCCESS in 1h 46m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b77bf1517144a2dae070d039bbf3987 : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5b51d8bbb1ce4b4cbce4047a3436d978 : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5016dfb8d464bb59ed056b1ad210c54 : SUCCESS in 57m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234ca2c380db46888a678674964ccc68 : SUCCESS in 5m 33s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77e9ce5b7b434056b86ecb177bb6e087 : SUCCESS in 45m 05s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f08e8b973094e0f856f3339d2e96290 : SUCCESS in 43m 38s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9b045d876444c9faffbc4ec62a4c6d3 : FAILURE in 1h 25m 33s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee1473c2fbfd4059acca8da2c4fba322 : SUCCESS in 47m 00s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b250293c84d74b93a0357cdb89c12a79 : SUCCESS in 46m 54s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7fed155b95e405fa6907a7bd2573d42 : SUCCESS in 47m 34s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e425d8f79f2c4493858c9449daa1b046 : SUCCESS in 42m 56s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2658dcc78eb14189b98f8ef78ba6dbb3 : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da8ae6bfacacabd665e3f31c5659937148ed32f5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-07 17:27:22.000000000","message":"Patch Set 1:\n\ndpdk is special cased here https://opendev.org/openstack/openstack-ansible-os_neutron/src/branch/master/templates/plugins/ml2/openvswitch_agent.ini.j2#L33-L35\n\nShould the same logic be applied here too?","accounts_in_message":[],"_revision_number":1},{"id":"d9b1447d9ecd49480e49da2fafa56061b0f105d1","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-08 06:34:49.000000000","message":"Patch Set 1:\n\n\u003e dpdk is special cased here https://opendev.org/openstack/openstack-ansible-os_neutron/src/branch/master/templates/plugins/ml2/openvswitch_agent.ini.j2#L33-L35\n \u003e \n \u003e Should the same logic be applied here too?\n\nto the best of my knowledge about OVS with dpdk support, that uses the regular OVS for ML2 driver. The openvswitch_agent.ini.j2 file, while created a special case for this scenario, which in my opinion could create inconsistency in configuration files, but it will not cause any problem with this commit.\nBTW, this is my first contribution on OpenStack, could you please help me understand the reason of Zuul check failure? \nThanks in advance.","accounts_in_message":[],"_revision_number":1},{"id":"e8f416fa1701451b81530a4613ec07f0d688fe26","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-08 10:27:23.000000000","message":"Patch Set 1:\n\nRegarding zuul it\u0027s collection installation error, so just worth rechecking\nhttps://zuul.opendev.org/t/openstack/build/4536331c2ab544f1b51f7875e289cb0e/log/job-output.txt#2700","accounts_in_message":[],"_revision_number":1},{"id":"72f0b550e0cec3f38b274b86a17c7742ca2d3dd7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-08 10:27:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d0c2906a3e28cba5c0585854f4950bddd5c1e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 12:29:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2fb5e09f141a457895d854b4dd9e6610 : SUCCESS in 7m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5a987b8c2559464f805ce3da03bb5b60 : SUCCESS in 1h 04m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/868750761213456694ca5ca88b2e60ac : SUCCESS in 1h 43m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d264c9ebde9846eabffdf90f93ac5e3e : SUCCESS in 1h 56m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f58bbb6c0f44d59a0918799b53dff3 : SUCCESS in 1h 08m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2ef6f300cfe5483c9f3b892a7dd14c54 : SUCCESS in 1h 36m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/291ece6743334a888e63b336e17ff769 : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b86de5dedcce4d8494fcd3eae6e4faa7 : SUCCESS in 56m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/520e2ba124c543ae9c8b99ad5c01b639 : SUCCESS in 1h 50m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7676eb9be24141848677e917aec784b3 : SUCCESS in 5m 37s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9e2d7808fd649dda2f67dd86cc7658f : SUCCESS in 43m 07s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/176db525c6544aa4aca4678a12807299 : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a2fcc6903a04b84b1f88246aba56ba4 : FAILURE in 1h 05m 45s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbd7e0db24e04f9697eb0d1954ad0c85 : SUCCESS in 44m 56s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea3436b8f8384c46a15adc7ca9443a89 : SUCCESS in 42m 13s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14204274cc3148bfa22ffdf06ee1bb3d : SUCCESS in 46m 25s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4230874e03044d3fa4c56ba2f6334215 : SUCCESS in 1h 27m 35s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c50c166d01584161a8a729a84dd25cb4 : SUCCESS in 50m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66086ea6907a98f9bc8856d49d42b3f7b500b563","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-09 10:15:28.000000000","message":"Patch Set 1:\n\n@jamesdenton can we get you to take a look at this?","accounts_in_message":[],"_revision_number":1},{"id":"37d990091cffa6a803a8467271cd2fa42f86ed06","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-09 14:46:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0a09e4d77e8746dd3fe5616a4b640e28f91d859","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-09 14:50:59.000000000","message":"Patch Set 1:\n\nJust to add to this, I see TripleO has removed the firewall_driver config option from ml2_conf.ini: https://github.com/openstack/tripleo-heat-templates/commit/e65b8309ff79341a7b4f51b5a22ac6dea5408d53. Perhaps we should consider that as well.","accounts_in_message":[],"_revision_number":1},{"id":"3ad6218e6ca9f5d958861c08e8766d27f800e118","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-09 15:25:42.000000000","message":"Patch Set 1:\n\nThank you James for your input.\nSo with James comment and the fact that other agent config files already have securitygroup section with proper firewall_driver, I think it is safe to remove entire securitygroup section from ml2_conf.ini file.","accounts_in_message":[],"_revision_number":1},{"id":"bb9b6d79851142a95f22a1958f2aed82f8b7840e","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-09 15:42:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d97cf8ad69b611e0f4205e413f88dd97b0426701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 18:30:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bdb92871f0f64921a361d3b6b1631216 : SUCCESS in 10m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0708e10ad5114ae68bd91526a29dccee : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90762340899c4460a58686662dd1c2fe : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae59c90bb85b446c976bf23c11dd58a3 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc3538bc8967492e844116ef71344055 : SUCCESS in 1h 21m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/75cee188ba774b42b826ff5aaedef4ed : SUCCESS in 1h 48m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1486edab6ae42ab85b51d6f6b39a5bf : FAILURE in 2h 41m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fc10432e0e934f549378ede53a42e409 : SUCCESS in 1h 10m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd46cdd990424297ac6b20fb6ca6ef6b : SUCCESS in 1h 01m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bdeb056c0341778dd89a1bda22948c : SUCCESS in 6m 46s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14d6cae839a94101bc681465c85b043c : SUCCESS in 1h 05m 12s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb32a182f84c4514bc383f586d347cf5 : SUCCESS in 45m 32s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f2024b8a7f54a36907080bf98887c9b : FAILURE in 1h 11m 13s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/251d1e462d7a4880ad6936fe2218dab8 : SUCCESS in 48m 33s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bda37acce493462996977b7e42ee3f1a : SUCCESS in 47m 57s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdcf69e8b655491199c89596e0fb4857 : SUCCESS in 52m 23s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a2ab2d12a8a4cf88abddd81de51dc18 : SUCCESS in 46m 59s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5550ec39874d44199006095437e64a5b : SUCCESS in 52m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4dc1cd34de50dc895c8ea56920dddb2f09ef5e5d","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-10 03:59:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3cb6882fc3ce06649edfa41c990e9ed755edac61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 05:56:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98dfcf99f43e4e9c99252498cfb838ef : SUCCESS in 8m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b76c485b951648719b524618ee8415c2 : SUCCESS in 1h 07m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/72859779693f4a16a303fa7978dd63bd : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c024802de4954146aa1fadfc388c149c : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fba234da454f959e3ffe181c41e3ef : SUCCESS in 1h 15m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/239a26cfac6840dc9f23bb9d0f3358c1 : SUCCESS in 1h 53m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85e9304967dc4d1da64bc28f6654d38c : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/38a404e249124135aba762971afd0d85 : SUCCESS in 1h 01m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d2bc4d7b30457591476d22839cf632 : SUCCESS in 1h 00m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da48ab4aee514e87b1534f93ee33f092 : SUCCESS in 4m 57s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd5437e6b3fc4c228aba7927aaf96926 : SUCCESS in 54m 25s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aae3e8906b6a4af982d6c95ec76fc0cd : SUCCESS in 42m 24s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e8c6388c1b94f59b404e65135a6f43c : FAILURE in 1h 10m 22s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4671e20a7c1c4a14989331d26660cb81 : SUCCESS in 58m 38s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70729da5716a486a87bea28c8d143aeb : SUCCESS in 46m 47s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6786529a994343de81f8f370a5e72940 : SUCCESS in 46m 57s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e34d1b76534a9bae6e31e7f338c224 : SUCCESS in 42m 07s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29a6548306864b50aca840de083dbc83 : SUCCESS in 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47d357ac23be7e14cbdbf9cdec2d96cd77d1810e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 10:31:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8f2ac02d0829d0b7edfc1ccf11b2efc0e31f9eb4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-10 11:37:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2f37071d14c48d1be5055947bdc41ac877fc661e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:37:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"115b4d59e31d32b7acf3972b21a34e8f26f11ff0","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-10 12:01:09.000000000","message":"Patch Set 2: Code-Review-2\n\nWe can\u0027t remove the entire [security_group] block, only firewall_driver. Mind fixing that?","accounts_in_message":[],"_revision_number":2},{"id":"dfa24773ebb3f25bc35228a6c3ac7867e1711363","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 12:04:06.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9b40eb97be6ef719ae1444bfc4381fcaa67dbd26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:30:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/701968f843c54c71a33eab657d67bb1f : SUCCESS in 10m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b9de3c62860c42b2bb431937b3885e3e : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bf2e230e4309412ebb228d268baf8c4c : SUCCESS in 1h 41m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cab0c2b1f98a4d9ab4b54de6bfef9f88 : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df8b9d366050439f9a4fb5b7a8570e70 : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ba874465a0df420ebfa7c74bf4b72d06 : SUCCESS in 1h 08m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f5a97df381e4925b519ffaacca531f4 : SUCCESS in 1h 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cba434c72dd460c8c84f16d8669c17c : SUCCESS in 5m 33s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71115aa60bfd4b5fa556094d7d549c91 : SUCCESS in 43m 30s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de5b985f7d64ff8abdc08c049a3e8ff : SUCCESS in 44m 25s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/613fc46d7aa649e7981a6dac7036d6e3 : FAILURE in 1h 07m 49s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/891b37a15199472f88e98a5f70456902 : SUCCESS in 47m 01s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09811f43f5904116bb0c9bd0d74fd056 : SUCCESS in 48m 56s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/220a4bc6ea2d495dbff48169c1b5353b : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a46147558afbe8d4f08a2ba4a1ed8bd28998a08e","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-10 13:42:49.000000000","message":"Patch Set 2:\n\n\u003e We can\u0027t remove the entire [security_group] block, only\n \u003e firewall_driver. Mind fixing that?\n\noh OK, I thought it was duplicate, I guess I misunderstand your comment  though.","accounts_in_message":[],"_revision_number":2},{"id":"1fe5e413ff0fbfcde95cc6a5da7032899d452fa5","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-10 13:46:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c0e9610204177f314dc5bbff5af6415e1ce0f8f","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-10 13:58:12.000000000","message":"Removed Code-Review-2 by James Denton \u003cjames.denton@outlook.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"c7818c063cc8998ad12f73695f3d284c7e87b57d","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-10 13:58:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9fc329e046bc461bdb6860fd51624aeff509e8da","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-11-10 13:59:45.000000000","message":"Patch Set 3:\n\n\u003e oh OK, I thought it was duplicate, I guess I misunderstand your comment  though.\n\nWell, the block itself is duplicated between the ml2_conf and agent config files, but depending on the service referencing them different components may be needed. In the case of ml2_conf.ini, it is read by neutron-server and we need *some* of the components of that [security_group] block","accounts_in_message":[],"_revision_number":3},{"id":"90ee2b7b39b1ebfafe28cbb35bc0032e72dd9c79","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-10 14:11:42.000000000","message":"Patch Set 3:\n\n\u003e Well, the block itself is duplicated between the ml2_conf and agent\n \u003e config files, but depending on the service referencing them\n \u003e different components may be needed. In the case of ml2_conf.ini, it\n \u003e is read by neutron-server and we need *some* of the components of\n \u003e that [security_group] block\n\nThanks for explanation.","accounts_in_message":[],"_revision_number":3},{"id":"5621ce0e7a5d5399a9d9bf9f320ed43fc25187e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 16:41:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bcccbbc586f6450d9a4b42d8081d9542 : SUCCESS in 12m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9f1183d0e98344309481f7448d038622 : FAILURE in 1h 45m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7d8a581d253b43fe8fc294b84d65aebd : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adccc26a861e46ecadfe501ab50a00c3 : SUCCESS in 1h 19m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ef736ecd4fa4e91b564ed6cb9d85214 : SUCCESS in 1h 09m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/52ce0a4f488144c68c933f2ba8492681 : SUCCESS in 1h 50m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e49c36959168429c8e3e7411c92f499b : FAILURE in 1h 15m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4b061829bca647e3b2d4f43f80dc0373 : SUCCESS in 1h 11m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b272b15a9484184877ddf6d8ef8e0bd : SUCCESS in 59m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b1cabcd28f4d9181010490f512d6c0 : SUCCESS in 5m 25s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddf271558aa64c71bcd3c17e7a489b86 : SUCCESS in 53m 04s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63caa1676a28411ea7c269728454874d : SUCCESS in 40m 25s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d5f1a8f67434ce2a9fa571ce9a6dd31 : FAILURE in 2h 48m 48s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4d7e6b9f63a40ca8b81c2de7e92be5f : TIMED_OUT in 1h 41m 49s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/243b22018f8f45d393f8d6dd93b03798 : SUCCESS in 45m 45s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b078e096e1a04df4a45bddcbe232de00 : SUCCESS in 48m 17s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a5309fd4f9c405e94c1e8cb60ba6407 : SUCCESS in 52m 47s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dbb72038a664a9ba2c1c741bdd1b14e : SUCCESS in 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49c58fdc4814cb43dcce859ea90d1ffb849010e4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 16:45:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6a138a80e06be4a4f7c88d6c28b666fe0ef00e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 19:27:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf28be5e7ae54eefa0762eccf26f4da9 : SUCCESS in 11m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3875047406b942249d1c822fca8b74ac : SUCCESS in 1h 33m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0010ccc471874a519649c01c55f15211 : SUCCESS in 1h 40m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30f16ed5c9ab4cbf9755fe22689cfbb7 : SUCCESS in 1h 29m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1336743464d46519bc4e3b9dbd27882 : SUCCESS in 1h 38m 26s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3fb1eca5717a4bbdb3da4cdd1fead9ed : SUCCESS in 2h 22m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8dffc6f94bd4c25988f91f265d9e6ff : SUCCESS in 2h 06m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/06ea3701ee334ee59170ad7a7a508e19 : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5bdc97bd6904aabb99ab59f2f914831 : SUCCESS in 1h 22m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9821eb9638ff42c69adafa4f175294d5 : SUCCESS in 5m 37s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79bb395aa3ef452a8d6776d404185411 : SUCCESS in 1h 01m 24s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96956b9daa1f41a18b9b8d1767aa4246 : SUCCESS in 55m 54s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cb7a107a7004676822849150d82f052 : FAILURE in 1h 24m 23s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4d2fe47b7064cdbbe661317e2ab70c1 : SUCCESS in 1h 08m 35s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b345664666654ac69ef52d6196651f93 : SUCCESS in 1h 05m 49s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1662455221ef4250bbb7d74ca7c3d388 : SUCCESS in 1h 13m 45s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bddd1b82ec85494bb8ff5d7d86bb8e12 : SUCCESS in 52m 56s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/473deab2648d42d2bf0ee713e8f9be2a : SUCCESS in 1h 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3d815928486179a8f0ce8984fd9921edd41675","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-16 07:34:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb27c71049a1f4f419fa640e89944c3d5523d2ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 07:35:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5473015c0859fe080959dd72e5c7ba6a52bcf3cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 09:12:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df1bed3d27fc4539a0d957adc52b81f4 : SUCCESS in 9m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6e54a466ae2d4a7bb24d50e2308561c1 : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6568ceee1b554538ae7f5978755d63b5 : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acfd33b82c1c466f8cd2c414d0ea86c2 : FAILURE in 21m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8715671513a844809bfedb2ada1b0ad3 : SUCCESS in 59m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5c244ec057bc404692cf24f87448be35 : SUCCESS in 59m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/623b3151bf96449ca81ef2942299e58c : SUCCESS in 1h 06m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3dbcb13c40340da803da89bf127a212 : SUCCESS in 6m 22s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a82089093e77490a9edb3a3c35f9b9e7 : SUCCESS in 43m 43s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9647beb57b549f7b6f7b08bf177a751 : SUCCESS in 42m 54s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea26e7aa6b344b44be2538827c344083 : FAILURE in 1h 07m 23s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f7fc8aa9c144ed89c251f75de2e2c89 : SUCCESS in 46m 08s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/692362c1520d4fc79a72525074f1dda4 : SUCCESS in 42m 52s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71b36c9bb3134483aa7dc70b3acfc488 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4e0af7513dba8d4f577b4bb27d12b38128fe6477","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-16 10:22:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f3f3c26e79139391af429ff9a036a6986d8823d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 13:21:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c89f9762cea4815b05228419eed99b5 : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7df85922308b4e9c929115c82bf63451 : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5f949afabc0c4a6f97810540d64c2628 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95fd5cc61822469e89de136face58f4a : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0a4fea8f04f420daadd75109058bf48 : SUCCESS in 1h 18m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f0b699829b7e4099a9305a20806a2383 : SUCCESS in 2h 52m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/797204cad694405f8ed798dcbae4fefb : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/557f487ed14d46efa7d38f1e3a37c856 : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35542bc5f0c948b0af2db05d16629934 : SUCCESS in 1h 49m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753f49f3fd60441180b5f89e365bcd2e : SUCCESS in 5m 49s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49717cb4d3e74d71abd445ff50eacc16 : SUCCESS in 41m 15s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2e9d39b7d144655bdf9af43210682cb : SUCCESS in 47m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b8b8c62d1c445b8ae900746561f58b : FAILURE in 2h 00m 52s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25441264653c4b1dbbb3801fe19f59c6 : SUCCESS in 47m 25s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8efa110e8864c759844f26de08d36f5 : SUCCESS in 50m 17s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2c7df568d2841bd9d6a3ab11b984c93 : SUCCESS in 1h 29m 33s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123cbea47e3c415dbbc62f780721a4c9 : SUCCESS in 1h 31m 31s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfca7f3f9f8545dd81aed1a133ae03e0 : SUCCESS in 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a21e5032f276363b6e44a1d0593f411cb858bdc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 13:21:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e0d1779f3a3d9b9884e362c26e0fc49e50cf74f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 15:11:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"55308dd0d7309d37fabb5e6a463609f2d24b6198","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 15:11:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17d15fac782c402a9f4d8ec20b770e81 : SUCCESS in 10m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b18b470450e944378cb37f6cdaac1ec7 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8c53f2d2f3e9459284e959c3a3b29c96 : SUCCESS in 1h 37m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c196a2d66c48aba591ade2a99d85ca : SUCCESS in 1h 32m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147774654f2c41008139b711ca1f99a7 : SUCCESS in 1h 43m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/298ecf65314f4e9ca00515c95ce5cf72 : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84cbbe3ff8904d1ca91575664cd25dfc : SUCCESS in 1h 08m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c333480a78584207a5303084a2361db1 : SUCCESS in 5m 44s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ea826b4e854674a5f7d7799a8dde78 : SUCCESS in 1h 02m 46s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7deb088b97ca4ec0b43dbd407130646b : SUCCESS in 44m 56s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85ad1386f9f84546b84acfd0b47a40bc : FAILURE in 1h 11m 38s (non-voting)\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4dd4e57d7b44fbb8e92c9607b00b329 : SUCCESS in 50m 11s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6047f59dcdf54c17bee2b57e954c5269 : SUCCESS in 1h 01m 30s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee03f3aefe444fa29234c4a83fb7c171 : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fbe5c417c1b24743a33ea77e723dd416e013eb31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 15:13:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3198513aef34e3bab2654bb356cff85 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"225ddf5c0a88d35da3679678264c1e275391e450","revisions":{"3c5a26cca9d125a927d9707119a6ce6be7c72eab":{"kind":"REWORK","_number":1,"created":"2020-11-07 14:45:29.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/15/761815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/15/761815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/1"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-07 14:43:48.000000000","tz":210},"committer":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-07 14:43:48.000000000","tz":210},"subject":"Add ovs firewall_driver support for ovs mech driver","message":"Add ovs firewall_driver support for ovs mech driver\n\nChange-Id: I4469dc572acf7d53bcf759e28138cc6733101ed4\nCloses-Bug: 1903390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3c5a26cca9d125a927d9707119a6ce6be7c72eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3c5a26cca9d125a927d9707119a6ce6be7c72eab"}]},"branch":"refs/heads/master"},"745345a18b8d7a8d78b313b527dcccec4ddcbf13":{"kind":"REWORK","_number":2,"created":"2020-11-09 15:42:59.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/15/761815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/15/761815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/2"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-07 14:43:48.000000000","tz":210},"committer":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-09 15:41:31.000000000","tz":210},"subject":"Remove securitygoup section due to duplication in agents config file","message":"Remove securitygoup section due to duplication in agents config file\n\nChange-Id: I4469dc572acf7d53bcf759e28138cc6733101ed4\nCloses-Bug: 1903390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/745345a18b8d7a8d78b313b527dcccec4ddcbf13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/745345a18b8d7a8d78b313b527dcccec4ddcbf13"}]},"branch":"refs/heads/master"},"225ddf5c0a88d35da3679678264c1e275391e450":{"kind":"REWORK","_number":3,"created":"2020-11-10 13:46:15.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/15/761815/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/15/761815/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/761815/3"}}},"commit":{"parents":[{"commit":"e3f04ef863c1af90143655e7640df8b943ff12ff","subject":"Enable notifications when Designate is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e3f04ef863c1af90143655e7640df8b943ff12ff"}]}],"author":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-07 14:43:48.000000000","tz":210},"committer":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-10 13:45:29.000000000","tz":210},"subject":"Remove firewall_driver from securitygoup section due to duplication in agents config file","message":"Remove firewall_driver from securitygoup section due to duplication in agents config file\n\nChange-Id: I4469dc572acf7d53bcf759e28138cc6733101ed4\nCloses-Bug: 1903390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/225ddf5c0a88d35da3679678264c1e275391e450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/225ddf5c0a88d35da3679678264c1e275391e450"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
