)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~773979","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ia48b3737718ec3928394e332ba89475ab2960fdb","project":"openstack/openstack-ansible-os_neutron","branch":"master","hashtags":[],"change_id":"Ia48b3737718ec3928394e332ba89475ab2960fdb","subject":"Switch neutron functional jobs to focal","status":"ABANDONED","created":"2021-02-03 18:26:55.000000000","updated":"2021-03-11 14:38:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9dc9ca5af65dd73027c048ed257fd4ef9252c7bc","_number":773979,"virtual_id_number":773979,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-09 19:35:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":-1,"date":"2021-03-11 14:35:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 20:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 14:35:18.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"2c8d10ff234925e13302c46405fde938e269ccbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-03 18:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f3d2fcd44bdeb761884c7f8e2604a80f8ebb42a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 20:49:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4806116768ef4f47b01deb4550446aae : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/41925d96d574494982f56d7dbdc4a328 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/25d4fad0026a42649ba4a99d7fc828b2 : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/927c7edcbf184366937fe7232b8472fe : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c77fac384884435b8f9f05a4d52b0f9 : SUCCESS in 1h 07m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/75c5b64fc41d49aa855d6f1d7d5f288f : FAILURE in 2h 02m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d5e4a6cfbdb413ea9c25c023d719900 : FAILURE in 52m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24b159b999014a6caccf329e7cd98a4e : FAILURE in 58m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05ce87c864044ad68520984cf28dcef5 : SUCCESS in 52m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05caf7784a2742c48a27169ad6223289 : SUCCESS in 6m 04s\n- openstack-ansible-ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b76179382e1b41d88e70881ee84089a0 : FAILURE in 22m 51s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f7f111e79d4e450d87f3343695244327 : FAILURE in 31m 00s\n- openstack-ansible-ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8e06ad6f82f414880ada0b8e1ad188a : SUCCESS in 41m 57s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/9478241e0b8941fa933367aa3b9b8de2 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1146b325a5504039914385ac39897f3c : FAILURE in 1h 22m 11s\n- openstack-ansible-opendaylight-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffbec2d74ae2498697b3cf093c43cb95 : FAILURE in 37m 48s\n- openstack-ansible-ovs-nsh-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08706581f0a04cd1b5eca63fd30729b9 : FAILURE in 37m 43s\n- openstack-ansible-odl-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d062669c094b838fc1c0f8a73a5e55 : FAILURE in 32m 08s\n- openstack-ansible-ovs-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6aaf6b96e2042e6bd7939e19a254b7d : FAILURE in 34m 04s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c782e4483954bd6a0e216583d933b6d : FAILURE in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7946b9757e483b7056f69ccce72dc0b37b5cb257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-09 14:54:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ecea95a6535f4fa331f3822796d9a49d5ab4b022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-09 14:54:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1a9f9475eff6b26c6d376e2d2f27a67023bc9772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:56:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3fc2195be2894222a884fead3ae23661 : SUCCESS in 8m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/acb7c5013ced445d8991031ad281cee4 : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d13511bff6c406db04d56cb8dae430a : SUCCESS in 1h 20m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98971d2af65c4a1496174d86cc31732d : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0b2de75019b44e5893628bd3aab7208 : SUCCESS in 1h 12m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e1a2d42fef15484eb05552344a19c747 : SUCCESS in 1h 51m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b207d978c50e4f7a873720601657ae14 : SUCCESS in 1h 46m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9ddc69dddbd14de98893cb4e5fd49dc4 : SUCCESS in 1h 03m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4beddf909bf4d50b0950d3dd266675f : SUCCESS in 51m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c8d74e217b4f208e50fa4e29da26dd : SUCCESS in 5m 48s\n- openstack-ansible-ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd437d9fd54d4acda29ccd9ea8cc5a28 : FAILURE in 32m 26s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/ba3ef5dcccff42acadc3ace88bb19b12 : SUCCESS in 46m 27s\n- openstack-ansible-ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e9d53404990483bbc2f5e0aff876eaf : SUCCESS in 40m 22s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/aa9380816d724ea2b94959f405d36a54 : FAILURE in 34m 40s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec20cd9ce8b64ac38c1357c594e7ac22 : FAILURE in 1h 14m 02s\n- openstack-ansible-opendaylight-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4617a66f41c04d479bd0705cc8931809 : FAILURE in 31m 34s\n- openstack-ansible-ovs-nsh-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9b8fa1f336042bab027442ee643da57 : FAILURE in 31m 57s\n- openstack-ansible-odl-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d957c2866bd472f939abf4d9dc5fe2e : FAILURE in 37m 09s\n- openstack-ansible-ovs-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d337537a32d4996bf610164b7d194c5 : FAILURE in 38m 12s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e542d132bb1454ab88b5be7cd7f639f : FAILURE in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ceb91cb1e51f77ffdf6f40dd0856da0ff6145a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-09 17:12:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"34694f8f1eb3888e6230a28f561f6488352f9785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 19:35:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57e13c46d0064a4392e6013f6296d6e6 : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7365f9d78ac34eb5b1f3447373fb678b : SUCCESS in 1h 10m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3f565f066bea4f5b86988865e02e8f4a : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26d133c76dd94cc4ad41dbd403ec7fd0 : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/962afdd5c64f47dfbd752574bb483324 : SUCCESS in 1h 23m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/02f9b1547327478da37059a30fe244c0 : SUCCESS in 1h 55m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7311cecfd0b460fb3318253ad6ce730 : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e98e2f92dede4010ac59e0f17cb7be1a : SUCCESS in 1h 00m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f11c09620def4d60ad5b58124bf536e4 : SUCCESS in 56m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4914acc0a60d40e896db5e415d196b8b : SUCCESS in 5m 36s\n- openstack-ansible-ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a190185dad944ecab9041ea52b91b37 : FAILURE in 36m 49s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f367cc4f189c4024917e24ac52b2c157 : SUCCESS in 48m 03s\n- openstack-ansible-ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad0a21b5235c4376b51577ac9c88778c : SUCCESS in 46m 37s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/9250dab49a4643579a7e9093a8319c61 : FAILURE in 36m 37s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c3de338bb94f5f9022e672a82f151d : SUCCESS in 1h 18m 34s\n- openstack-ansible-opendaylight-ubuntu-focal https://zuul.opendev.org/t/openstack/build/990e1882ba2a4f76b59b8c670586ab05 : FAILURE in 32m 21s\n- openstack-ansible-ovs-nsh-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98bd32841ddc4e9ab3f1857cb1bcf4ce : FAILURE in 36m 04s\n- openstack-ansible-odl-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4f9a2d3a84e4e209bbfc760a460dd4d : FAILURE in 32m 25s\n- openstack-ansible-ovs-sfc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be8f95a57434832a19bc42f1b235ab1 : FAILURE in 31m 51s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e522126fb65843dcb8df0711a95fe092 : FAILURE in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9683a77c3420addb654ba851d4d5765fed4b9018","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-11 14:35:18.000000000","message":"Patch Set 4: Workflow-1\n\nThis patch has been superseded by 778015.","accounts_in_message":[],"_revision_number":4},{"id":"9dc9ca5af65dd73027c048ed257fd4ef9252c7bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-11 14:38:41.000000000","message":"Abandoned\n\nAddressed already in 778015","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8efa5822004fc6e492fe615d8dae09b40219e936","revisions":{"205b7f7e7dfef96262642bdc82ecca7cc3666369":{"kind":"REWORK","_number":1,"created":"2021-02-03 18:26:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/773979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/79/773979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/1"}}},"commit":{"parents":[{"commit":"4b080bb3f940128c6b6572e634ed4070e88d81e0","subject":"L3 agent cleanup_on_shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4b080bb3f940128c6b6572e634ed4070e88d81e0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-03 18:26:43.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-03 18:26:43.000000000","tz":120},"subject":"Switch neutron functional jobs to focal","message":"Switch neutron functional jobs to focal\n\nChange-Id: Ia48b3737718ec3928394e332ba89475ab2960fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/205b7f7e7dfef96262642bdc82ecca7cc3666369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/205b7f7e7dfef96262642bdc82ecca7cc3666369"}]},"branch":"refs/heads/master"},"938ff1d12b3bd0701ecb993ede5081806be979aa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-09 14:54:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/773979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/79/773979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/2"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-03 18:26:43.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-09 14:54:20.000000000","tz":0},"subject":"Switch neutron functional jobs to focal","message":"Switch neutron functional jobs to focal\n\nChange-Id: Ia48b3737718ec3928394e332ba89475ab2960fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/938ff1d12b3bd0701ecb993ede5081806be979aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/938ff1d12b3bd0701ecb993ede5081806be979aa"}]},"branch":"refs/heads/master","description":"Rebase"},"6ac1f85edd531978add784da15f189bb9f4ef8c5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-09 14:54:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/773979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/79/773979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/3"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-03 18:26:43.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-09 14:54:58.000000000","tz":0},"subject":"Switch neutron functional jobs to focal","message":"Switch neutron functional jobs to focal\n\nDepends-On: I351b3f2ae458abc14a899768a04999ca10c86ea4\nChange-Id: Ia48b3737718ec3928394e332ba89475ab2960fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6ac1f85edd531978add784da15f189bb9f4ef8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6ac1f85edd531978add784da15f189bb9f4ef8c5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8efa5822004fc6e492fe615d8dae09b40219e936":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-09 17:12:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/79/773979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/79/773979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/79/773979/4"}}},"commit":{"parents":[{"commit":"27f7b8b4460272ef6a23b2841ee32166c682c06a","subject":"Move neutron pip packages from constraints to requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/27f7b8b4460272ef6a23b2841ee32166c682c06a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-03 18:26:43.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-09 17:12:36.000000000","tz":0},"subject":"Switch neutron functional jobs to focal","message":"Switch neutron functional jobs to focal\n\nDepends-On: I351b3f2ae458abc14a899768a04999ca10c86ea4\nChange-Id: Ia48b3737718ec3928394e332ba89475ab2960fdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8efa5822004fc6e492fe615d8dae09b40219e936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8efa5822004fc6e492fe615d8dae09b40219e936"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
