)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~777650","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I9aaeee50a4d07159a7a8a886c27eceee04c162b9","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa-facts","hashtags":[],"change_id":"I9aaeee50a4d07159a7a8a886c27eceee04c162b9","subject":"Use ansible_facts[] instead of fact variables","status":"MERGED","created":"2021-02-25 18:42:14.000000000","updated":"2021-05-17 14:57:27.000000000","submitted":"2021-05-17 14:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"777650-osa-facts","meta_rev_id":"04f231ef06938944853469caf26b5c875a464a94","_number":777650,"virtual_id_number":777650,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 14:56:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-05-13 06:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-05-17 09:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-05-17 09:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 21:03:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 13:11:41.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-05-11 13:29:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-05-17 09:11:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"2a65c3d4d46e8adad46423628f66359b5f95b970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-25 18:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"227621274c2accb1dea0bc81664b2c89eec7384d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 21:03:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/25bf1e1adde649bca91663850903135b : SUCCESS in 9m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e3598950a01f408b800092dd9223fd55 : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b2f1773c8b7640ebad7c827e603882a9 : SUCCESS in 1h 13m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c89e3eb678db475cbcb1e1a4c5cc7c25 : SUCCESS in 1h 10m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13a2e44cd8f49a28e1e3e5067135e9a : SUCCESS in 1h 06m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/731682690ad64561b3bb6ea85d007f6d : SUCCESS in 2h 05m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04c1f47560ad4cca898404fe0c86a189 : SUCCESS in 1h 53m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d0ca8e52afe24e148651e7812de01994 : FAILURE in 43m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5998c1b4c14d452f829aa5e1ee1f76b5 : SUCCESS in 1h 06m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f172d6790b1a4e40a5a02126b90b4387 : SUCCESS in 12m 59s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a6138df974943a299c920456c27edf6 : SUCCESS in 55m 51s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/589e5bc589e24edc914c77b95e8d3e11 : SUCCESS in 51m 57s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc3d9b4449614415b9bfc5ca4bb66272 : FAILURE in 38m 36s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/6c32cb7e9232411e998b80bf73b623e6 : FAILURE in 31m 49s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00bff0ca3461409385cf15eaf291dc9c : SUCCESS in 1h 09m 15s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebb5797c45ae47afa4be0675a52c7b9a : FAILURE in 42m 35s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c78476164e8d4b599690a1cc1fe9a159 : FAILURE in 35m 51s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3aea0338dbae49f3829780e91ddb51b9 : TIMED_OUT in 1h 35m 07s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b7fb5c489a94fc6b7620111fdcd4712 : TIMED_OUT in 1h 34m 43s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/466048c0ee8849d2b943c9d651ba827f : FAILURE in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"695a71679dafe19c93ae0013dcce56b86642a6fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-01 13:38:23.000000000","message":"Topic set to osa-facts","accounts_in_message":[],"_revision_number":1},{"id":"b369483d6e99f12362875bb1542a724e95322dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-15 18:48:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2f6e10aa5a3bad680eeb19b935e851a406e2a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:35:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/674555e9166845e7b0776f9ea4d3884a : SUCCESS in 9m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8720357f2f6f41ee8458904ddff81b17 : SUCCESS in 59m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e8a1245e5484fc480f0938552ad7ac0 : SUCCESS in 1h 07m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/974399f2b4f94368a6e36043527759ce : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f43966f35cf4fe28183b55854a8c624 : SUCCESS in 55m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3c575b659d6b44169067d6096428bf9f : SUCCESS in 1h 22m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd3c003b4d94095bddc5d85d739b230 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5d3a4b45a505425c9f0f23a447287bb4 : FAILURE in 31m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddc17eb908324a3eb7474ed6ec499f96 : SUCCESS in 46m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3349225148044b18b4a5ca2d88ba1862 : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce6619111d6a47a2ad0cc92091f3e862 : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c568fc2f764602a73fa8cf8b05a5fd : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c19739b488444092b67f95f40cea3bfa : SUCCESS in 53m 38s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0f271437efda4479b03a659a6ef12760 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/bc241b374fa740838ce058ef2ffa5743 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/cb56db8e451a43efbb4034e737ec5334 : SUCCESS in 1h 42m 34s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/6d4627688f634516bbd1fb94fdc95660 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3412c39edf82e87a8f6a4f3b15fdb14a1cdb3f4f","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-16 13:11:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nWhy are the calico overrides being deleted in this patch?","accounts_in_message":[],"_revision_number":2},{"id":"2f4ee31db37d421329c8a59b59f8ce2ac968331a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 13:20:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16348923d3f70717d042419c4ac1b952b5c4e7c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 13:28:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db3cec6563b6f11132962b4a615ba7c1d2f3a319","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-16 14:25:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"73f6dfe7b48d04a76734c5b36261b20818aca365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:42:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3cca72bd332f4bf49fc2601f59fa20e9 : SUCCESS in 12m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c3989f3c3c5a4001b7bf48dfa4631bbd : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b75ffa2570ca4c90bc7be1d2820d02c8 : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bee6961e3914da299bb119de359a5e1 : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdca5d5c85554dac9a83a102074e7f94 : SUCCESS in 1h 13m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6d8e6594d5bd4b938271c5463caa98ef : SUCCESS in 2h 00m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3fc24f5fe24c32a6b4d4db9bad1551 : FAILURE in 1h 10m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8c872c8ff5bb4e449245cdbbed194d4f : FAILURE in 41m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3237ea933c0243f39dc5fcb5d9c2a138 : SUCCESS in 56m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1c30b71fd64ef6a073ff89a01d622c : SUCCESS in 7m 27s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e0feed6d4942cc93a1b33939d4834d : FAILURE in 1h 11m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/884a5f8bdd534d8598532929502ba2d1 : SUCCESS in 1h 49m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951b2ed9c35f46cd93dd505d14d16e54 : SUCCESS in 1h 04m 03s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/a0713c77cddc49969e3e7cdfbb355dfd : SUCCESS in 1h 54m 04s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/ce4085cde1d6418dba523d829b14d1f9 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/10c8320626f74d7d9fef491914228ca9 : SUCCESS in 1h 58m 17s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/8671ca5f313d4cd78add860b46e586bc : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":3},{"id":"741c296b3c1756d91ddf75d9fad6df4031e20b6f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 19:40:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1289885e87aa430ee53f881fd2a154913e8ddf09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 19:41:10.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"a95dca7a261cb4b9f94194803143e9873ce3c11c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-13 22:35:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d88d2034214a45c928a8851e99c605ba120d3d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 01:42:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/913081a1bb194f6a99cedef3104915d1 : SUCCESS in 8m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ab649f4fb6814d3aab9b8bbd32bea874 : SUCCESS in 1h 09m 50s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b08aebb17d2341c99eb1a354dbe80053 : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f82c383a494c4a1ba8fdfcbcafabbd4b : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b320a1639744ccc8e2267524cfd5444 : SUCCESS in 1h 12m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9eebaa08350f4e6b91390e1426da2aab : SUCCESS in 1h 50m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/508b3f2c24f54e9990251b1434cbf525 : SUCCESS in 1h 36m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/04b135a6abeb4fb3aae83f2b568db7ad : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a728dd06d8c545c39069214732301cc6 : SUCCESS in 59m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314e2d4af1b24e6892f63a1792463f84 : SUCCESS in 6m 29s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e87b6310246d47e6bed0db8082e0502c : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4d52ec3e7804389adcd344cd8075d48 : SUCCESS in 1h 48m 00s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09573f79a23b45af9da97b7187b72e1e : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/170ca9b3c5d44c4ca54cc97b5c7151c3 : FAILURE in 1h 49m 47s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f16f591468c34c0097445a579598cae2 : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/35d5136375134a2db7f8cb3062e588a1 : TIMED_OUT in 3h 04m 24s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/7d02b91558ca49c9b0ff3f8cadf97f79 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b4c4bcb3cc9dbaa198fefe54f4575066b129b9a0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-15 16:19:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cdb00efbd8abd660ac708db5b182daa892f58d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 18:29:27.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/74d5259a31024c98b97322ae99ec5264 : FAILURE in 7m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/09d4eca94e7c49f4aa3230febfcbcfcf : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1a3e1844c28f49399ebeb8a0bbbf4cf8 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6edf7b75b9c2420988d896b9921fd6fc : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11888b43b1964cae99dceb7435e7b3c1 : SUCCESS in 1h 11m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3fee32d09f0241a99a420446d0d0e61b : SUCCESS in 1h 56m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/352f35fe8a684af589e79b10493893c9 : FAILURE in 10m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/32cbbc77a06a490f979aec4a6d0454d1 : SUCCESS in 59m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50e034374ed54d4abb8d68c3f373ef3b : SUCCESS in 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12563b8457f34334a629e4a249d72a28 : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6109b7ec2f24f0a8091858b57cfd670 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c54a491119ff41ccb6865bfd0874c9f0 : SUCCESS in 1h 46m 56s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15ff7e4820614819875b174d353e354a : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/8524f945768440a984197fc61c9d643f : SUCCESS in 1h 51m 55s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/149a75f2e84d416199d574c0ea3ab0a9 : SUCCESS in 1h 23m 46s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/11d88ac32c6943ba9eb28d33e3f17454 : SUCCESS in 1h 42m 55s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b9d80b5cc67b4a98b49950afc44734f7 : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":4},{"id":"78ee49f9e292b97f5c91b391948ec02f173059da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-16 17:04:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d6873e4b308506de68d997d940105a35fe80427d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 20:11:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9a149dd6cf5c459ba47237efb126cf19 : SUCCESS in 9m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b90e1cd3b05d4c29afc71898f36ffc00 : SUCCESS in 1h 08m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b0e505b2fd504989ad5ed56724891a26 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/963ec256ae5c42bca7413a61d8f54e8b : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/960af0099f13465ba8a53153cdc3298a : SUCCESS in 1h 07m 09s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f3da4208d3774efea1455135831330f9 : SUCCESS in 1h 53m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89615dc732d4d729ae4059e1b0d3cdd : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/25c98fca5c7147648ff331fdd7138d70 : SUCCESS in 58m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac42280c51754f9aa9cb23905e5bcf0c : TIMED_OUT in 3h 01m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae2b88fb83249e98d410fcf39eac58f : SUCCESS in 7m 44s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee35b142cbb84a538bbdeb696672135c : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/625abe86b551425691be1b786d23bd7c : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c95d49e4484c039542fbb0cba0b264 : SUCCESS in 1h 03m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e7eb7e5e95f0418fb07c1268e091a6ac : SUCCESS in 1h 50m 08s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/783d9825c707449f8c58e309f71611a2 : SUCCESS in 1h 18m 24s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/2e67c5ee17de4f078f7f34cdb36bdec7 : SUCCESS in 2h 34m 10s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/560b6f774822481787e7ac8ba95ce47a : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":4},{"id":"259455301f72069656e05abd51f31a7959445cb6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-11 06:30:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c1b95ea93cf86b01cbe1779978d3f6668027bc8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:35:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b03d62f9640f4f7e838e16dccde662c1 : SUCCESS in 14m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/da203efc863d4fd7aa038043e6084192 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/64eee9ee0cda42329f0af3cca95f4a62 : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9ab6b980e0f4e22ae5dc5993b62aa3a : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1324bdca9da446e1be990cb9587072cc : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f76926d59d14b90a5fab11772c5b9c2 : SUCCESS in 1h 18m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/77c68a6506d64aa490c9f196d138fd68 : SUCCESS in 2h 01m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45d9ded07a394b409545c047a149778d : SUCCESS in 1h 46m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ed1002f8b1b49ad8a5655e16d9b6560 : SUCCESS in 1h 05m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6194611f07bb4fc5bc11740325bff3f2 : SUCCESS in 58m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8197698d56a4dc2acd0f70a3d06dc70 : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e85a6c68c84e4ae9896c336660ee76d2 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6693e6b227f34ec6842cd2ab97889c6d : SUCCESS in 1h 39m 02s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d99309b40294937beffe3af8a02daf5 : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f9d219e309dd4fd790116c7c26db37c8 : SUCCESS in 1h 41m 33s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/377549d4aa8e44bfbf96367fcbfb20cf : SUCCESS in 1h 11m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/be4dc22e5c004ba990695926a79d21f1 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/3795d0d562a54947af1c7383da27ff40 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4863ec1d65fb8145066b07a5295ba641c081384c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-11 13:29:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e66b0d48457d280fe8fe602788bc1891f0e5e74e","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-05-13 00:50:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI don\u0027t think dependent_neutron_roles.yml should be removed in this commit, or?","accounts_in_message":[],"_revision_number":4},{"id":"0069748d1e1147a2b64f150823ab9d14ad4e9ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-05-13 05:36:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7479b193bb941c46c80639e768d76f2e486c4bc3","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-05-13 06:38:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"92a5e4faf9647648277883f4b33726c55def94fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 07:35:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/730b37c6832f47e19cf5d7536c76602d : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0cb751b95edc4fa29685844d89322932 : FAILURE in 50m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c3b2e01f9c0b480d838f3cbdef9145f9 : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5de79fa42fc14d6691ab97dd6b9f1ea0 : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2086b80216a438684ca04357991fd64 : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/703c642c513c479f94e6f1397fbb8929 : SUCCESS in 1h 11m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b8da614620b9464385d9f9b060c10027 : SUCCESS in 1h 50m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964abd86592f41929e91092b2618018f : SUCCESS in 1h 37m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fdf017bcd8454f94a81644f7b13a6dd2 : SUCCESS in 1h 04m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a73e6c89e04fadbef160fc95dc9dee : SUCCESS in 59m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac255c6d47c4c06a97fe898f5971e12 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ad1ebb451f4b2eb1f15c81cee18bbb : FAILURE in 1h 23m 08s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80dcc5e76cbb4d28885fd486a98bca0a : SUCCESS in 1h 46m 12s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67caa6116b9b40219db1445bc1e303bc : FAILURE in 43m 25s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/94b8c5c0426e486085d37db4db3f7bb0 : SUCCESS in 1h 46m 53s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/5f40a27f2271443e9f469fce7ab154f5 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c9c8f9301cf24db093a905150b6f0393 : SUCCESS in 1h 52m 13s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/069d22a0cacc4ffcb992717c396731c4 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":5},{"id":"1ac2d48fe3634e284db1dc26730cd1297ef0298b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-13 09:05:58.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a5fa97165349870d5215ec0592c49a4f7f63d161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 11:17:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/467edfca551345b397a164a1377b4156 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b7a5b8d146ac4c4195a99c33832d4ee1 : FAILURE in 44m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/212d3835e1744078a2c6eef8350cc001 : SUCCESS in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e70a3e6da0b4d5d8344c8f85256dbe0 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df9456fb0de4408a2ef94cc9d763095 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd23173fa2264050814328cbaeca4fa9 : FAILURE in 44m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7bb176f580424cadbd08a8a520ccc153 : SUCCESS in 2h 05m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62e8865881874a44b241bb55e77e4945 : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1dea663e1b3d49db9b1ad053ebd64262 : SUCCESS in 54m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/577c6d2149024f388ad2a21c14a2fa78 : SUCCESS in 51m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236db107b1fe475bba1e8a96cb98f801 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f415e836e8824c9782a3cf9f681a8cfa : FAILURE in 46m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2f455b795314d1998d213a5e0907b92 : FAILURE in 1h 07m 13s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deed260604d34de79c1b550183cc24ed : FAILURE in 47m 43s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/822c4bd016dd4ae3878b8975bf013feb : SUCCESS in 2h 02m 24s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/06b5a2c63a1e4a10a603b5ce72e738f6 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/624a4c38f67243ecbed973009c18b84e : SUCCESS in 1h 46m 37s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f884d72513ff46a78097a14e59031ac6 : FAILURE in 41m 08s","accounts_in_message":[],"_revision_number":5},{"id":"67e4db061a5f81a698ef487123645230c9314994","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-14 12:07:24.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bdcb6a0b33488d383e49057e0e3beb253b1f5ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 14:04:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/241b2a44c7b74694bfb16f4403a84ee5 : SUCCESS in 13m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/919bd5badd914f5393a75e5f96079a2b : SUCCESS in 1h 09m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b457934e9a9d4627a2eaca127d4910f8 : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bbde3050149436da660c636c6e99898 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91a1f10b3adb40c290b5ce38125bfd13 : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac7d2aa7877046deaf48fe5bde6e897b : SUCCESS in 1h 07m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9dee0dfe80c5418995a102dfd2d6725b : SUCCESS in 1h 48m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac5a93a67b64c21950433561b627ed1 : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ca0e06f468e4bfda15649d9bcc72fad : SUCCESS in 57m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/980f51d7a6074a81bdabaee0a136927f : SUCCESS in 58m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d9f362b6d3473d871611ba53c7ece7 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdb431d8dc954e54beb8da4e85a247a0 : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f03f55049d5e42e5889a9d27aae38174 : SUCCESS in 1h 49m 40s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0372480d25940c4a3cf55956ba98652 : SUCCESS in 1h 02m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b6cdebf92ad44cce888acae8a1e00642 : SUCCESS in 1h 44m 11s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/c2f3b38be6a44667a87277f2c203c854 : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/725c2c7bddf241f0bdef2d824e680292 : SUCCESS in 1h 40m 33s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/6718dedb48d847808b15a0c5e093917d : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":5},{"id":"80812384ad4acb2b5d736f2378a76ea236c11f94","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-05-17 09:11:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"270603c4fffd290aa3bcedd8a375124ff378b80d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:15:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"24673065cc4d4462af7198edcc3adcca3287b02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:13:21.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d85a2fdafb8b4e63a43793ce65b86fef : SUCCESS in 12m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d6867b374f9e4e6385dd512f0688f6de : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/85dd494f261b4786a747a994e8d93e3d : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e7680d0748d4b02b88e2b2f027df5d5 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ff1038ebf1d4a57969dbe4a5a32495e : SUCCESS in 1h 09m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b6114574fbe438c922bfb49ebfafc8c : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6b530c468374dd29f7f25b77482fd0e : SUCCESS in 1h 04m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/743a00e5999841b7865d4071b7862847 : SUCCESS in 57m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7064f5bee7f74c918d147184ad707842 : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbc027979ecb427c998982d2e7f14117 : FAILURE in 1h 14m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/975f6406b78f40d19ad725b6796fb366 : SUCCESS in 1h 44m 54s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71a0f10f090349f797c919ea5f702e06 : SUCCESS in 1h 06m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/35cf89cf4ef84b18be637f8e731f6f7e : SUCCESS in 1h 53m 31s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f7b2af352b8d401b8e0312624bfefe69 : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/ff229fe64ad44753b2a59ceca8f5675b : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b7f11c6e48054bae838d178c1c0bfb86 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":5},{"id":"38572a8aeffbec555afc194f4bb76df017c39283","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-05-17 11:17:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d6b9e49491d6bbc50b4999ad4e42405dbff1841d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:16:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd1b7d4931b54e7387ce195984f42543 : SUCCESS in 11m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a28a3b95f66f442da1fb58fd351be8bf : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3f26f4d5e5e042bfa7619658f4172410 : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c4ef55880454a628867cfb6cb8a21da : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/994551fe2ea34522ac436d55fa9645af : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05d76241ce7d465b9c39df4cc3d38f4f : SUCCESS in 1h 05m 26s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3aa2d28a4a6546d288f31435701825fd : SUCCESS in 1h 52m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e532b9eb1ba4fa981e72aced211b051 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/694442bae8734123b08bca2e0d2fe3ad : SUCCESS in 1h 04m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7008a930cefd4296bbe4e405634268fc : SUCCESS in 1h 09m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d75558c1b848f59e7c01b39aa55990 : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c2e1b449cc44fcfb245bfc12935cd5b : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b4456ecfb6b4ec5866bad915ee47d93 : SUCCESS in 1h 44m 00s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67151d4a82d04346a0ffa409b63c998a : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b5ab8e8ede8840bbb17290670503d4ba : SUCCESS in 1h 48m 51s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/41af9cffae4a41a9b5d140de06c67163 : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/d0db9619ad134007866bbcd730f021de : SUCCESS in 1h 48m 08s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/99fa45fa15d24c72a4fcdfb0d57b299d : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":5},{"id":"4c7c07515c5ba117b5172cb9f70c3b7ceed3c564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:19:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dcd6e9afa2927a70b5d898495de05db6d9500738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:56:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d65f6e4e354247bca2dc4109efb6a6ab : SUCCESS in 9m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96c3a9e7442d4a588332b8b2aa0ab4a3 : SUCCESS in 1h 04m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/818a6749c0354a2283f699307687c0bf : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0dbbf53b61c467a8f6a2e7806bc98f1 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37fd6ad187a0429c983961ebe3ef6012 : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/665181aed3c4426d9a363623a3b6715d : SUCCESS in 1h 04m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d910c3dd41be4802be89167c3b797aaa : SUCCESS in 57m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/864a110e18304f0a96b41a89c05a82e1 : SUCCESS in 42m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ef63e9bc6044e8a65962e4ca9a6bdd : SUCCESS in 5m 23s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/843c1d68d7854894a17dd5c9bbfcda34 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3625cb17f684adfb918a38f9d7bb43e : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b8d36e3d5ba4676b1903fc04656864c : SUCCESS in 1h 01m 06s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3b1609142e2d4bcc9e23b223b113cffd : SUCCESS in 1h 33m 45s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/c7ce79224edf4ff1b324a3d0fdca371a : SUCCESS in 1h 06m 07s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/90f88e201dc840bcb0137a173df8aedc : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/00ac54b13ddb4e84a94baf0b970b35e3 : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":5},{"id":"1c9487e2b4968a16ebb9c14519d305c0c4a6a758","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:56:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"04f231ef06938944853469caf26b5c875a464a94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:57:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee9fc3351f241ab9649807bf7aa166a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0495f54875eccf95078c7944dd8da0338a30ee7f","revisions":{"0887c17c20b94167d88917b9589515014c3ee334":{"kind":"REWORK","_number":1,"created":"2021-02-25 18:42:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/777650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/777650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/1"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:41:28.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:42:12.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I9aaeee50a4d07159a7a8a886c27eceee04c162b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0887c17c20b94167d88917b9589515014c3ee334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0887c17c20b94167d88917b9589515014c3ee334"}]},"branch":"refs/heads/master"},"002b96052700018f75bbbaf735462a52335ad7e0":{"kind":"REWORK","_number":2,"created":"2021-03-15 18:48:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/777650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/777650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/2"}}},"commit":{"parents":[{"commit":"754a1c03998ce5639cec1a9d82689fcfad0a2f3c","subject":"Merge \"Prevent neutron-l3-agent restart killing metadata service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/754a1c03998ce5639cec1a9d82689fcfad0a2f3c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:41:28.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-15 18:47:12.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I9aaeee50a4d07159a7a8a886c27eceee04c162b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/002b96052700018f75bbbaf735462a52335ad7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/002b96052700018f75bbbaf735462a52335ad7e0"}]},"branch":"refs/heads/master"},"3d23e0d12e3ea5f5247fd4b9a21ba9e270e0d766":{"kind":"REWORK","_number":3,"created":"2021-03-16 13:28:45.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/777650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/777650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/3"}}},"commit":{"parents":[{"commit":"754a1c03998ce5639cec1a9d82689fcfad0a2f3c","subject":"Merge \"Prevent neutron-l3-agent restart killing metadata service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/754a1c03998ce5639cec1a9d82689fcfad0a2f3c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:41:28.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-16 13:28:34.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I9aaeee50a4d07159a7a8a886c27eceee04c162b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3d23e0d12e3ea5f5247fd4b9a21ba9e270e0d766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3d23e0d12e3ea5f5247fd4b9a21ba9e270e0d766"}]},"branch":"refs/heads/master"},"79499dfb875b08cf0cdecea55cefa88e6ebf4b62":{"kind":"REWORK","_number":4,"created":"2021-04-13 22:35:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/777650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/777650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/4"}}},"commit":{"parents":[{"commit":"b5d34fe8fdd6bd836bc3d566b770b47109b228fc","subject":"Merge \"Fix u-c filter regex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b5d34fe8fdd6bd836bc3d566b770b47109b228fc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:41:28.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-04-13 22:34:52.000000000","tz":60},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I9aaeee50a4d07159a7a8a886c27eceee04c162b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/79499dfb875b08cf0cdecea55cefa88e6ebf4b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/79499dfb875b08cf0cdecea55cefa88e6ebf4b62"}]},"branch":"refs/heads/master"},"0495f54875eccf95078c7944dd8da0338a30ee7f":{"kind":"REWORK","_number":5,"created":"2021-05-13 05:36:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/50/777650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/50/777650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/50/777650/5"}}},"commit":{"parents":[{"commit":"4e123a134db6f5675d25abcfcb732263270c9f68","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4e123a134db6f5675d25abcfcb732263270c9f68"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-25 18:41:28.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-05-13 05:35:49.000000000","tz":60},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I9aaeee50a4d07159a7a8a886c27eceee04c162b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0495f54875eccf95078c7944dd8da0338a30ee7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0495f54875eccf95078c7944dd8da0338a30ee7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
