)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~778075","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I8f1407f895ffef7573e8b3b9c72514e2009da594","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"ovs-conditional","hashtags":[],"change_id":"I8f1407f895ffef7573e8b3b9c72514e2009da594","subject":"Fix conditional for ovs","status":"MERGED","created":"2021-03-01 21:23:39.000000000","updated":"2021-03-11 02:01:56.000000000","submitted":"2021-03-11 01:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778075-ovs-conditional","meta_rev_id":"f36e5d4b497319bcae2632dc31175b1245c001bf","_number":778075,"virtual_id_number":778075,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 01:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-10 15:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-10 23:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-10 23:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 23:59:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 15:34:24.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-03-10 23:12:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"610693f03dc98790f39c27bdfe3afbd60d9675c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-01 21:23:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c04807267afdd46c08a47275b38470916fa8a9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-01 21:24:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ea9d0f663d235477792b026404ca7170fda52fc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 23:59:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bbc67dd92c7e44f991c6513d770afda8 : SUCCESS in 12m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/34cc43f483eb448a95c04e792a47daef : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1bfcba3ead1548eebf79ae064a6f4f29 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d9819111eb54417b9739375de5bfd7c : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b295d93b2f845788d2e4cc1480bb0a3 : SUCCESS in 1h 05m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8b59ceefb3b4417db15c2e6e5e05c9f4 : SUCCESS in 2h 24m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f0ee5e5987a461383d78ac2de4d7db1 : SUCCESS in 1h 58m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/60d43e53a5334207881f578a3b0a05da : FAILURE in 35m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/769eb284fe3a4878ac3c9601d65541a7 : SUCCESS in 1h 22m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7d46d07a5547d5bd56ffee1f96a09f : SUCCESS in 6m 43s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b13fa5477df1408a9902fcb37e03d6e3 : SUCCESS in 45m 58s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/194d0e4cbebb4a6aa0d6fba0a925053c : SUCCESS in 1h 00m 17s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e5be5df5d18454590a0312b814db480 : FAILURE in 34m 14s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/c480e66a1b8e4cba8ce4b33cd690173f : FAILURE in 35m 57s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44102cd50bfb4a39b9abff21bd50c898 : SUCCESS in 1h 30m 22s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6167d1fc8f7451ab2c34aaa3a4ef9e5 : FAILURE in 35m 11s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eed0b6a2a4da4ff3b33559a691030e7f : FAILURE in 37m 17s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90f3b1658f8542af92ac18066ddcd47b : FAILURE in 34m 46s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba929f3e4d804f4e81b11c727cc56c83 : SUCCESS in 45m 45s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bca6fccfe2e947af95af282809996f8e : FAILURE in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f16b5d61b03fa1783018232908f77c79e278a3d6","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-02 14:04:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9868a2be9b5f14a5c3205a031e74f6c0bc7808da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:13:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/55c00763217d4bc89a87340d8ef6876f : SUCCESS in 12m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4c2f364a98c74e08a86f1011c1b3c07e : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6fc853f6815d4b9fb940f7e38ea98cee : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b985b11f4c84a469a36c47a18a3353f : SUCCESS in 1h 17m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a5999569d0d4b098a064545cd94f84f : SUCCESS in 1h 10m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/073731127b4d4ee383a2ca99968a7208 : SUCCESS in 2h 02m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04e15039d0c84888803ad46674402268 : SUCCESS in 1h 39m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/116dc057c01b40b697ba017c0beb08c4 : FAILURE in 39m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdac101fe18a497f9dc65e18e8bb3925 : SUCCESS in 58m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2409bd690364f2a84b29e500cba53d6 : SUCCESS in 7m 06s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bdedb8c03f146d59c15855d63951062 : SUCCESS in 49m 48s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/a1ccc9fc8d83471ea62938bc4c6bc7dd : FAILURE in 47m 26s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50e84f1895424a2d9b5c61a8b0412a04 : FAILURE in 36m 52s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/82fe101960a3425584af7873d1a6084a : FAILURE in 38m 42s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb64c1c86b6843019ae88996f0375bea : SUCCESS in 1h 18m 15s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7818a44e5ae40c9b2d872f3faaebeef : FAILURE in 39m 33s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee3f1ef6d8ad476998433b2340a3b8cf : FAILURE in 40m 12s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/329daa994ad641389cbee552c26cd600 : FAILURE in 39m 31s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47bbf8e7bed6428d99af6d68a7d081eb : SUCCESS in 52m 53s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80f648fde61242f78cac940dbbb958e7 : FAILURE in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"718988d9b2f6bdccd3aa667db16b8284ec949782","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-03 04:08:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f8771db17032ac311399e049202dac6ff2928cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 06:11:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1b8b05d8c71463899b514601e254c70 : SUCCESS in 8m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a289144d5bd7430aa5e2071d8e1eed5b : SUCCESS in 59m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9baf7c23a07d4f24863d794644d1ed03 : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90757828650549349e1cd018149b0fdf : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e2c65bd6984b65b03ef9dee5e0de62 : SUCCESS in 1h 04m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a3e21d86ea0a4e85893d7a518c94489e : SUCCESS in 1h 55m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30c2f88b789c49e29fdd93f7e0fa55a3 : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a73e28aa3a2e4c1b9b0ac34fbd140af6 : FAILURE in 32m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55174d5bb9d146d790d2c621d0cb2a7f : SUCCESS in 51m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9ced7786424becbf4a5fe8a52cd7a3 : SUCCESS in 5m 13s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee6e702f057c40b0b3e82ef2313b4f3d : SUCCESS in 49m 17s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/e4d158d44fb54da9b7934aed36e1be9a : FAILURE in 53m 38s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44d085518b5a40a5a671843994aa5fc1 : FAILURE in 34m 36s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/4e9e6dbd31d8402e84bb8d6f7eb5342c : FAILURE in 37m 05s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba8afb12ea7f43279044a658eacdc541 : SUCCESS in 1h 09m 14s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88e05286c67e4f079b5372f096a55dc3 : FAILURE in 35m 12s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2e65c83b25f448491ed5bc5ad3df46d : FAILURE in 35m 18s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b450c009750402bb5b95c49c237c002 : FAILURE in 35m 08s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe5ec2d13371472e880f0deba47f83af : SUCCESS in 44m 02s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/410aa42b4d104327b0ef9a3b6984b328 : FAILURE in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95371adae671c9721cc2fb68928d8614c7d5c563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-04 14:30:20.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d24ef938ca2490e6c5a38296eeaa498b2e895ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:39:06.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"174b5d6cee6906bb16ffdecee4f74ab115b6f79f","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-05 21:35:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"23e64b74911493a3b5f9426341f4c28ffe42c966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 23:27:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/227e5701e55b4781a1027c21040f832f : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83c404143bc4442fb3a70f08936f60b2 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d7a4ca9f1f149f3b0dcd06340f83e4d : SUCCESS in 1h 03m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da61a4b8d79d4732b49eeb9a95f9726c : SUCCESS in 1h 11m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6bc2a56cdbd40fb9e8c4f038d1ac0f8 : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/308512c80e1b456396a7a80290e90d1a : SUCCESS in 1h 38m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b2f3631d40e4e6b84540a949df62537 : SUCCESS in 1h 37m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/550e3cae63c8410493de4c2accad3ad8 : FAILURE in 38m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/004f8a322d2f400fae2741c7a9f64d3c : SUCCESS in 48m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d641500c28439eb54f9e7a2ce0432c : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63b4db45c2194e5fac869f050bd1c493 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1b9ef1852cb443796a8fdc716848298 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7691a4c84e52423ebcea9d8399f2f691 : SUCCESS in 51m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f75e18ea9b1041cebcb7807072390c02 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f40ca7bef92a4b62a512e8ebc9d2b199 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/89c3a1aa04044745a465f7ba2342f963 : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/dc7d506f472d45609b1c8afe97383b9b : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":3},{"id":"65c4c282f9c7d1afc0cc8516384d50a76dad864d","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-10 15:34:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cd625435598f4847dde6d8d3b1d0159d42c4e843","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 23:12:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"25333f5f5d42d55485a852e30c1eeced3fb2072b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 23:13:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bbc829b423cf350f725e0fa2e639726ee3314ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:55:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cf09eb30fb074f3b83d5e9d3f281b933 : SUCCESS in 10m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8aabe662f8341019bf38259f6ced149 : SUCCESS in 1h 10m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e70ea6ea068940b09e92578402654ce3 : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d23ba5fb7b994c5caca922b3410d2200 : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c93e12b263a44c2a7b49d4e20f45666 : SUCCESS in 1h 04m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/22380e86c08e4c928f4abf915992528c : FAILURE in 35m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c214fd8f7b8241e79722a7cc17b57a49 : SUCCESS in 59m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1249e0acb0b4a3bb69b93599131b6ae : SUCCESS in 7m 37s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e854ac56e3834e8f8da78b56fff93b8d : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51c37980b21a459aaee424728193de95 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee4297e32fc74c2e8c00a9f9f829b4fd : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/45063e7731e24726ac0ca6aa73a558ea : SUCCESS in 1h 53m 44s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/0c8f4adc6b754701a9b5c0528864ccc0 : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/61bf5ef9bfc348efaa5e8f66f9d09621 : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/9e5ca12207c34a0fb6ff61cc586fe307 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":3},{"id":"bc707b7af1fd571cae4228cc748d3fc7dfeddc25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f36e5d4b497319bcae2632dc31175b1245c001bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 02:01:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a4e78d0e9640b4807ce8320c45f7af : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d","revisions":{"a59735d44aedfe664edbd49d8b1f35dca98110cb":{"kind":"REWORK","_number":1,"created":"2021-03-01 21:23:39.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/75/778075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/75/778075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/1"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"subject":"Fix conditionali for ovs","message":"Fix conditionali for ovs\n\nwhen deploying ovs on compute the mapping might be equals to : \"network_mappings\": \"\".\nthis prevent the tasks to fail.\n\nChange-Id: I8f1407f895ffef7573e8b3b9c72514e2009da594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a59735d44aedfe664edbd49d8b1f35dca98110cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a59735d44aedfe664edbd49d8b1f35dca98110cb"}]},"branch":"refs/heads/master"},"e583e255a74dd5ad020445b3d95d79467848d1cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-01 21:24:04.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/75/778075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/75/778075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/2"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:23:54.000000000","tz":-300},"subject":"Fix conditional for ovs","message":"Fix conditional for ovs\n\nwhen deploying ovs on compute the mapping might be equals to : \"network_mappings\": \"\".\nthis prevent the tasks to fail.\n\nChange-Id: I8f1407f895ffef7573e8b3b9c72514e2009da594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e583e255a74dd5ad020445b3d95d79467848d1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e583e255a74dd5ad020445b3d95d79467848d1cf"}]},"branch":"refs/heads/master"},"55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-04 14:30:20.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/75/778075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/75/778075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/75/778075/3"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-04 14:29:51.000000000","tz":-300},"subject":"Fix conditional for ovs","message":"Fix conditional for ovs\n\nwhen deploying ovs on compute the mapping might be equals to : \"network_mappings\": \"\".\nthis prevent the tasks to fail.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/778008/\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/778015/\n\nChange-Id: I8f1407f895ffef7573e8b3b9c72514e2009da594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
