)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~779743","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I746e9c3246a257ca57c56b63c25a3cf9fcc49c9a","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa-cleanup","hashtags":[],"change_id":"I746e9c3246a257ca57c56b63c25a3cf9fcc49c9a","subject":"Remove references to unsupported operating systems","status":"MERGED","created":"2021-03-10 12:22:13.000000000","updated":"2021-03-12 16:38:42.000000000","submitted":"2021-03-12 16:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779743-osa-cleanup","meta_rev_id":"6b4f3739ca1a0ae9d855d67da04993920de798c5","_number":779743,"virtual_id_number":779743,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 16:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-12 16:38:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-12 14:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-03-12 14:16:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 13:31:30.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-03-10 14:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 14:33:23.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2021-03-12 14:16:01.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"f3c9faa59978f25d604c0b2a2191cd2d48ffd153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 12:22:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7340512563c20bd4081d91a8ce44357da4fd6c7","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-10 13:31:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nFeel free to also remove powervm :) https://review.opendev.org/#/q/Iea84648c7ff65b5a4d8b97957fb92716578f714d","accounts_in_message":[],"_revision_number":1},{"id":"a8b320e5c52e3b87adbb05ebc4e5f9df2c6bc5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:54:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/20fcb51b34f64c27a68b3782c4e1dd17 : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9b357f7fb1cd4dd3aa90740f32595fa3 : SUCCESS in 55m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bba767fa200e4131bdd8fb451a30e4e5 : SUCCESS in 1h 07m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd09d7a9502b4171893b5e8fb6e72cf5 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d28f17a81df4c3c80f5024bea17d47a : SUCCESS in 1h 03m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c7fe5cc429048a7b9f52a86eff61edc : SUCCESS in 2h 13m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f324db6311a4608841cf31d7e6d16f0 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3a4a1661e1c64b398aec57a32b56594c : FAILURE in 33m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c1bcb8b6314e34a4a1d95673f4a850 : SUCCESS in 55m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eaebbfe51bf49ae988beb5b7427bf40 : SUCCESS in 6m 51s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77132334ae2749979a3d43a2cdd423cc : SUCCESS in 52m 43s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/42363b8007ab4525b524ac4ea13b59f0 : FAILURE in 48m 43s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f960c584ac84e829011bbe9b4b455a4 : FAILURE in 35m 46s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/d887bef85c514c40a63a4fef29050110 : FAILURE in 35m 47s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d915ce4d99e4e8cbbd97f9fd083bacd : SUCCESS in 1h 22m 20s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be1e074f8c2b4d1b99ff068d846e6e43 : FAILURE in 40m 38s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcb322c7982b462fafa5b61917c39a3b : FAILURE in 36m 18s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d269f20b4670484c9d67adc1e991fb98 : FAILURE in 38m 25s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc17e011507a4330afd4321593ded86b : SUCCESS in 53m 19s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c7b0fb623a94b9985efc6b767b7a011 : FAILURE in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc4c0fad1cf572a800d51f153a6c51deae141e8a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 23:15:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"371c6c53b55ef44d5768214db96a0d8e3f282a81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 23:15:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bd53b61e01b916a7c3eee4cfaaa2f046b343944f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:27:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5056b65780d64280a5a367b7bd30c0ca : SUCCESS in 9m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b226038fb7d44c2b83371af279d016ce : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/41ee23328146496e85ad2f6a2c81c0bc : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78095ff065f743a883d564677041b533 : SUCCESS in 1h 15m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4bf9a2cb69e4b68a5575d63cc6124f9 : SUCCESS in 1h 13m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/45f32371cddb4a0cb0c54aa736978539 : SUCCESS in 2h 02m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff8855ac38b24e2ea3c9e855b6977f3c : SUCCESS in 1h 53m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/de52f4ef1f1b4de680d18a9ab3fc1607 : FAILURE in 38m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f1c7a9f83be4174af0d76b2cd789bfd : SUCCESS in 53m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a394dc048ccb47279fc44474c6017603 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39abf07f9e5843fd8f578f03b3829f20 : FAILURE in 1h 08m 57s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31fe91a27af74fffae714a79fbc50875 : SUCCESS in 1h 39m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5cf905e47064efd9924d24356e44693 : SUCCESS in 1h 05m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/baec54d21efe4b708d2aaaff0cd452bc : SUCCESS in 1h 52m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/d6d07610f4244df8a54241a6765a5fac : SUCCESS in 1h 19m 00s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c2bf98ef3f5f4fbb993abdab4f47f0e7 : SUCCESS in 1h 48m 35s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/75fec195e5a1435da99b69e867a0a66b : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":2},{"id":"664d7fa0340d840634f18e87e87fdf977cd90e81","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-11 06:39:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e62b21f8485715bec043a0ed2e9e9ebf30b7a214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 08:42:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce6cb6af70a74ebface930d8d248ca17 : SUCCESS in 12m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a8317a659b124d198a36716f447a31ae : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cf59a298dcd241278b927d63ca2706b2 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0025d37e3c9940ea883f5ffd401b7716 : SUCCESS in 1h 13m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fa37d606b4f477fb0483bccffe14e91 : SUCCESS in 55m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/006accc3e9be4ed18fb734e4d7282ee7 : SUCCESS in 1h 48m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea321a8802964364b947930eae8cb62b : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12c0fa82d78243209e0f4565a72c3526 : FAILURE in 37m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/062ed8b6527547e4ae69a91768544c9e : SUCCESS in 55m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac7f0cd35f6443485663f9efd644bd1 : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25153a4c61de4c6e8be8c199751f41e9 : FAILURE in 1h 52m 51s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04a87eac4f3647008a04575f48102f58 : SUCCESS in 1h 37m 46s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334c88720ee94923ab1cc8126e51e4ee : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/5807fba2ccc24763858721a058cb7494 : SUCCESS in 1h 56m 01s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/c29c35b74ce2454e8b0ce4f118857005 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/6b44e153990a44bc8841ab737da1116b : SUCCESS in 1h 49m 46s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/e6be8bae4a4e428d83aec8ab7c055029 : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":2},{"id":"dba19d78634381d8c45e9c73c94ef02711280ed8","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-11 14:33:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9620fb6186b9b32e2c44cb5107b5e7570ae61607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:36:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/53c0f269c3b84ece8763381a03b91ed7 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0c9c463ebc674f4aa49b731066dbb48b : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/909cd35fe95645e6b106acb51dd663ea : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5bdcfcbe9124dc7a28cff14509da332 : SUCCESS in 1h 20m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cf88cad29a409bbfe07bad4e2d1c3b : SUCCESS in 1h 06m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/20a67512c666408481642533d09e2dc0 : SUCCESS in 1h 52m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41cea5236fbc446796518b74ea6ca95b : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/10899852bd0c4178be743340b8a1ef88 : FAILURE in 30m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ba9be7ddc946ecb59c7285b4f4195f : SUCCESS in 54m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d698656f04a43f2b4ce2be761976a96 : SUCCESS in 7m 08s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/677453b8a2d241e7ae20306aa31cf1bf : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd6c1dd96d674800a676699692d3a066 : SUCCESS in 1h 40m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c3936d7c0b74c41812d193f1903f554 : SUCCESS in 1h 06m 05s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/45fe46a787984d7ba92780820f56fc2d : SUCCESS in 1h 49m 32s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/f95bb3085383465886927a62bee092ae : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/a515e03e699f4ff4812e64c106b8f9d0 : SUCCESS in 1h 40m 14s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/a1992d6695b944acad3562d5b598e968 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e35d0020f5079cd415897f04d13fa48fe571a9b7","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-12 14:16:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8bb7c3a9d6a846fbd7bb36e023d80589138779d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:16:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68f70e77b57d8fb21d3a26b70d958d971661b75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:35:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8a8b183c2224ebabb3658f163ddd924 : SUCCESS in 9m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a1878b20a8a8477eb8eeb6156d05308d : SUCCESS in 54m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c3e7455d5db2439cbd158a114da9d3a9 : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58b1a7386bd34061a459935ebbbac7b4 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcd5c114af4a472c8f24cc7654bcc76d : SUCCESS in 1h 00m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c84c34cf2cc488580bd2621be3769db : FAILURE in 33m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e12ddfab832e4363813ed7f266e9be87 : SUCCESS in 52m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0fd2cae99347638f43259be8d80906 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b9261f53e66439e8cfb066cd55fc477 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa2b4ab23a2741a1a6637912b31493a9 : SUCCESS in 1h 26m 37s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb02b5bb45449ec82fb564073ea256f : SUCCESS in 1h 08m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/b010e53525b843b59dc215c13ac34252 : SUCCESS in 2h 09m 17s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/421b8e48ffdb4b1f9d2b1d8a6bf06e93 : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/cc9411f7ce804273b1ade3ac40b5851e : SUCCESS in 1h 33m 51s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/599cb327e7464facabc88aa9b5d0b5bc : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c92c60c37535ef20e9983c7938c205074e3f743c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:35:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b8e6ac53208c7f56149e8a865e4edd5773fa4dba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:38:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f32bbc6c0643ba81cbf7ddff46a271 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d8c643eeb2c02fb8c450be71fff5f9fddcb75481","revisions":{"b5dc0cd83983c3adcf41c2a9da6a4c7f8c43ac8f":{"kind":"REWORK","_number":1,"created":"2021-03-10 12:22:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/43/779743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/43/779743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/1"}}},"commit":{"parents":[{"commit":"f441b212f5786f07472c103b36aa6224b19bbf25","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f441b212f5786f07472c103b36aa6224b19bbf25"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 12:16:39.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 12:16:39.000000000","tz":0},"subject":"Remove references to unsupported operating systems","message":"Remove references to unsupported operating systems\n\nAll references to Gentoo, SUSE, Debian stretch and Centos-7  are removed.\nConditional tasks, ternary operators and variables are simplified where possible\nOS specific variables files are generalised where possible\n\nChange-Id: I746e9c3246a257ca57c56b63c25a3cf9fcc49c9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b5dc0cd83983c3adcf41c2a9da6a4c7f8c43ac8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b5dc0cd83983c3adcf41c2a9da6a4c7f8c43ac8f"}]},"branch":"refs/heads/master"},"d8c643eeb2c02fb8c450be71fff5f9fddcb75481":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-10 23:15:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/43/779743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/43/779743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/43/779743/2"}}},"commit":{"parents":[{"commit":"69845d63a023f58beef172a03e48400ed3671742","subject":"Add integrated tests various Neutron deployment scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/69845d63a023f58beef172a03e48400ed3671742"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 12:16:39.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 23:15:59.000000000","tz":0},"subject":"Remove references to unsupported operating systems","message":"Remove references to unsupported operating systems\n\nAll references to Gentoo, SUSE, Debian stretch and Centos-7  are removed.\nConditional tasks, ternary operators and variables are simplified where possible\nOS specific variables files are generalised where possible\n\nChange-Id: I746e9c3246a257ca57c56b63c25a3cf9fcc49c9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d8c643eeb2c02fb8c450be71fff5f9fddcb75481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d8c643eeb2c02fb8c450be71fff5f9fddcb75481"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
