)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~779928","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fvictoria~I8f1407f895ffef7573e8b3b9c72514e2009da594","project":"openstack/openstack-ansible-os_neutron","branch":"stable/victoria","topic":"ovs-conditional-stable/victoria","hashtags":[],"change_id":"I8f1407f895ffef7573e8b3b9c72514e2009da594","subject":"Fix conditional for ovs","status":"MERGED","created":"2021-03-11 14:12:37.000000000","updated":"2021-03-18 09:30:31.000000000","submitted":"2021-03-18 09:28:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779928-ovs-conditional-stable/victoria","cherry_pick_of_change":778075,"cherry_pick_of_patch_set":3,"meta_rev_id":"730d3dcc82554985a430b598e7bffb74fe35c2d7","_number":779928,"virtual_id_number":779928,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 09:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-17 20:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-18 09:30:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2021-03-17 20:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-18 07:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 14:16:15.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2021-03-11 17:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 20:42:47.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2021-03-15 09:06:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-03-17 20:35:41.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-03-18 07:23:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"386b3d9030e8b8874933b338e01eb7d9bd90a952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-11 14:12:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0ebf4992dc43a04ee150443b0c25766d1c5c2f76","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2021-03-11 14:16:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd8bf98d98801a1076a7609fd997f61bbf1c01d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:01:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6183ddfef5ec4f1e982c0132d34d20ef : FAILURE in 5m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/461ce29c473d4f98a317733c93e062f4 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ff2b006af1dd4bbd9d34f14375d40e36 : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57ae87072b7f4b5eb633050fca96276c : SUCCESS in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18b21fb5105e4864b9cd2819cb7bb40b : SUCCESS in 1h 08m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dfd6cb236de4444eb6740bf4d9e84ebe : SUCCESS in 1h 57m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2424371f0c25497285de9fe44a698d86 : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2ab5c3e195714450924c899886d666b7 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/345ce01e64ca49c783c68c81d11692ec : SUCCESS in 56m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7149289fac24940bda3975320a5cd5d : SUCCESS in 6m 34s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38842c170a914fa6afb0847ca9ec8326 : FAILURE in 47m 52s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/16434ce6088a41f2b5f20c7d60f56da9 : FAILURE in 55m 17s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb2d23abe299482a8aa113e118ec1759 : FAILURE in 11m 12s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/ffa9fd7784514eef925ca4a8c93fd053 : FAILURE in 50m 20s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2560e2e4958416a8b9c315211a6918b : SUCCESS in 1h 13m 59s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b57219a918394ee8b7cbb628bf1f886c : FAILURE in 53m 08s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a059fef7d1d4a2897a5b35ec57b9cdd : FAILURE in 57m 45s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44aeb5f2249c40a68f6255aeb3c0ef23 : FAILURE in 53m 05s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d5174ad5bba4af5bd70ef204fbc1754 : FAILURE in 47m 43s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/478300caad9548c59928d96fed033127 : FAILURE in 53m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eca5c98b1c6a2c1f53c3bf6811f1cc8fc30eed26","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-11 19:28:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbb25402a412d104f9f688c38c6da3f9d154e1ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:46:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4061c8e6d20441f0a59c8ea549e3d5bc : SUCCESS in 14m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83e77b6d5407414fbadef4f7e234dea2 : SUCCESS in 1h 18m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e6794950f574e63b18d0f38e7aae98b : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f3d03a4d22492e8c4ce8efda2ab9ea : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e225e61a1449434782b2f9ad4e8ff199 : SUCCESS in 1h 31m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/65be8f20ba054f8597eee1eddfddf102 : SUCCESS in 2h 11m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6427538e59774abd946f6fb30021e17a : SUCCESS in 1h 47m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fed76058fcb347d89d30ce08d105d830 : SUCCESS in 1h 00m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/338a6d778db3425787b38bc5dc6ea37a : SUCCESS in 1h 01m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ecd236ef034527b68702028e4393dc : SUCCESS in 5m 47s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9d7deff25714eec8d3f135de58d3fd3 : FAILURE in 47m 37s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/6bbd69638e9f4b6591410ea5d443df15 : FAILURE in 59m 18s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/febc85c4735044739766647153bf61c2 : FAILURE in 47m 25s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/42523d4996f344518d9661668b6d7869 : FAILURE in 48m 38s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc8c5cac8d6d4fdf9b5c29fd5c42b43f : SUCCESS in 1h 21m 22s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efd4c3da012c4ca8b7c711432aecfd83 : FAILURE in 50m 11s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ae193c06a264be39bcc97f796744d99 : FAILURE in 54m 54s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fef520d673464ac38b59d52415386214 : FAILURE in 50m 40s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09cedad40d814af1aadb172955994a18 : FAILURE in 50m 37s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93d97b4b4e2d45049521c98fcbcc384e : FAILURE in 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06125791db07e4ef0a9674ef3b0de99650cbb0af","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-12 20:42:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e12a337be420a43d44a661f1404bfeb9da59fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 23:33:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/432bf04e3dae4de1a4e22fe68bf7ed6c : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7a47fcff2a434bc18483562d41653b4f : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4eab153a0e741b9afffc0aed8d6f179 : SUCCESS in 1h 19m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5467ad542d924ccf8290dbb035455ab1 : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d44de132df524136bce51fbbd2a55f9e : FAILURE in 52m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/506d659c522d4635a5f554eb3fae49e8 : SUCCESS in 2h 05m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bd64c50155e45629cef141e86b48af8 : SUCCESS in 1h 40m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/91a24f48bbc944d8a44cdb4053433aef : SUCCESS in 1h 02m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35bc8f07d3d4097bbdf7900521f98d8 : SUCCESS in 1h 02m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918898f72e104e778deb54840d0ba95b : SUCCESS in 9m 45s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ee5663e0d114ad5a34c836b7ff333b3 : TIMED_OUT in 1h 33m 36s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/44d06adedf534f12ba01b9c2fbd9240f : FAILURE in 52m 05s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47e243ca429849bfb916f120dc69859f : FAILURE in 40m 40s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/a877c84d0bba4fcf8ce19356f0c44ab1 : FAILURE in 50m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/148d242b88b6461d864b44b7f9b17429 : SUCCESS in 1h 18m 38s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e42d2f2b03c4224bb08effd9a3def8a : FAILURE in 52m 18s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce7e61bf6e1a400d9c2b9a3495500610 : FAILURE in 9m 48s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1d47c37935c4028b381cfb7b038e91d : FAILURE in 57m 54s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6c8b68450e24527add441ca298a2830 : FAILURE in 47m 59s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de43f00321c24c4490f2cbe4c19bc941 : FAILURE in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"131ea3f9d51ab260acfb780d0c85ff3fe9d693cc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-15 09:06:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab5ed6e1cd9167b1baf457296c4a1b4addb63e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:23:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d96d3495d1bc402c80ed338fd9856c40 : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/508e78018a1a4677b271c1f41c665639 : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e879e0951654ca5876c679372bee9e1 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fab97801d1b642a98fbc7e1ab722e67e : SUCCESS in 1h 11m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3877a1a9e504324b361f74df6adb2e1 : SUCCESS in 1h 21m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fc3d323ca702488baa0f1646b7fb041e : SUCCESS in 1h 49m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa733cf7c062454d87f7057b692d2cf1 : SUCCESS in 1h 31m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/254dbc6968d44affaac872ccf084b076 : SUCCESS in 1h 05m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58351d684d14f368d860a6f023b333b : SUCCESS in 1h 03m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3d40d7fca7431782d9a7e17468d5f8 : SUCCESS in 7m 41s\n- openstack-ansible-ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24f4822a47934cfa90737fedc432d78d : FAILURE in 47m 30s\n- openstack-ansible-ovs-centos-8 https://zuul.opendev.org/t/openstack/build/da248fee3cee45f09d92c73d6586ab7a : FAILURE in 51m 21s\n- openstack-ansible-ovn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7188004aa27f4403ae70e9ffb17142a6 : FAILURE in 46m 33s\n- openstack-ansible-ovn-centos-8 https://zuul.opendev.org/t/openstack/build/2d08bb2f88ef4ad8875257572bd09c6b : FAILURE in 41m 36s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d82168e64f04fda9ef90e574d3ec2ca : SUCCESS in 1h 11m 26s\n- openstack-ansible-opendaylight-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4c65e003a6a41eb967f9890be503d63 : TIMED_OUT in 1h 34m 05s\n- openstack-ansible-ovs-nsh-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24f42fd2f6e3441c98303194e81c01f3 : FAILURE in 52m 51s\n- openstack-ansible-odl-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ed2308a8aee422faf0ca22bacb922b0 : FAILURE in 51m 54s\n- openstack-ansible-ovs-sfc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f6278ff681423889a82a4395d6328d : FAILURE in 46m 02s (non-voting)\n- openstack-ansible-opendaylight-bgpvpn-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ccd1c68c0b44b259258a266725b7386 : FAILURE in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c55d33a6539745e2b9a8c643d9e820788c4d07c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-17 17:47:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e5e5ed7bb4d8b642c01f3cbecd64d9044097516b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:43:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2018d7e2c21f4d16bff444dee04e5fb2 : SUCCESS in 8m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3e6fe2dfb20742678411178031b4d95a : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4133e90c9dd94254b4511e68604ed7a3 : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b7782b45ee4746843059435aec1ec8 : SUCCESS in 1h 15m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/004fa052257a4a1ab75f6a61aef9a8f8 : SUCCESS in 1h 08m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a9f79a8af6a84e448a2663c6da408393 : SUCCESS in 1h 28m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69fa39ad996140a79f072f6c7f54431b : SUCCESS in 1h 34m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c86835d0eb0c43fe83449c0ddfb34bdb : SUCCESS in 1h 03m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f32cdd48a54cdb806de4b2b902df44 : SUCCESS in 1h 00m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc30643f80b044b6ac507a234c0cb697 : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a89293f877e487b81710fee1672f591 : SUCCESS in 1h 15m 21s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e255bbdcd8824f6e82bb916aaa942d9d : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-bionic2 https://zuul.opendev.org/t/openstack/build/d232b9b56097479b87bc07c1d603069a : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c89eecce1474a00bf658f4671a22c4c : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9574068037b84e06b3c57df0956453b0 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0049147780c44b21bda89e273193a028 : SUCCESS in 1h 50m 18s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/6c2252e3cf0f40c7b87c82ea9152a43e : SUCCESS in 1h 31m 32s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/d49d14601b224b3e9096f1ea05611fad : SUCCESS in 1h 41m 55s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/222635c538bb4ee0897ee0a479e57e35 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":2},{"id":"659dd66f5fbfde476fd3c299c83894568d7e8a59","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-17 20:35:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b618cf0799d3236a8aafac5d28193949f1785821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 07:23:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a15490d1acb0990a608611270ea070cbd0be398","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-18 07:23:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b50bed9c3de8ead172e8f06e7cb39569f54f7cb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:28:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fabd8ad4b8964aaa97ce7b62ea1f98e7 : SUCCESS in 10m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/208e9ab331b34b9cba398c21c04d60aa : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b9b7c15fb82340b7ac17c379fed30325 : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d54c00b3eb134946b90c95a293b6a61d : SUCCESS in 1h 08m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ac593a49d154613a63c31500cec8605 : SUCCESS in 1h 03m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/64378c6aa4ba466b8e0d43e068f69a8b : SUCCESS in 1h 00m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbea90f206b481fb01f26a3d455e27d : SUCCESS in 50m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90387c3998dd46f5971197337c3f279d : SUCCESS in 5m 51s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb1cca78e3ff4c249077c3c35dc87475 : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53fe67453eee4ed58ee9bf74ec3d7087 : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-bionic2 https://zuul.opendev.org/t/openstack/build/03763a613e134c4689ac5c550ce2611a : SUCCESS in 1h 13m 08s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c046f7076eb4dbd87cfae8aaf33522a : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9732942c676d4f748f90425dbf4539a1 : SUCCESS in 56m 53s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f2b3e9e6862e4d6196be90ac0ed5a520 : SUCCESS in 1h 54m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/3ea04e9e793c40d4ba1f73b35999a3f6 : SUCCESS in 1h 11m 56s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b97715ad01434b15af2c1f6e342c988e : SUCCESS in 1h 40m 42s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/85f0db378a5a409ebee2b1ded97dc43d : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2f1c0f9680c19b705aa88f8befd28e9f5d51c80a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:28:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"015a57ab215456dea98ed047612415f63bd99e2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 09:30:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc49bf4328f454dba4ad038ff9c1a6a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"596f7115e30ae813513637534a8ef5b0aa4684a3","revisions":{"2ae61fdb06fc4579bb0e51ac871dfeda22ffb7c7":{"kind":"REWORK","_number":1,"created":"2021-03-11 14:12:37.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/28/779928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/28/779928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/1"}}},"commit":{"parents":[{"commit":"f3eb23c7fa8a3fde67535cd8e36f15047cf22e4a","subject":"Fix neutron_keepalived_no_track default logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f3eb23c7fa8a3fde67535cd8e36f15047cf22e4a"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2021-03-11 14:12:37.000000000","tz":0},"subject":"Fix conditional for ovs","message":"Fix conditional for ovs\n\nwhen deploying ovs on compute the mapping might be equals to : \"network_mappings\": \"\".\nthis prevent the tasks to fail.\n\n\nChange-Id: I8f1407f895ffef7573e8b3b9c72514e2009da594\n(cherry picked from commit 55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2ae61fdb06fc4579bb0e51ac871dfeda22ffb7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2ae61fdb06fc4579bb0e51ac871dfeda22ffb7c7"}]},"branch":"refs/heads/stable/victoria"},"596f7115e30ae813513637534a8ef5b0aa4684a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-17 17:47:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/28/779928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/28/779928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/28/779928/2"}}},"commit":{"parents":[{"commit":"499660d736d37bab3a2662913991d2f2120f81cf","subject":"Add integrated tests various Neutron deployment scenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/499660d736d37bab3a2662913991d2f2120f81cf"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2021-03-01 21:14:47.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-17 17:47:20.000000000","tz":0},"subject":"Fix conditional for ovs","message":"Fix conditional for ovs\n\nwhen deploying ovs on compute the mapping might be equals to : \"network_mappings\": \"\".\nthis prevent the tasks to fail.\n\n\nChange-Id: I8f1407f895ffef7573e8b3b9c72514e2009da594\n(cherry picked from commit 55d6b6d74c93a11fe7e1f6aaa0ab045249efda0d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/596f7115e30ae813513637534a8ef5b0aa4684a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/596f7115e30ae813513637534a8ef5b0aa4684a3"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
