)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~798556","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I4d86bdfbecb16b3d2a4df01f06cedd17637d70b6","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"reno-wallaby","hashtags":[],"change_id":"I4d86bdfbecb16b3d2a4df01f06cedd17637d70b6","subject":"Update master for stable/wallaby","status":"ABANDONED","created":"2021-06-29 08:48:37.000000000","updated":"2021-07-01 10:06:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff19ee774c353c72a1e0062698fc4496ab63c7db","_number":798556,"virtual_id_number":798556,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-29 09:35:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 09:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0306d6fdd01be209a06fbb2fa29f9359728a4d04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-06-29 08:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e39532372720f4a9040bef79ea57650580c7898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:35:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5bde1a91a3b42fd98cbdd15ab3dd6b1 : SUCCESS in 17m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4efb6d74f90d42889e043161faba302b : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34467945e9d948c09d5cc89b6b633bc6 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ff19ee774c353c72a1e0062698fc4496ab63c7db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-07-01 10:06:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62644fd069b97ab9257389df9d64f0d6a249ebb4","revisions":{"62644fd069b97ab9257389df9d64f0d6a249ebb4":{"kind":"REWORK","_number":1,"created":"2021-06-29 08:48:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/56/798556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/798556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/798556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/798556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/798556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/798556/1"}}},"commit":{"parents":[{"commit":"61459c64bb28cdfff01c3166f3a0b79009b8d473","subject":"Add centos-8 support for ovs-dpdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61459c64bb28cdfff01c3166f3a0b79009b8d473"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-06-29 08:48:36.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-06-29 08:48:36.000000000","tz":0},"subject":"Update master for stable/wallaby","message":"Update master for stable/wallaby\n\nAdd file to the reno documentation build to show release notes for\nstable/wallaby.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/wallaby.\n\nSem-Ver: feature\nChange-Id: I4d86bdfbecb16b3d2a4df01f06cedd17637d70b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/62644fd069b97ab9257389df9d64f0d6a249ebb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/62644fd069b97ab9257389df9d64f0d6a249ebb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
