)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~802134","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ic482cda1616dd37f533e40513cb439ddf223cfb8","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/ovn-hostname","hashtags":[],"change_id":"Ic482cda1616dd37f533e40513cb439ddf223cfb8","subject":"Set ovn hostname using nodename facts","status":"ABANDONED","created":"2021-07-23 22:41:35.000000000","updated":"2021-08-09 20:46:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd5de62bc50a64e57c4a5e2a619cff177729a9bb","_number":802134,"virtual_id_number":802134,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-09 16:15:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-24 00:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-03 13:45:21.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-08-03 19:46:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"72263076fdd419d6411281a64744f3ba92d62d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-23 22:41:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d60b380092ceb6660c18ec79e687c815a7dbd24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-23 22:57:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"16223075e8e90e8e9e7ae00388bf8b506c4f5c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 00:51:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67a694a1246649cba19dd20b5ac137bf : SUCCESS in 19m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5d3ad84e0b284cfeb98061e99fc5a46a : SUCCESS in 1h 16m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b02e89f7ee9e4af1a993c11570df0020 : SUCCESS in 1h 14m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/585e23c24f404a47b0abe0fc7529c33b : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea3cb4228db54c5886c52c2cb395f248 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88550ab6df7d42aab44c3fb26ce95b05 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3906ad3ffe4a43e7b8dc7e6daa832ebe : SUCCESS in 1h 10m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b630c3c773b34bb79024d59f4dcddfbc : FAILURE in 17m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ddfe5cc11b43d780224ea2fecce8d9 : FAILURE in 21m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/71edcce2dcc340049e51b936d64f0d2d : FAILURE in 44m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8b341dfcccd4a228f621d53289a93da : SUCCESS in 59m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e943d5d33bf44838e29c58c06f96efd : SUCCESS in 6m 32s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec5320d4d6c342c3a8a2b0303d8486bb : FAILURE in 1h 19m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7affcfbd01ad4e2d93b07bb011ef9c97 : SUCCESS in 1h 48m 01s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd992997b8df40b68d958656f26a2aad : SUCCESS in 1h 06m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/996832f8684c44f693effa358c5cd6c7 : FAILURE in 1h 36m 18s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/88911a205eaa49fba45430380ae7c76e : FAILURE in 1h 06m 33s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/279bef821c7c4d40b31423382882b00c : SUCCESS in 1h 51m 54s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/4b39306bf9854a52ade26169f30c0800 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":2},{"id":"ab42134903c84d8e8a9c9756e85d9a05a030ce98","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-24 02:33:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0bf724b139bd85650b218d672c4cb21d645f12bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 04:29:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1eccab14aa94f609395d02e93968cf3 : SUCCESS in 17m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c45d46d687424f83845958c951510eca : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6c37f5367e1543249b950228c4617e85 : FAILURE in 48m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0633d957b29b4ce991fe9481764028bb : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/726b8c6ce6ba4583ae6bdc21ac750f21 : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53712c3ecf0c4099a63590fe638435cb : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aca0b00c97b4461493865572a8880079 : SUCCESS in 1h 07m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1de021d095bd4dbd8f030bfb822b8f56 : FAILURE in 15m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88082f5f1fe6400385c5682ae947edbb : FAILURE in 21m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6d41849e75f4ed1b6e1e35477c6328e : FAILURE in 42m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/605699119a98487cb582060d32edf3f6 : SUCCESS in 1h 03m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00c208072a140c786f8931233fef37d : SUCCESS in 9m 09s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b1ccfd8563406aaa9cbdc9e8de6cb1 : FAILURE in 1h 22m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67538c3042dd408ba1ea0e072d4aa259 : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ac322363f74f5db50bd16b9f3d026f : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f0753893b2c14b03891ce54bdb790f89 : FAILURE in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/71ceb9e493454e33b8e97396835ed8c0 : FAILURE in 1h 02m 22s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/e9979209b62c41919e6c42c8d7a06842 : SUCCESS in 1h 45m 22s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/4c4e12841112448282c958946664d137 : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d7ddd92028114e9ba059a8caa5885f30cffa2b02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-27 12:58:45.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9e847c06348a58c8ea60f2db4c04a9dcbbad6252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 15:04:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cdb7cdc36404694abcedac73704f980 : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eea252fb1d9740b9b7749c3a32902418 : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c7527e0f71449e89ac8b8bc41860ad5 : SUCCESS in 1h 16m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ded6ec0e03584e72ad8a178577617a32 : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7949ca7aa1504dfcb79342dd5a5dcff0 : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c53464580766440fb35dbc6a0700da68 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a649116abde043ccb4974bbbda94e82d : FAILURE in 45m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5d183f3d0bbf4c6a94882da3c66701c7 : SUCCESS in 2h 01m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59f1b5d2a97e4e08b735525d5cc2a689 : FAILURE in 45m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/373c2eb19938433689770e9dba7414da : FAILURE in 42m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6207e4a598e94390a6bce2ef62b34ed4 : FAILURE in 47m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc2386ca8b245a4b7d6781ebededbf3 : SUCCESS in 6m 39s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ebd40d3110c40ebabc6baf2bc7ee3e3 : FAILURE in 1h 22m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74aec5fd6d274bf08938262bdac39021 : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfdd73e7bd8844238eadd0b622780fb5 : FAILURE in 43m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/10d76ad760024f87b90c890c257b7756 : FAILURE in 1h 40m 14s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/9861950476384d9eb63aa7acf2418306 : FAILURE in 1h 07m 41s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/9eb3613451f74a2faf4f83ef97d8fdd9 : SUCCESS in 2h 02m 28s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/1432d71a43c44c3aa04e9c5a5283b732 : FAILURE in 42m 16s","accounts_in_message":[],"_revision_number":3},{"id":"193e3a6b648200758d24a1b4ca393e144b5dc85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-27 16:17:11.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3b3e4dce0e5e642cc9524b9beae3b723ebde7c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 18:55:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce9c70965cc54db3b8bb79e5407259b7 : SUCCESS in 18m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83c3417da77a4602ab570d65dfa864b7 : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/22931cd9af654cb0bb83b82ca42d8e28 : SUCCESS in 1h 17m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6d50bfecd8924daf8ba019ea40bf5906 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/350ffcc0718e4f8ab1d69af3e14c2008 : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2473af05f7fe4df0945ebd16a27aad25 : SUCCESS in 1h 23m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31883b7cacb94e7da7f642bd73359001 : SUCCESS in 1h 16m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/18307d8434174f30ab3a8e36f257e0ce : SUCCESS in 2h 31m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e0256a2e04d4052b210eeb84fb21446 : SUCCESS in 1h 46m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f838192a40df4d73a2168473c2a2ed92 : FAILURE in 44m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27bbe158ee0e4a5fafe347ccc87cedd6 : SUCCESS in 1h 03m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a022611925448f285d01a4148d33dba : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6c1e881c9a04cbda0865a3f3b710d63 : FAILURE in 1h 26m 42s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aad9f5e01ff546de846765e5d312cd9d : SUCCESS in 1h 46m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36e71be1aa0f45df82ee2e76595b9616 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/f51f8ab6bc5346c1ad0099e15122f534 : SUCCESS in 1h 57m 11s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/625f4a22ade347a9855da9108041b72b : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/8a91a66711eb436881b37ca73604a956 : SUCCESS in 1h 58m 20s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/66a28cf920014cafa8a2a9fb16ccea46 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9cd611999e216deaf917d0f330c6d7c7c41d9e48","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-28 13:26:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f47b87acd84d02a3aa69f2ba0c88bc2cd1e6ee0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:34:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/195e367c956b43c586c15998bb5c543b : SUCCESS in 17m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d38b9f29fe44491e932ab073ed3e17b7 : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1a54acd8355b4ce09aac9c005c021b38 : SUCCESS in 1h 14m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ede384b5ba8d42308aaf3ac8fe485e14 : SUCCESS in 1h 51m 22s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6af27f9ccf4f41f5adaf1afd4a05ef6b : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/749b37eceb1d461e97ddcaa88d2757dd : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87c0eaf2ec9b4283af4cae77b987da08 : SUCCESS in 1h 06m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fcfce37ee45a483aa3807afbb65f7ffb : SUCCESS in 1h 39m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc6714bdb5943b486cf4c526f3f622c : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f60086f015b4b32b42f90df9a22d4eb : FAILURE in 45m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b21a7a1dac8f4d2980c1083e120bdb46 : SUCCESS in 51m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3725826efaa46caba5f55396e420645 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/626528c99340402eb4a4cf786e497207 : FAILURE in 1h 22m 00s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a1a2e5371748a2a3e99561191e1dde : SUCCESS in 1h 41m 47s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20cb708f23f541778d49192c53c597cb : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/aa8e271e37cb496c9ebe690576d6fd23 : SUCCESS in 1h 59m 58s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/ad22ab89fa63460a89ae20322abc2ce7 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/548918aa1f044bd4b612883f6a46aa12 : SUCCESS in 1h 47m 22s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/5808f864992b446c935bad5ec79b7f19 : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":4},{"id":"cc73db0e6c6f1488ab18bed74ff2f3596b936223","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-07-30 19:14:05.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6794a159ee4581576c7a5ff0a24fc0a79377db94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 21:15:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2874f43e0d9549d5b89f472175103ddd : SUCCESS in 14m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0a95bef0461c46c48179fdb5b84abb7a : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/80d199ba4aaa4689bd85d457bf7ee1f7 : SUCCESS in 1h 28m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1eba1fb4fadc41b28b42a6589fe8cf41 : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a6f2070a6ab420ba308fc295770df0e : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84634154845d4e038e09d116e33c463f : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b195d42d8f914b05a96275b688858b66 : SUCCESS in 1h 13m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d11751e3fd94e5bba37586c20e14f10 : SUCCESS in 1h 46m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9592dffe9f4242dd829e26108d2fb613 : SUCCESS in 1h 43m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ae62d1376a143d1b4b498db45a302af : FAILURE in 41m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3f5717124fb491d82173d875d77d4ca : SUCCESS in 1h 07m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44412537b38346e1b4d4017a3ca7f996 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10eb9e907bbe48d3a4d67ed9eff57b7e : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03d7ffb5007644dbbf9ebca4eeeb99d5 : SUCCESS in 1h 52m 31s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc6b797f6ea84b21b875564d24b79f3f : SUCCESS in 1h 07m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/68e2c6fc1e0745d4a62787ab671684bd : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/596ecc918b184b7db39a48cb53304514 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/36c278864ec248138e1a7875179d1180 : SUCCESS in 1h 51m 57s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b3b1aa1d3b6b495fb76425323d0afe0b : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":5},{"id":"590de04750dbd3836059213d233aa51c498b3433","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-01 16:17:44.000000000","message":"Patch Set 5:\n\nLet\u0027s have some review and voting here to merge this patch","accounts_in_message":[],"_revision_number":5},{"id":"7ccb2a6f111e0e7bb2fbcc38d6fd5012c29b513e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-08-03 13:45:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cb2b00b23e0e9f111b73741155df8ad4e0e79d28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-03 19:46:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ad3ddea6740af263b337b6e677c8215cda505487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-09 14:15:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"64282eed48f533603cda262abc058510e904b1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 16:15:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/174a80fa95c54785983fcec862fc6857 : SUCCESS in 16m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/89943dc95e2c46d9859a9a16fe739242 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3d46cb710dee4f40a18c8ac04f0899f7 : SUCCESS in 1h 18m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a0e449d38a2b4d01b6838c820be28b7d : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/80a9ec592c04404292eb9d1f71fe9990 : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e3d5c2f6c7b4ec5a22d2b61185aa409 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a09895b949c24ba7bb4edeae340e2be3 : SUCCESS in 1h 21m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c143e0a29a94eb2801fa0c0b796069f : SUCCESS in 1h 56m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa63a52849bd46979380291010b7c28b : SUCCESS in 1h 47m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d198b0b7541e4cfc9cca750905ea2bb4 : FAILURE in 47m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a2b523438bf45adad4088633669a671 : SUCCESS in 59m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053b6686a8c14a5a8da9b3e7c18fca3f : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4743098476640b9b0ef2a56c61ade98 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12bae99e65254e6cb2bcb6e64f5444a1 : SUCCESS in 1h 36m 53s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33178cf8b3014657a1ba408d484d2921 : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/40c5fe2272b247f9879bd119139a5b4a : FAILURE in 1h 40m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-8 https://zuul.opendev.org/t/openstack/build/cba3061108a74a51afd24a1f2fd73861 : FAILURE in 1h 06m 05s (non-voting)\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/26c996fb0c6a4bf091b6664530c78efe : SUCCESS in 1h 54m 20s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/6505bb1be14b43b89566fc00929bd16d : SUCCESS in 1h 14m 37s","accounts_in_message":[],"_revision_number":6},{"id":"fd5de62bc50a64e57c4a5e2a619cff177729a9bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2021-08-09 20:46:24.000000000","message":"Abandoned\n\nWe don\u0027t need this patch because replace with https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/803988","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"29b9d54834462c8fb301c9c3bd5348a9b6607716","revisions":{"17a94c3b666a85ec66e73ffe4232bb56e20722bf":{"kind":"REWORK","_number":1,"created":"2021-07-23 22:41:35.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/1"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/17a94c3b666a85ec66e73ffe4232bb56e20722bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/17a94c3b666a85ec66e73ffe4232bb56e20722bf"}]},"branch":"refs/heads/master"},"e11ef8ba24a5310b9c425d8e02af0d7be43fa2cc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-23 22:57:20.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/2"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:56:58.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802135\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e11ef8ba24a5310b9c425d8e02af0d7be43fa2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/e11ef8ba24a5310b9c425d8e02af0d7be43fa2cc"}]},"branch":"refs/heads/master"},"a31384cdafe8aa9d78f5c52f2e5cde2524734b45":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-27 12:58:45.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/3"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-27 12:57:54.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802135\nDeoends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802402\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a31384cdafe8aa9d78f5c52f2e5cde2524734b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a31384cdafe8aa9d78f5c52f2e5cde2524734b45"}]},"branch":"refs/heads/master"},"a830396262fe979cd37cfb1071b10a7f5e717e67":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-07-27 16:17:11.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/4"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-27 16:16:48.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802135\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802402\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a830396262fe979cd37cfb1071b10a7f5e717e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a830396262fe979cd37cfb1071b10a7f5e717e67"}]},"branch":"refs/heads/master"},"f46997c04525fdc262048b7a9f6e40dde76a6146":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-07-30 19:14:05.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/5"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-30 19:13:17.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802135\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802402\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/803041\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f46997c04525fdc262048b7a9f6e40dde76a6146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f46997c04525fdc262048b7a9f6e40dde76a6146"}]},"branch":"refs/heads/master"},"29b9d54834462c8fb301c9c3bd5348a9b6607716":{"kind":"REWORK","_number":6,"created":"2021-08-09 14:15:40.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/34/802134/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/34/802134/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/34/802134/6"}}},"commit":{"parents":[{"commit":"ffb4a81461facb02ed75567e46a98ea069d548b3","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ffb4a81461facb02ed75567e46a98ea069d548b3"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-07-23 22:39:48.000000000","tz":0},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-09 14:15:24.000000000","tz":0},"subject":"Set ovn hostname using nodename facts","message":"Set ovn hostname using nodename facts\n\nansible hostname facts using short hostname which may not work for\nother senario where hostname is FQDN. This may help in that senario\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/802135\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/803041\nChange-Id: Ic482cda1616dd37f533e40513cb439ddf223cfb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/29b9d54834462c8fb301c9c3bd5348a9b6607716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/29b9d54834462c8fb301c9c3bd5348a9b6607716"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
