)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~804449","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fwallaby~Ia9189adae01d8515e392abdbede7fd7b3f89e02e","project":"openstack/openstack-ansible-os_neutron","branch":"stable/wallaby","topic":"osa/neutron-ovn-stable/wallaby","hashtags":[],"change_id":"Ia9189adae01d8515e392abdbede7fd7b3f89e02e","subject":"Fix OVN metadata protocol to point https","status":"MERGED","created":"2021-08-16 10:27:59.000000000","updated":"2021-08-20 23:18:35.000000000","submitted":"2021-08-20 23:16:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804449-osa/neutron-ovn-stable/wallaby","cherry_pick_of_change":803989,"cherry_pick_of_patch_set":4,"meta_rev_id":"e05a586b545bd92fc387bcd607bb13c06ba85841","_number":804449,"virtual_id_number":804449,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 23:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-08-17 07:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-20 19:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-20 19:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 10:27:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"},{"updated":"2021-08-16 12:45:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 19:19:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0bb7b7adbabf709f50d757752c5655512f6cd94b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-16 10:27:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"760f8df42854f324af13a800c0c12de6515cffd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-16 10:28:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"231a999b0523182c863a0d7d5d424f9cd9340605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:45:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b617b3acd7884464922c469d50987287 : SUCCESS in 18m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e593a6f753c84c45a2f029bc9f2d7064 : SUCCESS in 1h 13m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/56a01ade7dae4e41910d49d6ff6b582b : SUCCESS in 1h 37m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5b88d352864340189d06d35f08f041b4 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/246cd6ff9f2c46c1a499d4840ccdb9ef : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/900810e646954dd29f4f9782a0fa7aba : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f2c3f27a2d4f898961b756800e7ff7 : SUCCESS in 1h 10m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98b77eed897343fa9e52897c52963ce5 : SUCCESS in 1h 59m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c644930de424943a147f9bc4c76fec3 : SUCCESS in 1h 35m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a9adae1ab554304b53d1d0ceecb9cd3 : FAILURE in 43m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6244c10af5a64ca9b273926a2b54c513 : SUCCESS in 1h 07m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a24244da1f848d4b616d6a870e9d90b : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea10c02c7a4e4c5b85c2eb02ca32ca44 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b6c06daf79c41d485d5d6a54275c314 : SUCCESS in 1h 45m 13s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acf71dd3b8f14e338d6cbce5c48fd143 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb80405c83f8436981cdb3a58859e69f : SUCCESS in 1h 59m 45s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0c270286c8a48e9bc4e48c409435286 : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/beed513922634ee6b714e38aa3d06fe2 : FAILURE in 1h 45m 34s (non-voting)\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa38e69272144755b1f9cee97fdce0d3 : FAILURE in 1h 11m 23s (non-voting)\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/9dc6643e3fe648098b56db945f8fff15 : SUCCESS in 1h 50m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/95e4af497c684663bf20c18eab5648f7 : SUCCESS in 1h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0ab76f11d23f5afdc3662f6ada00d2e0c7141c1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-17 07:32:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf1d7ca6866d20b7d07f0b130d310c42c5c46680","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-08-20 19:19:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"418b567b74cfe044c139f8b4f39075d57bd1a24c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 19:21:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d254a3d4c1efb0b08027332a30740792cbb79a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 20:22:32.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6831a81ad5d0283e62cd50ff442c1a2cb1e19b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:16:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4041c40bf9584641b7a0f3507f5f949b : SUCCESS in 16m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fe56791c1eaa4697aa58fb0956262d50 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/be550d366f784eb3b41cc3fdca8d6bc5 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/25c4818847584d1cbca20694091bddbd : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3182f6ed2b4942afb1204ae76508cd2a : SUCCESS in 2h 03m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdc045e9f6c541d4b0bd40bf57713726 : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8ef9872b3f94be0b0a8b99c036ffb61 : SUCCESS in 55m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891f2b1bcfbf4bf4b76aaf9e9698a0c4 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5844e8cf2fe741689d07d65df36b9527 : SUCCESS in 1h 19m 54s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8abd496aed15478daae98b0d5a02416f : SUCCESS in 1h 39m 01s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d6cf1c03b3747f5b8ca96b92de67fd5 : SUCCESS in 1h 04m 06s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6246e112bbe341fe874f9506712bd539 : SUCCESS in 2h 00m 33s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/efd6a1f269d242e6bf7e24c689f44b8d : SUCCESS in 1h 19m 21s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d997664cb3a4e1d8fe0ce8653daabb4 : FAILURE in 1h 50m 05s (non-voting)\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6434f2aebe0c40e5b6a909f466d5c01e : FAILURE in 1h 11m 37s (non-voting)\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/ff44b2da99ef4916a24075ca9cbd5901 : SUCCESS in 1h 47m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/317be12ba7b24b7ab81e2e9ede0612c2 : SUCCESS in 1h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"093ef9ece190b7144e51787ea3340aa9e019725e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:16:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e05a586b545bd92fc387bcd607bb13c06ba85841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:18:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360116bad4494cafab0721c932aa906a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6d9c54a921c1e5f18d5ce300e0cbeb73ddfcc49","revisions":{"c8a33bda1ea1f80fd24a714a64188d72979b54d6":{"kind":"REWORK","_number":1,"created":"2021-08-16 10:27:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/804449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/804449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/1"}}},"commit":{"parents":[{"commit":"29e5fbf252074bd8cefb875911d88f58af9cb205","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/29e5fbf252074bd8cefb875911d88f58af9cb205"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-09 20:26:02.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-16 10:27:59.000000000","tz":0},"subject":"Fix OVN metadata protocol to point https","message":"Fix OVN metadata protocol to point https\n\nIn OSA recently we moved all our LB endpoint vips to https or all SSL\nand that break ovn metadata which was default using http. This patch\nwill fix that and allow us to override variable.\n\nChange-Id: Ia9189adae01d8515e392abdbede7fd7b3f89e02e\n(cherry picked from commit e5e3ccfae6dbbb164802d3b50c806783c3bd7362)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c8a33bda1ea1f80fd24a714a64188d72979b54d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c8a33bda1ea1f80fd24a714a64188d72979b54d6"}]},"branch":"refs/heads/stable/wallaby"},"f6d9c54a921c1e5f18d5ce300e0cbeb73ddfcc49":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-16 10:28:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/804449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/804449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/804449/2"}}},"commit":{"parents":[{"commit":"df35816daa06ab5c5e832758f091afebad85301a","subject":"Set ovn hostname using nodename facts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/df35816daa06ab5c5e832758f091afebad85301a"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2021-08-09 20:26:02.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-16 10:28:09.000000000","tz":0},"subject":"Fix OVN metadata protocol to point https","message":"Fix OVN metadata protocol to point https\n\nIn OSA recently we moved all our LB endpoint vips to https or all SSL\nand that break ovn metadata which was default using http. This patch\nwill fix that and allow us to override variable.\n\nChange-Id: Ia9189adae01d8515e392abdbede7fd7b3f89e02e\n(cherry picked from commit e5e3ccfae6dbbb164802d3b50c806783c3bd7362)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f6d9c54a921c1e5f18d5ce300e0cbeb73ddfcc49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f6d9c54a921c1e5f18d5ce300e0cbeb73ddfcc49"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
