)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~806084","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fwallaby~I82c2c35acbeb55ac5ae15121acbe457347de5f49","project":"openstack/openstack-ansible-os_neutron","branch":"stable/wallaby","topic":"osa/neutron-ovn-stable/wallaby","hashtags":[],"change_id":"I82c2c35acbeb55ac5ae15121acbe457347de5f49","subject":"Set OVN jobs to voting","status":"MERGED","created":"2021-08-26 11:34:22.000000000","updated":"2021-09-08 18:02:02.000000000","submitted":"2021-09-08 17:57:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806084-osa/neutron-ovn-stable/wallaby","cherry_pick_of_change":804324,"cherry_pick_of_patch_set":2,"meta_rev_id":"c12aadc8b93f3c6bdc29adc808d8d5ec5c464766","_number":806084,"virtual_id_number":806084,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 17:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-09-07 19:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2021-08-26 14:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-09-07 19:32:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 14:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 19:32:56.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"9d0e9663857b748f1a5729fa26fdaa034e87568d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 11:34:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ac682bdce632d34a9785c158b170e2e03384738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:14:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/444e5ccbc7a143b8bcfb9ac642b886fe : SUCCESS in 18m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ca7d3eb81af84b8ba910aa0c0ce01c95 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ed1f2d128ca44332a8eba3bf5033f0f0 : SUCCESS in 1h 20m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a0025ec58664411a721c098b734ae17 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d692cad4faa34c1496c724810d22f032 : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6788294f93524ce5998ffbb910ea4f36 : SUCCESS in 1h 33m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53562fa7c8f44b3695f49bcadd24dac8 : SUCCESS in 1h 10m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2f1dede004b94f80b8abf6a38ff62699 : SUCCESS in 1h 55m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38e854b2d19d4ff2bef31e95d5e53883 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d89e6c0d774495e912f0a76b31921a9 : FAILURE in 43m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/defe75aa400144efbcc7ce193e806787 : SUCCESS in 1h 00m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e095259827084c58ad10149368e219af : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a114e86938814fd7b2ca031f51038ae6 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2c1e6911b3e40b9a489339842de0602 : SUCCESS in 1h 52m 34s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/110c7eacf7d44a498e120ed21726f45b : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/810b406fed15474fb08ca28b7ca0f4ba : SUCCESS in 2h 01m 09s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2fd1625256b4029bff6951414899f09 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a64005f47802451f933581acbc11eecc : SUCCESS in 2h 19m 11s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8918160ea9ce44159ba1f32b6dd5c88f : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/738e500971e341afbf2d7d87de366957 : SUCCESS in 2h 32m 55s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/498b2c6e0fad4256b84da4c8a9764953 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ce3d2e7e31213b8d7129a04483ffb0d6d75a6dd4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-08-26 14:15:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f5cfa39408ccc0e0ec7cd1741c849fcb48c3f99","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-09-07 19:32:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"829362f3ba490b32c691ba540c847c83c326fdb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:36:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24da220ea83a87aa10cd9870d517069d085443cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 21:30:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f17f79727984c838b7514aabc9a2457 : SUCCESS in 17m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/83d17a54b262485aa1cc266d5f72051d : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/935247e6e4b84e088219f5047bffe93f : SUCCESS in 1h 29m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/480972a6980349d7adf2cb022178b816 : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbe15d092c1a46beb0ba19e180be337e : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621ae03b42494291b94de1bb86497d21 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/168c289b1ee248d784c5b8eaf6ae3580 : SUCCESS in 58m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50439e3801f74b59a8130fff89fa3739 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842e72b5a2664f33ab77f0e0531a7a53 : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bbd778421b84662a6e61ddbd9bd18e0 : SUCCESS in 1h 45m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24bf0ce39a034722afe8a99b3fa5d956 : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ec9d11c7a8c48d08d05fdd6df7393a5 : SUCCESS in 1h 53m 44s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/e17cae9931cf44158176baab693dacd6 : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0d1b452aae1413e835d24a830154618 : FAILURE in 1h 51m 39s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e6ab91284f249478abb48916de6fa9a : SUCCESS in 1h 44m 36s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/d809f43e39cd41b68d8820217710afa4 : SUCCESS in 1h 47m 07s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f0667439e46c42b68858fb26276f698a : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ad8839586d720df671919e0b64eabd29967c864b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-08 13:40:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8aa898ef1036266430482ec49fcb0c31eceea402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:44:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84ea2170e2534b2481e6afd21108161e : SUCCESS in 16m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/500fc7c3a03d4cfd9d62a692bf54aaa3 : SUCCESS in 1h 24m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5998d099633948c9a88f0bc0536034f0 : SUCCESS in 1h 07m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a3b8a11d368941388c8616d7fe2c95de : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83b1165ee2ec45fc8023f12d89ce770a : SUCCESS in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f5db18d45494ecd9491603c9e5a729b : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ab7f2abde3645af93f28086a4090cd5 : SUCCESS in 1h 10m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7f365beffc9a41ea8ccff72e85ac89b9 : SUCCESS in 1h 50m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8af832f988ca4cdbad0bafbd5972a2c9 : SUCCESS in 1h 42m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f833a808d70147958ca4088905ecfa65 : FAILURE in 43m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f495c13f9f4e495499f2a139b2e1543b : SUCCESS in 52m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528d1c6a3b9e429e9eac7c281541f538 : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f62160d627e428ca8edd4e65fbfae4a : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eafac709ff6248cd8baeeff5b9d46921 : SUCCESS in 1h 44m 16s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e72cab6e92b46b3ae7b91021f8865e2 : SUCCESS in 1h 06m 31s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/95a19962dadd4121833950343dc83516 : SUCCESS in 1h 58m 13s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/25a853171fc94cf1a171e51320420f59 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/654f87dec6d04608bc8357332db437a7 : SUCCESS in 1h 58m 57s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6033685cd1f4693b8e5d3321fdbf17f : SUCCESS in 1h 17m 13s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/dc2a0f1703ec4072b5c46421d7a6fa2f : SUCCESS in 1h 42m 58s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/eb01899bee7f46fb8083920564dbb815 : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0700263a0c3cffa7c3368f85db9febe0ecbd9110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:47:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9df1df208b0b498e0bc36a270c8ce5104a1fa8b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:57:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/915c12cbea694b9db5cdb2d2b8f191e2 : SUCCESS in 19m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0369bedb326847dd829bc17adb51c9e4 : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f95ec91fa26b42b4a39b645fe6a4b750 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0791da55ea6e45cfbba80af336b1b4b4 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d413594740f4aa2aa77bfce24d9d7d7 : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88aebe3fdd8b497a90df1a6d3fc5c7f8 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c644866e2c34cd0bfe9aabc83fc8d2f : SUCCESS in 1h 02m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e8a095274a4b0bbc17525b514057e2 : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b37db3c56bd4a889b95d23dd472f38a : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef03c6073f0e4eb7915a074782acd665 : SUCCESS in 1h 52m 26s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2e198c0d96c4bb9aa7c72dfc3d61199 : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d3c563818c54c53bc7ab85465bb0cf5 : SUCCESS in 1h 57m 50s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b2cf76c204748e285b3812547556a28 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd24abd22d4c42a987b241f1f0aada4b : SUCCESS in 2h 04m 46s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/27ad19afa36f481e82e4ee0fda9d263e : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/d68f2ee4f07948b4af534bf1462f2b87 : SUCCESS in 1h 39m 07s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/09e1bd392ab04e998ba41bd6bb652b27 : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"db455bb0ffba799c45620afb1bfd47d4989fd804","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:57:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c12aadc8b93f3c6bdc29adc808d8d5ec5c464766","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:02:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0805c7e8fd224f219e92a62db85eea4a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb7c10b090afadc191fadcce5c93035de2c9f55c","revisions":{"eb7c10b090afadc191fadcce5c93035de2c9f55c":{"kind":"REWORK","_number":1,"created":"2021-08-26 11:34:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/806084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/84/806084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/806084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/806084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/806084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/84/806084/1"}}},"commit":{"parents":[{"commit":"40e99aea192b50192ea888ad2640f33ad98da00f","subject":"Add support of OVN for CentOS-8-Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/40e99aea192b50192ea888ad2640f33ad98da00f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-08-12 03:50:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-08-26 11:34:22.000000000","tz":0},"subject":"Set OVN jobs to voting","message":"Set OVN jobs to voting\n\nChange-Id: I82c2c35acbeb55ac5ae15121acbe457347de5f49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/eb7c10b090afadc191fadcce5c93035de2c9f55c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/eb7c10b090afadc191fadcce5c93035de2c9f55c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
