)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~817913","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ibbd537281c716a1353fc20768ec1eb20e97edb62","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-11-26 15:48:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbd537281c716a1353fc20768ec1eb20e97edb62","subject":"Set default for neutron_local_ip","status":"MERGED","created":"2021-11-15 08:35:35.000000000","updated":"2021-11-26 15:49:53.000000000","submitted":"2021-11-26 15:48:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817913","meta_rev_id":"94a8f22a1dbda2ebc26c3a9e0b8c83769f60c705","_number":817913,"virtual_id_number":817913,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 15:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-11-26 13:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2021-11-23 22:34:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-23 12:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-11-26 13:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-15 10:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 15:07:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-11-23 12:53:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-11-23 22:34:48.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2021-11-26 13:26:29.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"6d1ef035c31aed989cf44372cfd2600f5ea06390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-15 08:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52cbafd41c1bd91e6ac3ef7ea070ad1120342a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 10:45:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8b33d66b6b44f10a73c8b76dab810b7 : SUCCESS in 13m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c4055f78684c40e3949b2996fa69ab3d : FAILURE in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5090765101eb4be9a7a47ffda3c727f5 : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6135f52f263447dda34129b27990eb8d : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/958b04a9ade748baba4289917080bb72 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/170122ca92944ca99eccc221a67b7cdd : SUCCESS in 1h 28m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57c2fcb55828466f816ecce5d031f6d7 : SUCCESS in 1h 32m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb2ebdd3234d418b9263c4c3daff3640 : SUCCESS in 1h 51m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31e70e83f16d4307813379b260dadb28 : SUCCESS in 2h 01m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa271313403546158a88377c8ecf43c5 : FAILURE in 41m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c34935efd5c4bdd9b5c589163bdb0f1 : SUCCESS in 57m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a456dad98b084ccd8da894d507b36b18 : SUCCESS in 5m 54s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b21fddcc98e40f68a85b5c5c5a7d755 : SUCCESS in 1h 21m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85668cf31f9d4234b2c9e7c8a44d8744 : SUCCESS in 1h 51m 17s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/207d543eca674defac2a8c419311eabe : SUCCESS in 2h 05m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c08cc9bf4f1b4ab8aeeeb91ca2aab357 : SUCCESS in 1h 51m 04s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4013bba88984e28b150687fb919f6c6 : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/54401824d7e1440bac201132592cc5c1 : SUCCESS in 2h 02m 09s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0771eb2a9cdb4e299525d66528122e66 : SUCCESS in 1h 57m 36s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/3e194be574124670aa3206d786d52b7f : SUCCESS in 1h 52m 51s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c5653a07a53040dca223166c7bd408f1 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8478bba10ce4b35c3e6546b030fcf1b2b8fc1eab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-11-15 11:06:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99bde496f5fce122bf7e448648d58001043935fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:17:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b91408af585348108ab203c20472b91b : SUCCESS in 19m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f6da72ccb89141338508710ee776478e : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/56bc993b75cd409f9764aed85c9f7c5b : SUCCESS in 58m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6c1062810e254e3a895107a580cef803 : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a774a84a02674fb6827e9f15d6202721 : SUCCESS in 1h 33m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a87af2ef0eba4615ad6340ea900bef24 : SUCCESS in 1h 38m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14e7a3a7f78c456396a44c4b4d044598 : SUCCESS in 1h 16m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/615e0679886944b9990ae0f6c2a2920e : SUCCESS in 1h 48m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217d56f3007540a197a7c2f7b5863ae2 : SUCCESS in 1h 52m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5ff1e12353a41b6839bb9f0e2d10b58 : FAILURE in 38m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad4ceea6ca4c4419bcf79fa8c45bb4b2 : SUCCESS in 1h 04m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79dbe346b795410aa6135c41ee7a4a57 : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ef53b35d3d9428fa53af171331aee11 : SUCCESS in 1h 36m 00s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547559e3b1e241ed8b187a40c99a61ec : SUCCESS in 1h 37m 33s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d89516b281ac436d8c274cb12ca15ce0 : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5975a1e743f42e085bcce33a7ae12ae : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/42095ecbf11a4c3fab03095f43c7f935 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/36249ee8e96f411ab3cec19d9f92f21f : SUCCESS in 2h 06m 11s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/286a6a8d248c452fb5c87feca8c9a7b5 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/d9c1eda3603c4213bc82972ce5f089e7 : SUCCESS in 1h 48m 52s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/372d292a6b6040a182f3b16491dff18e : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a8649a717ede12e7a41784d30c3d2c41bb38f21d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-15 15:07:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46fd2278c789e2d2f78c2c0abeea11105c549230","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-11-23 12:53:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73ee6dcb0c6ad7d1a65ce337fc8ca15e63be03f8","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2021-11-23 22:34:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5012ea8b825b873c9061602963084204d07e1236","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-11-26 13:26:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1096317208811d04e277f52abbc7f0ec0c0950f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 13:35:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c5ee23ed92490bd26cabae0045de374ecbce0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:48:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe4d09a55cd4dbab06b18fd985ba251 : SUCCESS in 16m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/45e3abbe987f4b099858ece747d992bb : SUCCESS in 1h 34m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4a18ef703bc849c9847feb724faf0b9b : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b53afd19fbc48748ba6aeb5ae85c417 : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b99fcbbbbe04859b021db21b75a5e81 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b2fe81848174b2e8a96985436fdf3dc : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d463d32774a146a2876dc51ea3aa9ab0 : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e25d74af5a14c119befc436a4332420 : SUCCESS in 49m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831969c925cf4709b1cb98d972162741 : SUCCESS in 6m 57s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc5db20d5ddf4af18a54f5a95876209e : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab18fb546a7402d8a6ad02e468fc344 : SUCCESS in 2h 01m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ab1dcb1027f4297a513aa70bb9c0a7b : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9342ac9702e04c3baeb320e1a4285187 : SUCCESS in 1h 54m 08s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/88721f9dea4a45b5aab5dcc49dd7c4c5 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6e5ce7a3a5a44178a8a34f60dbf6eb1 : SUCCESS in 2h 10m 50s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce03707472fe4c809078ab7d331b42f1 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/316154b07cab4e24be4401d0523e6cd3 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/ab7999c69900419e8b9c35818f7442b4 : SUCCESS in 1h 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"362d67ca64e50dc14bb71d450a07d79655159fbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:48:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94a8f22a1dbda2ebc26c3a9e0b8c83769f60c705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 15:49:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b48acfcf299441f88be00ce35774e0c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39dc90c99d945576bebeb1efdae1fdd76419898f","revisions":{"39dc90c99d945576bebeb1efdae1fdd76419898f":{"kind":"REWORK","_number":1,"created":"2021-11-15 08:35:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/817913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/13/817913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/817913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/817913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/817913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/817913/1"}}},"commit":{"parents":[{"commit":"a1d60f78b236c056c31bb92b185a0ef8cb38e677","subject":"Merge \"Add support for openvswitch interface driver with OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a1d60f78b236c056c31bb92b185a0ef8cb38e677"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-15 08:32:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-11-15 08:32:06.000000000","tz":120},"subject":"Set default for neutron_local_ip","message":"Set default for neutron_local_ip\n\nMove providing variable from integrated repo playbook to the role\ndefaults\n\nChange-Id: Ibbd537281c716a1353fc20768ec1eb20e97edb62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/39dc90c99d945576bebeb1efdae1fdd76419898f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/39dc90c99d945576bebeb1efdae1fdd76419898f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
