)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~828549","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fxena~Iac091d77abcc8075acff3981604747159e4c2884","project":"openstack/openstack-ansible-os_neutron","branch":"stable/xena","hashtags":[],"change_id":"Iac091d77abcc8075acff3981604747159e4c2884","subject":"DNM - test https://review.opendev.org/c/openstack/openstack-ansible/+/828548","status":"ABANDONED","created":"2022-02-09 15:07:15.000000000","updated":"2022-02-14 17:02:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d220011114c26d1b087cdb34c2fe55e5355f2ff","_number":828549,"virtual_id_number":828549,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-09 18:24:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 18:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4de62212b7b74b6470b4d58fa9c758b4143bfc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-09 15:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b49d0f7cd1a4c6ed7ac63e2857089d4a34e34494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-09 15:10:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a5ec194dcee7a3f31491f1a2904edce4b9fe3ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:24:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38e5f1991c244d24ad71ebdc8bdde2af : SUCCESS in 19m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ca774e2749054e9eb77b294175171598 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e99105df2a134a58b26160cc32f163c1 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/154a0a0c7b7941d1b3e6db7fa6afa72d : SUCCESS in 1h 35m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/784a362ef6e3405cb4aa781c848774ea : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/055d9fdec85f4cada2e79eb7b6014e8f : SUCCESS in 1h 21m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a89ddbc5b29a481691fba0c140220fd4 : FAILURE in 24m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adea363543ec4925a4a7d96bccf6b9e3 : SUCCESS in 1h 50m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/40048767068145679e8a4557c73ef430 : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e27ca631235418c9105f0615ded0688 : SUCCESS in 1h 00m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4488c9fe6d4324b22adfef99a68534 : SUCCESS in 8m 38s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d96eb801d424c49abeed78880b8ace9 : FAILURE in 1h 26m 11s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b5035f2c72475ea0bd81e0775f8f20 : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25976e6ee95d4d2bb2baf4bc390d83cf : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/23b53aeff3694f9d94682d05881925d4 : SUCCESS in 2h 05m 12s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/c93ad9a083d74c09ad0d96099604c977 : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd9b7e1ea62f48b99fc65163c628b209 : SUCCESS in 2h 08m 51s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d25d2bd3f34463a88bb15d194bffbf1 : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/712329e21f8443dfacd46fbf58510412 : SUCCESS in 2h 10m 41s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/7039a634873b45f0935dbe59eaf39562 : SUCCESS in 1h 52m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5d220011114c26d1b087cdb34c2fe55e5355f2ff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-02-14 17:02:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b777b5faf60636f2d3240d5fbcb870a161e166b1","revisions":{"060b7af7027599439daef41d94041bc24bc35e59":{"kind":"REWORK","_number":1,"created":"2022-02-09 15:07:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/49/828549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/828549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/1"}}},"commit":{"parents":[{"commit":"0362c51f8f1f627f57a3fada34d60eb919e4c261","subject":"Enable recursion in combine() filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0362c51f8f1f627f57a3fada34d60eb919e4c261"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-09 15:06:33.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-09 15:06:33.000000000","tz":0},"subject":"DNM - test https://review.opendev.org/c/openstack/openstack-ansible/+/828548","message":"DNM - test https://review.opendev.org/c/openstack/openstack-ansible/+/828548\n\nChange-Id: Iac091d77abcc8075acff3981604747159e4c2884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/060b7af7027599439daef41d94041bc24bc35e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/060b7af7027599439daef41d94041bc24bc35e59"}]},"branch":"refs/heads/stable/xena"},"b777b5faf60636f2d3240d5fbcb870a161e166b1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-09 15:10:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/49/828549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/828549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/828549/2"}}},"commit":{"parents":[{"commit":"0362c51f8f1f627f57a3fada34d60eb919e4c261","subject":"Enable recursion in combine() filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/0362c51f8f1f627f57a3fada34d60eb919e4c261"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-09 15:06:33.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-02-09 15:10:37.000000000","tz":0},"subject":"DNM - test https://review.opendev.org/c/openstack/openstack-ansible/+/828548","message":"DNM - test https://review.opendev.org/c/openstack/openstack-ansible/+/828548\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/828548\nChange-Id: Iac091d77abcc8075acff3981604747159e4c2884\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b777b5faf60636f2d3240d5fbcb870a161e166b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b777b5faf60636f2d3240d5fbcb870a161e166b1"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
