)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~855803","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ifd04358878aacce0b8236f01b9b53aaaba8a95d5","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2022-09-06 11:02:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd04358878aacce0b8236f01b9b53aaaba8a95d5","subject":"Convert include to include_tasks","status":"MERGED","created":"2022-09-04 02:14:35.000000000","updated":"2022-09-06 11:03:55.000000000","submitted":"2022-09-06 11:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855803","meta_rev_id":"db88640879d8d6dcd70d8e6ffcfaa8af85b53e78","_number":855803,"virtual_id_number":855803,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 11:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2022-09-06 08:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2022-09-05 07:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2022-09-06 08:24:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-04 04:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-04 14:33:11.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2022-09-05 07:23:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-09-06 08:24:07.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"42fdc1dafefc957b6f3aa76bea67e7a116284804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-09-04 02:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8af9e013d16f094b98db5a0b75354e4144503798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-04 04:46:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e044d9bf962e4e569cadc15a76dc17d5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e6f38b99dc4d17adfb9488a48febd1 : SUCCESS in 22m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4dcccd887a748718682d2080dfbb2c6 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bf2278c68454823859a128718d158d1 : SUCCESS in 1h 36m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f927b305323849788ce94ec40a34133e : SUCCESS in 1h 30m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d00d518a6a3a4faea85982af9fe8c880 : SUCCESS in 1h 28m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b99caa93ce64c9094d5cef7e62d3434 : SUCCESS in 2h 06m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e83ab40e0c00473f8fb9f7a17eb6b570 : SUCCESS in 2h 21m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4789711379248b0ae444171b9503f3d : FAILURE in 34m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76e3e2db5adb46ef80c6a10fa0a12a91 : FAILURE in 31m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7036eb1266470a8c8e2224c67e43a6 : SUCCESS in 5m 39s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f95dbc99bac40aebb7efc0d750c44d3 : FAILURE in 1h 30m 09s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64318bba36504feb9522f34cdbc7904f : SUCCESS in 2h 00m 10s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b786a63cefd49aaad406901e915a9fc : SUCCESS in 1h 29m 25s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ae6948145a24f77b36e92ab8900f4da : FAILURE in 29m 01s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cd3134b41ee44aea4f9f899d08ef189 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e967b657d8e43fa92b9514b823d64d5 : SUCCESS in 2h 11m 07s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9d14484db0f4bde8257d349c3187dd9 : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c040e24d717c484f9e44d833febfa019 : SUCCESS in 2h 04m 14s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e935e7791257491bbedd9624f685b6c3 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"835a134abecc50f9d459620af3ee23ba0ee80416","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-09-04 14:33:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce78ebc4790825f40fdb8395e8a5aaee69b6d09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-04 16:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cb9b9ea3ac44c782a30b7f77fd9319\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76bce71f513f463ea16a705dc6bff115 : SUCCESS in 17m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/92983874a8a04174aa666d821fb68489 : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d0a46dd8db4ef0a3e1b16703750a30 : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4eaa6fb46544ae584ab3509d6926414 : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04da365ff2e048329ec1e402b8c6dc72 : SUCCESS in 1h 19m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/026a715cc40945c5b4deddc71c7cd61b : SUCCESS in 1h 53m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5833c73c94f040f79baafac9f2d4f394 : SUCCESS in 1h 53m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/614be2928d71474e91d44d902ce1a4b0 : FAILURE in 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81f6e976f22a4cbea893a1730284d970 : FAILURE in 28m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adf9749490d4f5d8ad781771fa1dede : SUCCESS in 6m 57s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e8c12394f6b426cb6ad580e0d20eac2 : FAILURE in 1h 25m 24s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcbbd52ae0ee45cba2b7ebb49f0054cb : SUCCESS in 1h 50m 54s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6861d590f2a24311b08588afcfcef60b : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f91dc450f449c795749df0cf4d9560 : SUCCESS in 2h 00m 04s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/95d74702e632403a8664f3f449bae261 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5839667d8eac486e9f09f2d90540b25a : SUCCESS in 2h 06m 48s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8a127c287444a62afe7625b119aeff6 : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a830a85317488a8f92c6b751d11272 : SUCCESS in 2h 22m 41s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c99348ddae4231bd37c036052c3d1e : SUCCESS in 1h 27m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bccc20137626a20f1030a2f5e9c44e6c2fe16cee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-05 07:23:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4e9a213a2cd10b457dc0081d2acaef0802bd3d5","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-09-06 08:24:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a08219fc5fa3ce1bcad6f6948d9bb58badcf5021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 08:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0165637c841b69b01b5003ca329b5aa815221865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 11:02:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a434738b4735417cba5b3753211cb27f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0cd92b31b5148b3bdcd9facf0f7f5e4 : SUCCESS in 20m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3d0e97c9c19e42079e01de0f740252a0 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f06b9f5a80e0480ab895f6925796db0f : SUCCESS in 1h 31m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c441a525fcfc4aa881edb52652f29415 : SUCCESS in 1h 24m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a746fea2dd454d4b92837300609f9563 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc193b32bb2b4e8d8e29423163e48ea9 : SUCCESS in 2h 30m 11s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fde7bdf401fc488bb84b805b7fbb68b9 : SUCCESS in 1h 27m 39s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd1ab096327f479eb7723e5dbfaa0d1b : SUCCESS in 2h 09m 26s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a51666dcc124328a44fd6f7924db63b : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc010b102e6c4462b4997734832bbd1f : SUCCESS in 1h 59m 41s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/501df638904f444cafde7901806d42f2 : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813936258816477ba6e26164a42a9c24 : SUCCESS in 2h 05m 26s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d1c91ec054f45e88b6b42529361c112 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c658312c6b17b987c369a03ca028f87d14638124","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 11:02:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db88640879d8d6dcd70d8e6ffcfaa8af85b53e78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 11:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b69ef62d551e49a5943b0c80b5795b32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84719c212a842f8bad7519e4b8c42fe : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c88b486fc9bd8311d184ddd1cf0739a901346e8","revisions":{"4c88b486fc9bd8311d184ddd1cf0739a901346e8":{"kind":"REWORK","_number":1,"created":"2022-09-04 02:14:35.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/03/855803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/855803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/855803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/855803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/855803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/855803/1"}}},"commit":{"parents":[{"commit":"ef764770e2ebf79251ce6ffba5923fe0e358a5f9","subject":"Switch jobs from CentOS 8 Stream to CentOS 9 Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ef764770e2ebf79251ce6ffba5923fe0e358a5f9"}]}],"author":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-09-04 02:12:18.000000000","tz":-300},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-09-04 02:12:28.000000000","tz":-300},"subject":"Convert include to include_tasks","message":"Convert include to include_tasks\n\nInclude is deprecated, additionally include_tasks is now faster.\n\nSigned-off-by: Kevin Carter \u003ckevin@cloudnull.com\u003e\nChange-Id: Ifd04358878aacce0b8236f01b9b53aaaba8a95d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4c88b486fc9bd8311d184ddd1cf0739a901346e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4c88b486fc9bd8311d184ddd1cf0739a901346e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
