)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~858271","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I432adbc167dc03f1a492dc2dd6ef7451646b0cac","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"neutron_ovn_azs","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-29 19:06:14.000000000","reason":"Change was marked work in progress"},"34653":{"account":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"last_update":"2024-01-03 18:46:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I432adbc167dc03f1a492dc2dd6ef7451646b0cac","subject":"Neutron OVN plugin AZ support","status":"ABANDONED","created":"2022-09-19 09:26:48.000000000","updated":"2024-01-03 18:46:58.000000000","total_comment_count":8,"unresolved_comment_count":4,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"727116f418d49d36d9a2ed20bdf86fc0b8984b85","_number":858271,"virtual_id_number":858271,"owner":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-07 17:39:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-09-19 11:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-03 19:17:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"092fb4d5adcf38daccd804f5cde1fd3306d23d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2022-09-19 09:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a161e3d73357cb64fa0c2667fb3811ba993b341","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2022-09-19 09:53:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a393b70b9948a7b807132cd8649f457be48c12d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:50:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2b8235c9564498aef48fca652f144a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15aef603716e4ca4b1444c00d77d6157 : SUCCESS in 21m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57c50a7e9fa34be89b7413df166d2207 : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/93eb67f8b67941a691a5587729fbdfa3 : SUCCESS in 1h 33m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0156a10b6e6742918c9515f624164623 : SUCCESS in 1h 34m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f778029ab71345648e7c67ec9e701aa4 : SUCCESS in 1h 24m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/795d8a61eba149a3b2319697e9f01df6 : SUCCESS in 2h 07m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e71408671aa4f2f82d3ce46e76a98ef : SUCCESS in 2h 03m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/da3a6d91b6d947ac92b09292ae351902 : FAILURE in 38m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3446dad146b34ae19957dfe173b358e0 : FAILURE in 35m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cf1fb79b134f30818c0e5d43d855b8 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/527127f7cebb44b9b1a0d95213a73cbf : FAILURE in 1h 28m 38s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dd2185c7ff7478e9f3e66ae499cb44a : SUCCESS in 2h 12m 37s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c033750a6f3345e09a93a479cc9de219 : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d8cadd04d0440e89ca134224ac0dd69 : SUCCESS in 2h 17m 53s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d121bdc4e0904c1baffd634402acfb11 : SUCCESS in 1h 29m 17s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/da17c9140a3c4520914ad7d333dddf53 : SUCCESS in 2h 19m 54s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d990fe2d60043c383939997400ef00e : SUCCESS in 1h 29m 55s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9c1dd97b6674446b1f42d887cbb65f9 : SUCCESS in 2h 13m 48s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae0f5fc172c64f2589655fc001bc9a7a : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d33e13d798a37afc09bf850b788e899b40d0bee1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-03 19:17:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c74e371aa6299cbfae6b86847af7f312913f2205","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2023-06-19 14:45:40.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"f5d880636b6df11aa5c24bf1dd3a5aee0439c6c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2023-06-19 14:54:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcc32bba6d2449eba98e795c0eb2efa91ad0f7f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 17:37:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40f0a9a32387403c8d24c230da0c7373\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ae4f04f0b6427181250a6f83cff468 : SUCCESS in 12m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0483790af9644bbdb4ef67678b87b56e : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/314c4d21e37345eba691661f56591f7f : FAILURE in 1h 01m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8b8726da720f4fbfa7bad9279c6c83a4 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f37324178e0a4288b07a96cd04d691fa : SUCCESS in 1h 32m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d733b0de9bc54420b4b2e9633288906a : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c774096ce65456cae72e9ef360ef9c2 : SUCCESS in 1h 34m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d29b3cac801403885a9a12f284f0103 : SUCCESS in 1h 30m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/678efa5335414aba8a076c14316588b7 : SUCCESS in 2h 34m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3036702cbc148a4a1e726eb680cfd32 : SUCCESS in 1h 59m 01s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6edeb43af6a403bbb46a4df2a5e2372 : SUCCESS in 58m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa9b03b0e8b44be49dc14f10e9c4c2a6 : SUCCESS in 54m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2860eb0f704f11bd753a2e0bcba20f : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d46bcaaa9b745c7a693dd570de8157b : SUCCESS in 2h 04m 38s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/077dffaece044416a4f6b92951a8f5be : SUCCESS in 1h 36m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3efcabf12d9455faee928beb63f08f2 : FAILURE in 43m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bc65feaa4a84884a0cc52911102538f : FAILURE in 1h 02m 12s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3242230beb040ff8704599eaa1ed631 : FAILURE in 41m 36s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/118b1c9054b44368852c1d4d690d7bb8 : FAILURE in 1h 02m 55s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea2cdb045af54f1ba22ca12da2ae28c9 : SUCCESS in 2h 04m 14s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64ab50eb1c5f4baebc66db807ac06706 : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6355d12065d350abcb7b3c155cce1fa787db89c9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2023-11-29 19:06:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d99232f5f716e8fcb0ca0877412675585761a86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 21:54:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a174f882ae284e598319c848503caf91\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89e6f0b7aacb4584b322c52793bf8206 : SUCCESS in 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8dafd6a404284b80844dda147402f66a : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f2c9b7903e344adb9cc611c189da659 : SUCCESS in 1h 06m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5494d3421d0f41af8885875b1dd7154e : SUCCESS in 1h 12m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4753a480794443bdab62393981dcc223 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f425b16a2f7344c99a4539660f48307f : SUCCESS in 59m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64e0c2f101234f67b5183c4d96887ee2 : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4870e4d6c4a4a1f9f9c9ba9823d81b9 : SUCCESS in 1h 21m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9a11ed10c37f474f8287fdb152185b3f : SUCCESS in 2h 38m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/693c9652b0cf40d7ae000f17b6877ea6 : SUCCESS in 1h 50m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51992076faa044928e4de7f0899756ee : SUCCESS in 2h 10m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e658964389e74515836e729eff851069 : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f20e8695d378436a9b27cb7a64660cf5 : SUCCESS in 55m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec444c5ed00346dd8320bd654c4d9d12 : SUCCESS in 59m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5d0fa3d1b44d4b8e03c73fd9d7ac85 : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eeed8134f32b4f228aa1c13880273e8d : FAILURE in 1h 43m 12s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b573ed259bc497298ef535e6def98ed : FAILURE in 51m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/64b2acb80c4c4a6bb4ea9beb1f26ffe7 : FAILURE in 1h 44m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/515556b2aac84fadbe597f65778961d1 : FAILURE in 57m 53s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/72b00a20c89f4ddeab5b32dd12aa1623 : FAILURE in 1h 56m 42s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb0260a728cf48e8a3974f17ae6f87a7 : FAILURE in 1h 03m 53s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90d2c6c5abdb40e3947de8f90e4786e3 : FAILURE in 1h 33m 25s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88f502fb555a421ba91f2e696256f100 : FAILURE in 1h 07m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d5150c0971cbf21564a348db1b679d035fc1eda3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2023-12-07 14:32:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"12a2bfd1b14acfae304e5596a46e752250c7d7c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-07 14:43:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"31238d357bf5b91845d6ffb48312e6e87e08ed19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-07 15:08:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91b1354b9955d83b28e6d4a1d1852934f42b3700","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-07 15:11:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8db627329682884576d5cf29dd3a0afc17e8c645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:39:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fdda1b04794b07a19a5540b1d6a124\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b6a6e60acc246d3b6b9b2e1408429ea : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ee4c6d129c8b4a3dbea2a9e65649ba68 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3576c676a9884111aa01dd6499f86da1 : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b43fea6d04c4160bbe4772f2422d7b7 : SUCCESS in 1h 12m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d4ce8247a845455fbd561ddc5eef9bed : SUCCESS in 1h 19m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa739ccc9cc34963b7e6484a580bc432 : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2cffea3bd857429aa25d5421b173f527 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6c86e886c504b90a23c7c615bb15b3a : SUCCESS in 1h 33m 11s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3f65112337b4be5abfb848a42995c6d : SUCCESS in 1h 16m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2effd0e8cf6745ed997afff83612da22 : SUCCESS in 2h 56m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82a245a141b340279788866038479aa1 : SUCCESS in 2h 04m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad4ff8ff9e9c4e118325617b76c9dc30 : SUCCESS in 2h 15m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/709ded08f4b2464dba88cf547eace025 : SUCCESS in 2h 17m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/236fd531c1e944c48d9744ed95898463 : SUCCESS in 1h 08m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0b10fa553a4425dbd462591f5d4067a : SUCCESS in 56m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aee38befa024201bbaf63b3d7bb39dc : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae0ac038b5e540e4b3adc9b5ff2e3d9b : SUCCESS in 2h 00m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/277956f67afd475e9aea58ea66c5fc37 : SUCCESS in 1h 22m 40s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae152dc692f44635a44accb98855ffed : SUCCESS in 2h 06m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a00b5e203d74e8ebd855c19fc1fe10d : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a365928a1f441bf8b23baac3a9f94ba : SUCCESS in 2h 00m 07s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9185b1f2b09f439aa50516cf9e433f90 : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d29601cc1e3e456a8eafea55ea6ac2d9 : SUCCESS in 1h 52m 05s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c97549ffc7d4de889ba9db0b72c543c : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":4},{"id":"0a9f78a03f4e2e2580dd69080e16956f0165328b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-12-14 13:41:52.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fcea46630953801777912324a99fc09331f5c540","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-03 14:03:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"727116f418d49d36d9a2ed20bdf86fc0b8984b85","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2024-01-03 18:46:58.000000000","message":"Abandoned\n\nSupersed by https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/904572.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"592d8bd343681a852a99338cf4838353dd94ceba","revisions":{"2d405526116fd467ed73dd91047cebc5147cd05b":{"kind":"REWORK","_number":1,"created":"2022-09-19 09:26:48.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/71/858271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/71/858271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/1"}}},"commit":{"parents":[{"commit":"37b11c5b118504732da43ddbe0844fd112955f7f","subject":"Merge \"Convert include to include_tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/37b11c5b118504732da43ddbe0844fd112955f7f"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-09-19 09:26:25.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-09-19 09:26:25.000000000","tz":180},"subject":"WIP: neutron OVN plugin AZ support","message":"WIP: neutron OVN plugin AZ support\n\nChange-Id: I432adbc167dc03f1a492dc2dd6ef7451646b0cac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d405526116fd467ed73dd91047cebc5147cd05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d405526116fd467ed73dd91047cebc5147cd05b"}]},"branch":"refs/heads/master"},"1cf98cfdd38eb038e19323f1f38d2a870bc52caa":{"kind":"REWORK","_number":2,"created":"2023-06-19 14:54:54.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/71/858271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/71/858271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/2"}}},"commit":{"parents":[{"commit":"d0c37ede9fde4a25724fbe6ec4e57426bfe7416e","subject":"Drop OVN package installation from ovn_config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d0c37ede9fde4a25724fbe6ec4e57426bfe7416e"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-09-19 09:26:25.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2023-06-19 14:54:32.000000000","tz":180},"subject":"Neutron OVN plugin AZ support","message":"Neutron OVN plugin AZ support\n\nCloses-Bug: #2002040\nChange-Id: I432adbc167dc03f1a492dc2dd6ef7451646b0cac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1cf98cfdd38eb038e19323f1f38d2a870bc52caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1cf98cfdd38eb038e19323f1f38d2a870bc52caa"}]},"branch":"refs/heads/master"},"443435c426daf695ad4979de1a6810fd9b0c2e4c":{"kind":"REWORK","_number":3,"created":"2023-11-29 19:06:14.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/71/858271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/71/858271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/3"}}},"commit":{"parents":[{"commit":"ef4d3278556c6e34f79b195cfb3e0d300f7184f7","subject":"Update VPNaaS package for RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ef4d3278556c6e34f79b195cfb3e0d300f7184f7"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-09-19 09:26:25.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2023-11-29 19:06:11.000000000","tz":180},"subject":"Neutron OVN plugin AZ support","message":"Neutron OVN plugin AZ support\n\nCloses-Bug: #2002040\nChange-Id: I432adbc167dc03f1a492dc2dd6ef7451646b0cac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/443435c426daf695ad4979de1a6810fd9b0c2e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/443435c426daf695ad4979de1a6810fd9b0c2e4c"}]},"branch":"refs/heads/master"},"592d8bd343681a852a99338cf4838353dd94ceba":{"kind":"REWORK","_number":4,"created":"2023-12-07 14:32:39.000000000","uploader":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"ref":"refs/changes/71/858271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/71/858271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/71/858271/4"}}},"commit":{"parents":[{"commit":"ef4d3278556c6e34f79b195cfb3e0d300f7184f7","subject":"Update VPNaaS package for RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ef4d3278556c6e34f79b195cfb3e0d300f7184f7"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-09-19 09:26:25.000000000","tz":180},"committer":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2023-12-07 12:07:00.000000000","tz":180},"subject":"Neutron OVN plugin AZ support","message":"Neutron OVN plugin AZ support\n\nCloses-Bug: #2002040\nChange-Id: I432adbc167dc03f1a492dc2dd6ef7451646b0cac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/592d8bd343681a852a99338cf4838353dd94ceba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/592d8bd343681a852a99338cf4838353dd94ceba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
