)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~860902","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I45d6cd0f039c3fd2016e52df3607a5ac22956d0a","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"30356":{"account":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"last_update":"2022-10-17 20:42:49.000000000","reason":"Change was submitted"},"22165":{"account":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"last_update":"2022-10-17 20:42:49.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-17 20:42:49.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-10-17 18:19:40.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I45d6cd0f039c3fd2016e52df3607a5ac22956d0a","subject":"Revert \"Remove the neutron-fwaas since it retired\"","status":"MERGED","created":"2022-10-11 06:36:55.000000000","updated":"2022-10-17 20:43:42.000000000","submitted":"2022-10-17 20:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":736100,"submission_id":"860902","meta_rev_id":"0da4837adb462b190945f6dc86dced736a1a7a94","_number":860902,"virtual_id_number":860902,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-17 20:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-10-17 16:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-17 18:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-17 18:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"}],"reviewers":{"REVIEWER":[{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 06:36:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2022-10-11 06:36:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2022-10-11 06:36:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-10-11 06:36:55.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 16:59:09.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"23fb46f69258e12f436b5344d09d42b1cfcc7e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-11 06:36:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86516be508d03a7af7205df66b95351a3e5ae729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 06:37:15.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible-os_neutron for 860902,1","accounts_in_message":[],"_revision_number":1},{"id":"86bdcab7ddea4384a4f123a9bba8c6bdc81b5edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-11 07:30:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1782c33cf563ff9ece54fd4a65d0ff7a8d4bf828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-11 07:37:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e7ff4a0643bfe8b9b9e2fddb89b6bac491c0851a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:41:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19c60f93127849deac05585d99a9b6c9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4758977da74547950737396532cbc2 : SUCCESS in 13m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/28ac397344ea4f0183d24eaa23b6ad40 : FAILURE in 1h 06m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/efa9e9569d554d8c80a4c9a720609775 : FAILURE in 1h 04m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3eba68a8fc54401ea4b92701c0dcb29b : FAILURE in 1h 00m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5d3674c4914b9abf4b116caa035b7d : FAILURE in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b18d55ffef2149afab6ebfa3140eeac5 : FAILURE in 1h 08m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce3e55369c3d492c8c0342df98b37b6b : FAILURE in 1h 05m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f09b994595d4daa8adae6ef4ceb8314 : FAILURE in 1h 33m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fc5bd4db8a14c599a80c9a2f1d03b50 : FAILURE in 43m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc69980718a74df88dab502662276ba4 : FAILURE in 31m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9a7926918d43d8aab45346a305e2bd : FAILURE in 41m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ee691bbcfe4100aaf150f99e87158c : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63b6dfc81ab64505a2187b143202d676 : FAILURE in 1h 28m 10s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7102cf18bfc341989a4f0ecbf3a92ac3 : FAILURE in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c0edb0acb748e3b9f2fb0f53e9ce06 : FAILURE in 1h 19m 07s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f11ecbda207a4a2b8fb173e8539f0d7c : FAILURE in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/866ea5a9e0264a55ad17ae8a4b372b0e : FAILURE in 1h 14m 34s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b62aa049fda04bc88d5a3700c65b69bf : FAILURE in 1h 36m 52s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f591edb38f04fc1955f6d8a60f37c80 : FAILURE in 1h 21m 28s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51485b53f65145db83fbb7d7a738e78a : FAILURE in 1h 59m 57s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ded457403ea45ecb46aa055f33dad9f : FAILURE in 1h 27m 56s","accounts_in_message":[],"_revision_number":3},{"id":"61ad38654c033f97bb634ff3d11aa83c38e3d2bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-14 08:03:37.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"05da2a5e7f4199954ba3a54c564c6df102e8ab0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 10:27:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c53f8ed2a104ec49d7f7779a134bc6a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12bc199da09b4c99b4295244b6b7a197 : SUCCESS in 13m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1eaadde188ea44a488c5a603a1be82c5 : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7d5d3215c20426588eb5bc0d7f26957 : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0507d642bbcb403580dc1f34bc46c9c0 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d05c9fb4af748d79a1dca9b573856e1 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f2a5873f78b40cc978d48882feb0be8 : SUCCESS in 1h 20m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6805d5c9f83f40f7ab187c8e3b2f1f72 : FAILURE in 56m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71c6919213324ed6af12b99a0eefcaed : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2752433e3b14e8e8261f15beb7a32d5 : FAILURE in 36m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac72c9105ba647ec9f6bb3e1c0d65ee3 : FAILURE in 35m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bf9d4dda31949c4867624c263e6e967 : FAILURE in 39m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd41fa247b3f42dfba57a2dc916fc828 : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a037ae24c841ecbbd5bc4a6f832db8 : FAILURE in 1h 31m 12s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6555951094de44b6aacb214541a7dcd8 : SUCCESS in 1h 53m 43s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acdaa4a5c706434bb58d4b6946fdb287 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5153e6054f4442a98351425f50b1ed51 : SUCCESS in 1h 52m 57s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7cc6aaeef41443f9b8c69b2bf9cad48 : SUCCESS in 1h 12m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ea6c54d2b9845379de114e0eb50d431 : SUCCESS in 2h 08m 59s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/723e102007cc423dbbbcb073abc6f26d : SUCCESS in 1h 29m 12s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65414b9e2212414db60acaa997af6c86 : SUCCESS in 2h 20m 54s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db6328d089e043108e564f4deb123ec4 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4bc7442ddbaa1678c5ce04eba5aafdbd3eccf2fb","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-10-17 16:59:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f41ad518b953499559c88833bf5e128391166fd4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-17 18:19:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a68478760a91a9fc572c00d03f808a2c69228019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:21:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"99e122502fa6f48425fe3ee44d68728ee844ccef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 20:42:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c415ed767e474b18ae00c4712ec5dd1d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f197be78d5874005823ac4e7877aed29 : SUCCESS in 13m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1471d262f124b639b3998bd94a22827 : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e51c5e44172c49d782cc8227d3409e33 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9cb4b595def44be7a351418980b977d8 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25fd61ed233c4c95828716553a49febd : SUCCESS in 1h 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591461a0fede48458a676fcddffe564d : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b29ad488f264082a1a29e17cd241b1f : SUCCESS in 2h 16m 11s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/077d1f7d70fd401dbf211041288ebab3 : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a22f88a592047a9bfd9935fb9819150 : SUCCESS in 1h 49m 05s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d18d1be6fbe64f33b241794a8d88e44d : SUCCESS in 1h 35m 58s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f75f6f09ecf0473785995cbb4f1e0d8e : SUCCESS in 1h 46m 16s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfbc1d55dcc74734b930e48bb652358c : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d228bcbdf4924a9e8c5ec40797b37380 : SUCCESS in 1h 58m 53s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/406df57809a24fd5858c77ccb2551437 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":4},{"id":"e62ef17de8b2319008fb3e4b951f7703668d7f3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 20:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0da4837adb462b190945f6dc86dced736a1a7a94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 20:43:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae1863cfd4c433c959e6b5972eb8a66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017ce8446d9f45218455caf61f981790 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bd1db203a0a2a470486d2c235dcdf62ca9f0cb63","revisions":{"8300bcd2cbda4e406508147bdecb345082d443a5":{"kind":"REWORK","_number":1,"created":"2022-10-11 06:36:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/860902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/02/860902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/1"}}},"commit":{"parents":[{"commit":"5fb6ef370e106c13da9f14126710b9970ac0874a","subject":"Remove the neutron-fwaas since it retired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5fb6ef370e106c13da9f14126710b9970ac0874a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-11 06:36:55.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-11 06:36:55.000000000","tz":0},"subject":"Revert \"Remove the neutron-fwaas since it retired\"","message":"Revert \"Remove the neutron-fwaas since it retired\"\n\nThis reverts commit 5fb6ef370e106c13da9f14126710b9970ac0874a.\n\nReason for revert: FWAAS has been revived in Zed with I14f551c199d9badcf25b9e65c954c012326d27cd\n\nChange-Id: I45d6cd0f039c3fd2016e52df3607a5ac22956d0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8300bcd2cbda4e406508147bdecb345082d443a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8300bcd2cbda4e406508147bdecb345082d443a5"}]},"branch":"refs/heads/master"},"886d510565b9b06bfe25d76a4cff1ec193864d5d":{"kind":"REWORK","_number":2,"created":"2022-10-11 07:30:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/860902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/02/860902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/2"}}},"commit":{"parents":[{"commit":"748f23992605aa48490327711827037a6635c10e","subject":"Replace git-core with git for debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/748f23992605aa48490327711827037a6635c10e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-11 07:30:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-11 07:30:23.000000000","tz":120},"subject":"Revert \"Remove the neutron-fwaas since it retired\"","message":"Revert \"Remove the neutron-fwaas since it retired\"\n\nThis reverts commit 5fb6ef370e106c13da9f14126710b9970ac0874a.\n\nReason for revert: FWAAS has been revived in Zed with I14f551c199d9badcf25b9e65c954c012326d27cd\n\nChange-Id: I45d6cd0f039c3fd2016e52df3607a5ac22956d0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/886d510565b9b06bfe25d76a4cff1ec193864d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/886d510565b9b06bfe25d76a4cff1ec193864d5d"}]},"branch":"refs/heads/master"},"9a6bf9b2c2c99a5c9eb61f108499c4d139921e74":{"kind":"REWORK","_number":3,"created":"2022-10-11 07:37:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/860902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/02/860902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/3"}}},"commit":{"parents":[{"commit":"748f23992605aa48490327711827037a6635c10e","subject":"Replace git-core with git for debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/748f23992605aa48490327711827037a6635c10e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-11 07:30:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-11 07:37:39.000000000","tz":120},"subject":"Revert \"Remove the neutron-fwaas since it retired\"","message":"Revert \"Remove the neutron-fwaas since it retired\"\n\nThis reverts commit 5fb6ef370e106c13da9f14126710b9970ac0874a.\n\nReason for revert: FWAAS has been revived in Zed with I14f551c199d9badcf25b9e65c954c012326d27cd\n\nChange-Id: I45d6cd0f039c3fd2016e52df3607a5ac22956d0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9a6bf9b2c2c99a5c9eb61f108499c4d139921e74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9a6bf9b2c2c99a5c9eb61f108499c4d139921e74"}]},"branch":"refs/heads/master"},"bd1db203a0a2a470486d2c235dcdf62ca9f0cb63":{"kind":"REWORK","_number":4,"created":"2022-10-14 08:03:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/860902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/02/860902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/02/860902/4"}}},"commit":{"parents":[{"commit":"748f23992605aa48490327711827037a6635c10e","subject":"Replace git-core with git for debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/748f23992605aa48490327711827037a6635c10e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-11 07:30:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-10-14 08:03:37.000000000","tz":0},"subject":"Revert \"Remove the neutron-fwaas since it retired\"","message":"Revert \"Remove the neutron-fwaas since it retired\"\n\nThis reverts commit 5fb6ef370e106c13da9f14126710b9970ac0874a.\n\nReason for revert: FWAAS has been revived in Zed with I14f551c199d9badcf25b9e65c954c012326d27cd\n\nChange-Id: I45d6cd0f039c3fd2016e52df3607a5ac22956d0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bd1db203a0a2a470486d2c235dcdf62ca9f0cb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/bd1db203a0a2a470486d2c235dcdf62ca9f0cb63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
