)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~862403","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~If8ca3f2035ada97cff248ad49771eefab95c6c23","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/ovn","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-12-01 11:23:56.000000000","reason":"Change was submitted"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2022-12-01 11:23:56.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-01 08:58:54.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"If8ca3f2035ada97cff248ad49771eefab95c6c23","subject":"add ovn ssl config","status":"MERGED","created":"2022-10-21 19:50:49.000000000","updated":"2022-12-01 11:25:02.000000000","submitted":"2022-12-01 11:23:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"862403-osa/ovn","meta_rev_id":"3ea0f3b62b7a9056159ffb18aa06e986f5eb4336","_number":862403,"virtual_id_number":862403,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 11:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-12-01 08:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2022-11-30 18:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2022-12-01 08:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 22:20:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 14:44:19.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2022-11-22 14:44:19.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-11-24 15:43:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-11-29 08:39:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-11-29 20:06:57.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"2c730d9c26c064ac3acbadb3cd6f4236e759e284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-21 19:50:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"058a1e5e31cc9252073ae45bebc24c5c72f00d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 22:20:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af59c966c5094ba4b2c8d668884394f9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80144ab170064bef9a445d8220c34d34 : SUCCESS in 12m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c8a5e4398286413399fe1948eb152547 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aaae64b1456486bb8447b917288eae9 : FAILURE in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/33794dbffd4a40d0be9b1f337666dc5a : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8048bbd10054cc7a2e57ad785184419 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59370c7cb41740fdb87db699cd07ac65 : FAILURE in 1h 20m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7efe49ca929461e95798e4a1d386f5f : FAILURE in 59m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3795d5d576341cea8ab9bf6d6e0dc69 : SUCCESS in 1h 35m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/39b788df63d94edeb7ba4b45a5817ac8 : FAILURE in 40m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a12818b251741cdbcdef39fd7656ca2 : FAILURE in 32m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d73d005a3dc74ca29f253d8ebc16d251 : FAILURE in 41m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d1b2e592c34c79aadc9b7408d027cd : SUCCESS in 4m 48s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2181c05b93cc41b1a98bd97e9c34af1f : FAILURE in 1h 27m 14s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd8e554d62494378a0f4616040999046 : SUCCESS in 2h 13m 37s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5446cf07e17740ecbdb47afc30bc52c9 : SUCCESS in 1h 30m 42s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4f0925d3ede4888a2df27a5410b56c6 : SUCCESS in 2h 04m 12s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9085211f9ff41c5aa72d9aaa2e8633d : FAILURE in 1h 23m 53s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/636a505e8e9d42dab6ef72a8c3919d6b : FAILURE in 1h 59m 19s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0995a4cbe9a1437c8f212886e1e69d24 : FAILURE in 1h 27m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3a2dde5d6c848c1813739afb340f272 : FAILURE in 2h 21m 56s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/018b9f4adad742e0834f33f2037aa9f0 : FAILURE in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a0ac14092abfddcccfe0603bedc80d8102ac295a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 15:28:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"537fd5e1a988539dc1d7d47ccb043bbe3e4960e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 17:25:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af2f5dfd2745f16c549110b5a0ea2a250ca5b674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 20:09:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/331f4d8953fd4ee7973386af43241487\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3a52434d78f4f2b8975789023167ac9 : SUCCESS in 17m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a15a6f3ca07c48209a7fe29b4b00278a : SUCCESS in 1h 24m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3108eec2c3043aca2abe7102f81ee6c : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/148d610b34bb457f9305e8035d9adb2b : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52d260dc63194c5d8a4cf3bc3e4413c9 : SUCCESS in 1h 29m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf8d2bca0af043fba87b026d9a14aa63 : SUCCESS in 1h 31m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/65abcb65ea5b43548bff11ec2fabe1f8 : FAILURE in 1h 10m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89300c3bbacf417e9d3a63f659bf7ca4 : SUCCESS in 1h 55m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/de9dda4e246e42a193f339fc5179e8c5 : FAILURE in 40m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d2d49361cee48079c7d91c853b8c877 : FAILURE in 37m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69daa751c3b84f66abbfe68ca5293c67 : FAILURE in 43m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de2125b4d1d4c9ba5eba09db34ac489 : SUCCESS in 5m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a00d78a0651d4d6f97a6a47d4a4d0e56 : FAILURE in 1h 14m 58s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/444652a5154646feb7140d7b2c59877d : SUCCESS in 2h 04m 02s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac141dcb5d934a25b9a95a73a293fc20 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b61e0694a5df4e2ba17a866fb65a83aa : SUCCESS in 2h 05m 45s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0f07537d81f4030973629ec2e6b99d0 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8887660a5cd4365b66f212b7f80f772 : FAILURE in 1h 49m 02s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0379f70107874b8ab8a2bdd1f3267ab0 : FAILURE in 1h 04m 05s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6abc7f0d0a0c4f69842267d162218d55 : FAILURE in 2h 17m 32s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acd34d9fa1854444bd9f7df03971f617 : FAILURE in 1h 09m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c67f412b07f32a5c5e9afec15a4d5146f744678e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 20:20:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e7960b42094cecc127e2525003d0dc15411e60a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 20:25:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8dc4cdaf200586e482eb78d423d6fec0cd63c38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 20:26:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"da51ef038358c6af0bfb30e23faddcfa776ff77e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-28 20:28:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d4baadc8f6cdcdfc00616b647d4db41f68d9385e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 22:17:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3cdf91fe40e403ca459f9f2054ba37a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5359a84379a14f259978998f03cb4276 : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fe6c56b0f0d64ea5a14ce4bebde3a30b : FAILURE in 1h 00m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/eedab412a0254776be383631f3693b3e : FAILURE in 59m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef069c60c29848249d89c55fbf85711b : FAILURE in 1h 09m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a50c432677b04dc2bd0657914c061803 : FAILURE in 1h 08m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a52ee65ab635425ab6b9e24c94d8da36 : FAILURE in 1h 11m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c4e228338c843e6811dd6ab2546ff4a : FAILURE in 1h 10m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/373b5456900541ab810e82fad8e69775 : FAILURE in 1h 46m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/91f0eb6911d2451b8a1b57463a69cd45 : FAILURE in 35m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/edf7cda902d54398bd315cfe13bb15ef : FAILURE in 41m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/358a557aca9642178655e39e3703083f : FAILURE in 35m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bac2d2b15d84cf6a9257b9345c08537 : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b508769e730249b992e9fe9033001c8f : FAILURE in 1h 20m 35s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f707132c354286b8ab1ebe060297ae : FAILURE in 1h 45m 59s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3baa42cdc6d4982b36b1478a0af2373 : FAILURE in 1h 17m 09s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/66f178a52c6146c0b5dccb44626a8690 : FAILURE in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca07956aa7894270bb4c6c0b3c1a72b7 : FAILURE in 1h 12m 22s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3045ea5605148d59d509b6aa81b0950 : FAILURE in 1h 36m 59s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd29581571e74853aff7db43f9645bcb : FAILURE in 1h 11m 44s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c12c7dc0bd744533b2c0d8a3deb5d845 : FAILURE in 1h 47m 36s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c6089f34d3f474f8aacbe522ddcd38a : FAILURE in 1h 18m 10s","accounts_in_message":[],"_revision_number":6},{"id":"15c9f0956b5ddc76b1e79532241ad4840f4baeba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-10-31 14:22:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"10b1c082850ca5bb84e8678096684db7f488d313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 15:41:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a69699a82fba42569f09eb1ff2853808\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44e8b7f4795942faa2c84ae2681953fb : FAILURE in 15m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/abf2f750c4a04c24ad2e297ef901f2b9 : FAILURE in 40m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f45a4b8e5b44d04b598c3c940b745df : FAILURE in 41m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f4617265a59f4ecf9929c6dad28f33d2 : FAILURE in 42m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d44d1f91a63f47f985d6c3e212e3f1c2 : FAILURE in 40m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f966c36916b4759b5e0b0c34cb64005 : FAILURE in 42m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1a8ebacb6a44010b0db6b88c9eaf52e : FAILURE in 1h 12m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894110fea27a41d1b49ccd30428733b6 : FAILURE in 1h 12m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e79a6ff8a7247ab991dc3fa8a7fe9cf : FAILURE in 36m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a78dfa6c478749559c0d02c23ad5c27f : FAILURE in 30m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb3a97df8bba4e47b1c108004f77b41e : FAILURE in 39m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb4c428cb14401bb1f5c2dcc371d5ac : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a1943baff544b394276a0ea88bc477 : FAILURE in 41m 06s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c265a4b14c04f1d917e3dacbfcdf78d : FAILURE in 55m 33s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a802324e96dd493996eb22fdccf11165 : FAILURE in 41m 07s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e4655d00c16439aae0d5a2198cfbe72 : FAILURE in 56m 17s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5ea75a1e17f4b50b50e430d5bdb262f : FAILURE in 43m 06s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/21a60abaee7b4b628177374f0d75e16f : FAILURE in 56m 28s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c8ee7901d1149b0b886a24efddbdc70 : FAILURE in 45m 50s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11aba8ad54f44fd19f8a3e8f0d1e5b69 : FAILURE in 1h 02m 12s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/901ecd4ca5dd41ec986272c6a24c3a50 : FAILURE in 46m 16s","accounts_in_message":[],"_revision_number":7},{"id":"e584b8bc41b5af4c862f75167441aa933eaacdc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-02 18:58:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"acf59a5aeb9bc7ddc07869c162529eefe02ed7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-02 19:03:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"12dc26a3dc8d7483adba73c72d7624839facc0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 21:01:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a343c657f5b4226a9f14200e7f6623b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76fe31da484b48008953c2ed1aa50a12 : SUCCESS in 14m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88ba6684a46d4fe1a0e689651cbab799 : FAILURE in 57m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/00706edc8147448e9e866abcbb01b4d7 : FAILURE in 1h 20m 39s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21c5781515cb476d8f3c21696e020ab4 : FAILURE in 1h 14m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93adf9847aa74c469ab4b37f0287e8b8 : FAILURE in 1h 03m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71ade797767c4a71b7b146e0ba3cfd7c : FAILURE in 1h 14m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a0595fa2b7743828cd48f40df7f8636 : FAILURE in 1h 11m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93b5465bf08b44139bc45887b347a54e : FAILURE in 1h 43m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa99cd6ce744a85991cedb490457bce : FAILURE in 41m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e5c729c0cd643559a3aff6392f60539 : FAILURE in 37m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff8818c69d24f59b84dafcfc4e0c711 : FAILURE in 41m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5539e4e96545d081f2cef93f6830bf : SUCCESS in 5m 19s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c81ae1127ac34e1d9cef9f327973c783 : FAILURE in 1h 16m 32s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7627c459ef494a9aa51b606b5628173d : FAILURE in 1h 46m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4e0cc1c89e4bae9ae6540d26154ff2 : FAILURE in 1h 02m 49s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5dbffdda77e41008c3fbe7150dc666f : FAILURE in 1h 36m 44s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c5c33e30c954962b95915169bcb4617 : FAILURE in 1h 03m 17s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c18bf8b7d2a476da1a8e956966192f1 : FAILURE in 1h 50m 16s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/47fa1b4988eb4635a2aec9672943c977 : FAILURE in 1h 33m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cded04d58183449480963c98806f0a56 : FAILURE in 1h 44m 47s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e7a993511154722bcf3b8aa863f8934 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":9},{"id":"f92052f8b1831d092960f14d3c13a99d3d7de1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-03 14:19:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ba656735f6e5b1b2bd36873b77998fa7980d12e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 16:32:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1f855a9e6964964b4a307636494eca6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d18c7f91903b4716a29e0ba08dd52f8c : SUCCESS in 12m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/94caf997829d4ea7a4f79d522e4f10d9 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c668be4daed416e9cc9964b6d6161bb : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd96a63854314e7ca2e08edbfa4623e9 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82e6104830a04858b40b968426211529 : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a49c0c68506b498aa59c689397230617 : SUCCESS in 1h 27m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/875e6ba9c39f4ce09d7eafa9384be333 : FAILURE in 1h 13m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c768a9d9a2d45678e9cb11698543cd0 : SUCCESS in 1h 57m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b5cefd247884bc1b97be748221b89dc : FAILURE in 41m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cddbab0df2764a0f97683b816879bd9a : FAILURE in 33m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/911d9cb2a7db4f33b33237221e82a488 : FAILURE in 38m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea58a69f2fe41ad82373b0605d4a788 : SUCCESS in 4m 34s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81d48cb92468447b85ad4164d43a5ba9 : FAILURE in 1h 29m 31s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2c5f8ec3e62458d94fbcb5d2c70f079 : SUCCESS in 1h 54m 21s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aee7ef53ca1941dbb9ba37badd42aaf6 : SUCCESS in 1h 30m 25s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aadc34d28a6437497b10e1837e40855 : SUCCESS in 1h 46m 42s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/335bd93294df44a68790ffb3bb3ac052 : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a1f05eefe944212a4916a30aa01cb08 : FAILURE in 1h 36m 51s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/15cbc1cd6ed441349652c14158967aca : FAILURE in 1h 07m 11s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edb87f4e16bf49b287bb2fa46b3145c1 : FAILURE in 2h 06m 13s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f272a5a22fd485db28edfe043759614 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":10},{"id":"bd14daf5295952741ab87ce336ebdf2978bbb785","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-03 19:47:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bf2e819ce8ef28fd92f3e9d0e74439120e43a846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 21:58:51.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e23b3e0f07f04d87839adaf9097d0b0f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a7faa093ebb40c4afa6bdb6be74f5f9 : SUCCESS in 13m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbd72aeccb83418bbfda69b0f18cc0a3 : SUCCESS in 1h 27m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2d512a3746b48bdb2cea12ed0eef343 : SUCCESS in 1h 32m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4fc8cb22c53d4c28a9d41e8a289591bc : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5c383f1c0e44db0ab47e48ad3553ebb : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/444e0a2906684024aaaf0ca60d74e825 : SUCCESS in 1h 30m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/53fbe142b9b548eaaddbbcb0edd2fb8a : FAILURE in 1h 13m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24c77af4dce46e199522aeab7f36122 : SUCCESS in 2h 02m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/804f1faac0fa4b24840f721ce74eb784 : FAILURE in 41m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6be56f40ff554b95a015f7037be626ba : FAILURE in 33m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dfb46ba3b7a4ad3b8541e9a740beb13 : FAILURE in 38m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2eb1e43aa70417db5240f400b27940a : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e450f287fa5b483eb926cf55575d62c2 : FAILURE in 1h 30m 23s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89f227fddbcd4109bd83f83120c6ebaa : FAILURE in 2h 04m 42s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99608da6d288446fa8a88ba42157ad45 : FAILURE in 1h 26m 33s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ae643aab67b4ca4a182513f6a679c2e : FAILURE in 1h 56m 35s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6752845dc8d04a89af3fdd3851ee99bc : FAILURE in 1h 24m 49s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c4bfdf024f41a3928985a31f3b4759 : FAILURE in 1h 42m 26s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/824fb91780c147eb8004e3c392781b6e : FAILURE in 1h 20m 39s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d26255551ed14ef8b36d24712f892f00 : FAILURE in 2h 02m 19s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/484eab41e7e8450496027970c3640653 : FAILURE in 1h 23m 30s","accounts_in_message":[],"_revision_number":11},{"id":"455535cdef76d07cfdd269a7a633e7c968be8799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-04 11:21:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e92ff0b62ff08c1a1c3435b7a1b12b0a9ceb2104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:31:27.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4936b4b8534f459aab13e5e487f3dfd7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa9b0a01f23946efb420971e12c1a499 : SUCCESS in 12m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7d8d2562a7934dca9d95a6c097b6b976 : SUCCESS in 1h 18m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e84e9e2a65c44809eec3b4b2c22d15c : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68f697185059452684242d63f2d64991 : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92b58fdce78483bb3bb13ddf0ddf5eb : FAILURE in 1h 38m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc8822273dc643d0a098f94ea66e0ab9 : SUCCESS in 1h 15m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/46a450e445fe43959c2a60b8f65d5bb7 : FAILURE in 1h 09m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed2535a030b143a387a71a7e9dfa3563 : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/17d69327c0634f07bee1f22f56f4962b : FAILURE in 42m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ff45f82e84a4d8e81beb87a1030cf52 : FAILURE in 35m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e890e981a6684216a80a3c62a3a4ff06 : FAILURE in 42m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd45282fcbac42d9bbff99b89b3266aa : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d9b0dd2610341d5bd266c80424faec7 : FAILURE in 1h 26m 40s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e72898ba42f44e3380b438c2dfa6ae62 : FAILURE in 1h 47m 31s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30621ab3aadd46d5bdc8d860a2406321 : FAILURE in 1h 24m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/292ae94d7f134c668b5011d8dce91ed1 : FAILURE in 1h 49m 32s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ef5462fea9447c9859a9c2100bbe0b2 : FAILURE in 1h 16m 48s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/eaf349600d9b4767980da58131e3927d : FAILURE in 1h 39m 16s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/da6541aae801404f9bc851b00195a03d : FAILURE in 1h 08m 36s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7152fb16704948d19890dc435aff56d5 : FAILURE in 2h 06m 26s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dbafdba7b7b453c9e0c3c8b07372d48 : FAILURE in 1h 18m 28s","accounts_in_message":[],"_revision_number":12},{"id":"47b955a5d055162201d23e965ea33d30bf13648e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-07 13:53:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4e56bf337ce7c3dcde607e7410c8bf6f43418b83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:16:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33b37206645340f5aad259cbe3ba4e5a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f87d8b11bf04a11be8249f27f0f16ad : SUCCESS in 12m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d86e9619381f4aa6bd8025e80ae242da : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/33187dcb5b2e462aa1a16869031cf7e2 : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f768dc926f064207ab386e583d2edc68 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e658fa99771428cb7f3a798797e8f93 : SUCCESS in 1h 27m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4fda6faa0014d17b688a569bfb12f6b : SUCCESS in 1h 21m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/defdcfe10dac4578807f803a6a55d708 : FAILURE in 1h 02m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25d527f69af54b10b5f15d2469156856 : SUCCESS in 1h 41m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6521db96c0574f4a872dfb49580f5269 : FAILURE in 36m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fad8ee14bb054f5392cd0416baa35a85 : FAILURE in 39m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/859fe928d752499f8d9ac81fe060cc42 : FAILURE in 43m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333000b128124eaba7d737dd4c33e15b : SUCCESS in 5m 02s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef9203a213c94f7cbc0f60301b7666d8 : FAILURE in 1h 26m 35s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56090effbad4d34a5787d14f10f79b5 : SUCCESS in 2h 19m 40s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be4927db1ef949069bb79b1beb352b05 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7844686382424dcdab31e9fda9b34c17 : SUCCESS in 2h 05m 58s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/002382f6b00641208f1bbebfafcdc35c : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef8e99827d5b4ca8b9ec1dba121923a8 : FAILURE in 1h 43m 47s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/372642a835a44aeaaef6d0b345b18b48 : FAILURE in 1h 17m 18s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c66ab480b4184c97a6689fcdba9fed64 : FAILURE in 1h 56m 34s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88cc68305d0044958ec5bd6ab9fd2a34 : FAILURE in 1h 24m 04s","accounts_in_message":[],"_revision_number":13},{"id":"022f56e8b508c63323da32e1f71bb0016e91cd9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-07 18:43:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0b7a1a4034d0f40d2dac3484b66357121311f4dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-07 20:59:07.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"43e518d70e0265b1c4e831580c93023283644506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 23:25:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/690dcd12bc374f02a0ad0a1b1734b07f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d49a4d90e274b15b2d8afbeeb0e8b6f : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2f597608a8ff42f6ac95e51cdc1040dc : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f705552be7f4c42bfcadbe1bd7891f9 : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/008e932f381b48d496f062a6ccb14f09 : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e9dbf71a80f4e858f30018857d50d8c : SUCCESS in 1h 38m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c0815da7c33429289366be91c4279c8 : SUCCESS in 1h 33m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1d1e8178cba45bfb0f2f24bdf4f612f : FAILURE in 1h 20m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec17c06898a146e7b44738ee7afbfc70 : SUCCESS in 2h 01m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/183e71b77f8843d99b1ef014f1e1967f : FAILURE in 42m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/259ef27af8134063a204dafabf402aef : FAILURE in 36m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e28dea35be4d44c69effac0c1f63f8a8 : FAILURE in 39m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27a19a9114744d1bca5a6cf4e5557c6 : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e313c027c5914c6d92c77a5efccc3ecc : FAILURE in 1h 31m 44s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a31e2b2ddc024ce2a371952377cf0c8f : SUCCESS in 2h 23m 18s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e72df115e6c4a50914ac1f2e5afd018 : SUCCESS in 1h 34m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/64a5a161a7084b3ba88c6c3d03bd4325 : SUCCESS in 1h 59m 39s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/01df0ff8db824b889690c9b39f79ac44 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0390f83865b445af98979f12956adaa5 : FAILURE in 2h 01m 09s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3d3d2ed19d34fab98f8abc548f30a79 : FAILURE in 1h 23m 24s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea6aef9758f442f9a2d089605cd0fb6 : FAILURE in 2h 15m 53s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4e9c0156b264f3faffe0708bbd3fe66 : FAILURE in 1h 26m 02s","accounts_in_message":[],"_revision_number":15},{"id":"72d772e9bbf8d957a68db60fec65b8eb46824ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-11 14:19:09.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2ffe9e325e9990118cda4b7e7bb3ed43acd99f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-11 14:20:34.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"1401161d4753a9716efa4027e76fc06d908e7d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 17:27:58.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4203578bc5b74720acd8cf23b3ee212a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed2e7e459018454a94ff50788f9473d8 : SUCCESS in 16m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f0755c007ca4f08832f9a552e0c8204 : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4089bfa2e34948f8a616abcf4bcd1a2a : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64f4ba3d7fcf465cba13acd07f4898a4 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21dd22a6107a4477a259d88e6ac25b88 : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aebcd1a789141558b0505235c7657b5 : SUCCESS in 1h 32m 31s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/137096cec7144a5b9d573003459e6dba : FAILURE in 1h 16m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e61caf4df3748f4bda08bfed9a2969a : FAILURE in 2h 06m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/99f7d8b84874405d920a1af92dfa010e : FAILURE in 42m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76fc3520a74d471abb28028d7af48088 : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad4e62d179794633bc8e75c52654ec73 : FAILURE in 41m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854438d0690f4407825d47244109ac3a : SUCCESS in 5m 06s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dff89b7961b8428e9f03d69054ab54bf : FAILURE in 1h 16m 37s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3760e6aaf55141b582d80f5d02271864 : FAILURE in 2h 48m 57s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c771aa948d8241e78a8fd6b6c22b039c : SUCCESS in 1h 31m 45s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd7619d731f6433ea248dcc93ce1415c : SUCCESS in 1h 38m 01s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/777b944e6d6e43459a45646d2934d3b6 : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/70900af553ac4ae990b426dcc521b45b : FAILURE in 2h 05m 40s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/203a4167b4064a2ba016ab422ef2792c : FAILURE in 1h 22m 38s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9773f1b5e2b45aca4e3686c7c31bf99 : SUCCESS in 1h 53m 38s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/081ab777181648a99d1b55b34e651cdc : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":17},{"id":"539f6357ec669f38796a2b272380ed755756a160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-11 17:53:06.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"6cf473b52849d08452b3beb106f81a240160ad19","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-11 18:23:52.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"12a83d75adf4214a4ec24465791e2dfd30ecfdb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 20:21:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79505e8aed414de79b029d7a23066703\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dd1f9eaf2b74656921c2c00725fdf96 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1f6c02a263b0478098b86ea1aaf76157 : SUCCESS in 1h 08m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf4655edd5d84e13b086c779b9473091 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9d211915d0342fca18e9827623cb9dc : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79508eb0910944488833a0dac0e53749 : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa278922a33d4e0a90a13da774477965 : SUCCESS in 1h 15m 09s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbb4dc0131c44ee8a1737a68509d1efb : FAILURE in 1h 13m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4b9344abdd74dce82eb088038962fa9 : SUCCESS in 1h 38m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/317639ed45ef461191ea5f6dc4e9fa62 : FAILURE in 42m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7a698db3f264e97b7b88ea0854884cc : FAILURE in 35m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c35f7723d7284e8db543911dc079255c : FAILURE in 36m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8025c3665f7433eb3cafdb9cf273341 : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23f8c56c83a640dab01c88cc370ef10d : FAILURE in 1h 20m 47s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a653d70f0d864996bcb10415b6424e9c : SUCCESS in 2h 21m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26210cade1474386be22db777c972a3c : SUCCESS in 1h 23m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b493407077644bb58c4700df8de34045 : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/80a56f6f97ca4771b11fec66f0d423f2 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5f5f8278c9a44e3953a3a152a92d612 : SUCCESS in 2h 14m 06s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/772a679ad5fc4ed198cfd88b4f86b44b : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9d73738d4f846f088fc0ac80f7a4111 : SUCCESS in 2h 25m 48s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76533923b88f418f85c58b2a4cc9fa5e : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":18},{"id":"7511e88804b828afe08262a0fbdd7ce002fd83cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-22 14:29:16.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"ff3cffed8a06eb035e5b710a028411f3a3384326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 15:38:21.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eabb78278c5c45d0905566dc2177f669\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/734e614db10d408296655d11d1874ee6 : RETRY_LIMIT in 11m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d8589a48779644b5b91da0240be13a1f : RETRY_LIMIT in 9m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4b5567cc4ee471ab679ef50407fc949 : RETRY_LIMIT in 10m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/89fc4395754d40909dd89ec3ad872ac7 : RETRY_LIMIT in 10m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ad3c54e36474ba18f10776f7f68f6db : RETRY_LIMIT in 8m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/996bd4671a8b49208f8318c6a5378c57 : RETRY_LIMIT in 11m 10s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8be566ca12a44297be80105c01104d73 : FAILURE in 11m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2e9e5f061414fdc87a3f4ae3b686810 : FAILURE in 1h 06m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/62cb06b4e5784aca936d5a035ef11957 : RETRY_LIMIT in 6m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cc65b3fefee4fcc8d180adf8f9a05b2 : RETRY_LIMIT in 7m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2bbed72811c448183318935bac01479 : RETRY_LIMIT in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472e68afbdf3434784c5d31e2f83cf0c : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c77c3a67e9b048bd82d50010ccd1a054 : SUCCESS in 3m 50s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62fc041ee7a440cdb33538a79ed28821 : RETRY_LIMIT in 8m 27s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/515ff3e8c53c40e6aba317fbd47167b8 : RETRY_LIMIT in 8m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0914fc47448f4c259f19ff5f4c74155e : RETRY_LIMIT in 6m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e8f47db09a24e6fbb86b6f8b89a706f : RETRY_LIMIT in 9m 35s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/54ff61acdffd4387b72aa42eb6cee96f : RETRY_LIMIT in 9m 06s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d39a1ed83cf48bfa89f62eb3bcfc774 : RETRY_LIMIT in 8m 40s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/566c9fa1f2904c5e89390bb87016e3b7 : RETRY_LIMIT in 11m 42s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad329dbcab74d47ab6ea03a6e9ce037 : RETRY_LIMIT in 8m 41s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53efd33a56514fc299ecbc66baf15cb8 : RETRY_LIMIT in 8m 17s","accounts_in_message":[],"_revision_number":19},{"id":"7fdde66e94be3cbe3f3fc288e85f4d1c8d65f4e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 15:43:22.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"bf8b45cc2a549c0d8a83236096af4063c16b04a7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 15:53:17.000000000","message":"Topic ovn-ssl removed","accounts_in_message":[],"_revision_number":19},{"id":"a4159034795df8d5c7ee6831f5fa9ba7a9616844","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 15:53:20.000000000","message":"Topic set to osa/ovn","accounts_in_message":[],"_revision_number":19},{"id":"1308855fb6bdbef155bd179c3a792d9c991f9400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 18:02:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467ce45bfe52448485b27908f3fa29a3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dfadf2fa9b2489aaad7491af4463525 : SUCCESS in 13m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88e8a5efb3cf4cea81f1abbbded46ca7 : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec105f50fecb410782ca884740900964 : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/82314ddc56af4372995c1d3f3b7abdef : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/227f6d94ccca4cebbe27f755fe01ed3b : SUCCESS in 1h 37m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a726b7a7efed465e90cf2e39b3831b4e : SUCCESS in 1h 27m 24s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/925929e496614fa7990ee8ef6427f4a6 : FAILURE in 11m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c805360c2d146b49f3396bfd6483eb9 : SUCCESS in 1h 52m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cff87be3a86d46deb6f09493da0195fa : FAILURE in 42m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed117b990e0e47f190cab4c088bd1ff9 : FAILURE in 34m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52bf7e9af1ae4c7e8e45eacfcc3e2c2e : FAILURE in 37m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21c6f2858a547a099709038693efbfc : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f36863e0c3744f5e893628f8a9c6eaf6 : SUCCESS in 3m 11s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/543d6d48c8c5423f9e91ecf54eac95d9 : FAILURE in 1h 27m 38s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814761784abd48b598f109f142098df3 : SUCCESS in 2h 04m 57s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12118208864446b888cbeb8ebc59ff5a : SUCCESS in 1h 47m 17s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/eac1da3fff6c4b75b475b3506f7c0d85 : SUCCESS in 1h 59m 05s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d9e972b53b54aeda405674dc4e2c746 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/eab90b7308674b208084d03241227ae9 : SUCCESS in 2h 08m 26s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f771867c3245ed97f361ef98e2b918 : SUCCESS in 1h 30m 01s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83b46b612b084a88822428d90ba7db2b : SUCCESS in 2h 12m 40s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f34b94a079d24d7d8f32205c9be6f432 : SUCCESS in 1h 28m 29s","accounts_in_message":[],"_revision_number":19},{"id":"1cb21e2fdd95e6ac700f67cc73998e0acf46125e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 08:39:00.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"8f99aaefa5c3adfc560c33cd23ffb44029503f6a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 08:39:49.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"c81e9bbc79985a033213a9cc44982f043bc05810","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-29 20:06:57.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8ed030cff2a76a6c43643a0727cf48516ae5279b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-29 21:19:32.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"e1178b174a51420b6416a5306fe082f7953dbf20","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-11-29 21:20:49.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"6846360661fc7cdae116643eed999c878322411e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 23:45:40.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ffadee37514439ab4d36ff161211d4f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19c369ec9c5a480b8411009c34d09131 : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0fd3409f88e14d31b7aba9b96876ae0b : SUCCESS in 1h 19m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a75acdcd0e64e19bbae30aabaa69236 : SUCCESS in 1h 39m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df7bca57a29c415e819b019815b0e685 : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ee55dff15bc4d9ebac8feb39e6d5942 : SUCCESS in 1h 43m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f023bb6a6fa4ded8989945f174821c0 : SUCCESS in 1h 23m 33s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbc425366e914375a8f891acb7600b80 : SUCCESS in 2h 00m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56698f710304513877d11ea1294c230 : SUCCESS in 1h 46m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0ed10627519460289f4b62c9b0f561b : FAILURE in 46m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ac24348355d4d74b533d7f1119d2616 : FAILURE in 39m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2718e17eeeb940feb13b3a5ebcbd9680 : FAILURE in 38m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7d1208d7ea4ce4ad3c614bfd149373 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dbc7640de8141e5a6a59ab98d89f505 : SUCCESS in 2m 58s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb6defc4c7774ef39fbadc0175cfd3ab : FAILURE in 1h 12m 17s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97ab9277a9604d489ef11c806e0a4a0f : SUCCESS in 2h 19m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea97697601a943b487286b824acb7a95 : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a85d539c5d4c409c85af09a8a5d450b7 : SUCCESS in 1h 57m 06s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/af338e69b9084a88b6e23ba95e4a3c4f : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfc817ac626b4975b7ea25d749d58bb7 : FAILURE in 2h 00m 57s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/786cfcfa155544f18111d07c4181d5e3 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cb3235ec85c434fadbb0308bb5f7105 : FAILURE in 1h 57m 02s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace324925d504c2ba777fcc4d46e8a47 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":20},{"id":"4252b18dbcaf40b4baff65955397655119a74cd6","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-30 03:51:41.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"0bf20e5e633d6ae43318efb36b2aa55ee1a58493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 06:02:14.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4cdd67fd9ba465190f67ea952274d73\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a7d25115444843a40942823186ae6d : SUCCESS in 15m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d7f4b5d88c9f403ea13d22b32b92df75 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/02962edd15e14bfea2eec52b2508dcce : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/540d1209f4284a3793813cb643a59b3c : SUCCESS in 1h 18m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f78651eb29e4acb9fb099a5b5c5dee7 : SUCCESS in 1h 36m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc75834ef5894e23b2e52b6eb344a1e2 : SUCCESS in 1h 29m 32s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b88c9e5428534b00b3bc884d356ac335 : SUCCESS in 1h 39m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15c532786f6453d84ec1cd8e0e5dea7 : SUCCESS in 1h 43m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2e08e3e5f2c475b88106714935b4aa8 : FAILURE in 51m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2afbc24959f4d03b7d23f25ce3f7cfd : FAILURE in 48m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f1524db77143ef8aa2491f9834e231 : FAILURE in 38m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b895667d518744efbab5cd07b7665754 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7605c9239e34723b33c192be658bbc5 : SUCCESS in 3m 23s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb92927e9fbf4266972a2a2aa358768a : FAILURE in 1h 20m 35s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0b9e6c85ec8412687d1dbb7df502260 : SUCCESS in 2h 00m 00s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3041c564a1604efc9a167b16793ef1b1 : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ac3a2827fd147b9a931ee9459cf1388 : SUCCESS in 1h 51m 50s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/2541a4c3838f48f9bfc38664626f7a51 : SUCCESS in 1h 31m 39s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/80ba4306ec274a95a850da4c8de90338 : FAILURE in 1h 53m 51s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/07dff6c1cef54b0e876c4baccde10c45 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55d001cb429e48b78a5fb5a81e2db53f : FAILURE in 1h 57m 54s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7e44f57d1354ba69cf01bec336ee7b1 : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":20},{"id":"4d1aaeb951792111e815f0cadefbb4c692a00f3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-30 09:25:35.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"7940c1de9a607036a0020c09972de77c708b874f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 11:44:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48a88b2ce9e1456c871a8675a03c86f9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4510b2c849c64fa5b95c5ef4eb9afd74 : SUCCESS in 16m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a54a128d972e4ab6b317510cba2a84e7 : SUCCESS in 1h 30m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/88572bd2d0e5463385ab301e4e610478 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f4de7f27e3241ae8011d9629205b088 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/908372050d024d1ebe0f8225c6c85b63 : SUCCESS in 1h 39m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dbbed966dca4338b365a057ab71e836 : SUCCESS in 1h 36m 47s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5793b6249562444d9495f1edaf9ab1d6 : SUCCESS in 1h 50m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e066a972fac4ec79b7784378354f7c8 : SUCCESS in 2h 03m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/15610ad75053494da695c13cf0fa002a : FAILURE in 47m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5d112ee493444e48c1e391970c0e8f3 : FAILURE in 38m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce81a859687848b2b31207136c45ef92 : FAILURE in 43m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afb32dc84ce740fdae62375e20312a76 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/929e7c09440441049454acd0f9674d93 : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce77d8a741b847a1abd002046d079c3f : FAILURE in 1h 33m 23s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c979401fb053493b8462be74f4ab4509 : SUCCESS in 2h 14m 53s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/395181625e984847b97039ae6271eab8 : SUCCESS in 1h 37m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c744e06184104a3d862372e05b18daa3 : FAILURE in 1h 53m 22s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4d1fab4285248f1a17e3c2e5c3c789e : SUCCESS in 1h 31m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5137302b71fb4c8e87f83a55742d7e5b : SUCCESS in 2h 00m 43s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc8bb5214552406586d3822e8a59b3ce : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c707e0a0b1524fbc8ce08bbc1cd10902 : SUCCESS in 2h 12m 14s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db7e3ce32ac74439ac27c19b92e29989 : SUCCESS in 1h 34m 12s","accounts_in_message":[],"_revision_number":21},{"id":"526bad0861ceb2a5f4568e9ec14a65323fb0d6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-30 13:30:48.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"496860044e762aa82a9898adf549eab82693676d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 15:41:00.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55627e75aaa04d8aa471f154293d87ab\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eff072cfbeab4711ab4079a593f6a030 : SUCCESS in 15m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d70698e3458c40cc8507b2e73e4e2123 : SUCCESS in 1h 31m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8595480eb5de4bddae5d0a6a27636575 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba80bbfb30b34bef803113227ef53e85 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b0fbcea082f425ea15ee960e8cf0379 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cea7606170a43c7881f46f75f1fe032 : SUCCESS in 1h 30m 13s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ed167cb6d664cf58e359a66ad8efda4 : FAILURE in 1h 45m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59f27ced0a6e422da28f009bcaef9779 : SUCCESS in 1h 59m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6fe545423144c2fa9d710167fa0b126 : FAILURE in 44m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb4bdcec40d74de8b69bcc9fd0413b3c : FAILURE in 45m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dbff4aea45540bdbeb60ba75ab9033e : FAILURE in 39m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddfd66a3b3d4f61bd50a41af5d9c129 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be49eda9a0cb453dae4b1ac786a4649f : SUCCESS in 3m 03s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7228527681084874afaf01dda92e6e31 : FAILURE in 1h 36m 02s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ca03cd36dfb48cb9e6e53707fbae48a : SUCCESS in 2h 01m 49s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c6301ac2094098848094c2ad93ddc4 : SUCCESS in 1h 35m 39s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/23af913993bb4eec8add5b2150098fc8 : SUCCESS in 2h 05m 57s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/722cadb796f54326b2cac3a8a6b331aa : SUCCESS in 1h 30m 33s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6896ef2ae4f49aea51dd35607ca9188 : FAILURE in 1h 52m 20s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2ae930e881e473d8df23c22957d738b : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137ad627b3b94f0b8dc43e1f342c0a05 : FAILURE in 2h 06m 24s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c114a2290444c789b506299fd110115 : SUCCESS in 1h 35m 48s","accounts_in_message":[],"_revision_number":22},{"id":"750d55fffc6f9d7db5a506baf38fc8f856c77df0","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-30 16:03:54.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a5a243e924596b05d178268237a1484a2561eb3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-30 16:04:17.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"2124ed78bc6690a969cc7225227278fd775ddaa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 18:16:15.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221ce2888a2c4028822495c908f42596\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/809e8004d9b04fdd9753d56da067a4c8 : SUCCESS in 14m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a899f09192c443dbec0c24287803f7b : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/171948c05a0148718825572ad3643d9a : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcf7c0904f7a4900905243927266ab75 : SUCCESS in 1h 34m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da5ba6bc37624ca296ac690cea771292 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66318813dc4143ef9fbeb5a42a239854 : SUCCESS in 1h 22m 22s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1761513246cf455e848f4db1243a2f05 : FAILURE in 1h 53m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb7e8bcf341e40959f1f2c1daea342e9 : SUCCESS in 1h 40m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/545600ebd222437f91b1f92a31256d77 : FAILURE in 43m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/23f389f1ca954de1b0b411024fd1f3bf : FAILURE in 48m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0edba0dc91e488cb0101468bd8f2b8a : FAILURE in 37m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c879ddeb78cd49129a5ee07dbaad5246 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a84d33651c1247ba9a32dab01e02b7a5 : SUCCESS in 3m 59s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a535280be6984b358dd764748da81423 : FAILURE in 1h 21m 37s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159f9047e3cd423eaaca9786a539b586 : SUCCESS in 2h 01m 43s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f5efd1c5034ef9a73ce3c53f89b943 : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8598237a0e1b421a87bd935670325ceb : SUCCESS in 2h 06m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/49bb699addf84d2c9abac80fd9305afc : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcd736af23c2493d9d98063e841e8464 : SUCCESS in 1h 48m 39s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/33aa1b5e52aa45e8b358aaa0861f6969 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08e8c3c1416b40f6b463b39c13c1c77a : SUCCESS in 2h 06m 44s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89f1a07cc57b40b8afd043b56d4ab93e : SUCCESS in 1h 25m 30s","accounts_in_message":[],"_revision_number":23},{"id":"3353b41e2f0a1efacd051d9f6050a270b0eb85ca","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-11-30 18:57:35.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"af3a5d1f98468d45a62526c90eb0323040722d45","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-01 08:58:54.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"e11b3b5356c08b8823f52a29eb55494364ca9194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:00:53.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"b5399ec32014e8615be090549bbe1fa8497176e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:23:56.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfbe9538b86c4f118dc517d01a81208c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6404f9034fa94b87a85a6f6d7b540841 : SUCCESS in 14m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4c46e1b23efe48cfbd6add76de8fed9b : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/736546a5f08947cba29c3519dff48f1d : SUCCESS in 1h 31m 21s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c6d3609b7f9941e5b59613c3b53ed7c2 : SUCCESS in 1h 23m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79500b4a28c6459a80b551a9e543af4e : SUCCESS in 1h 19m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9ac4a989ac45b0a212d6e88c40fd17 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b07636d499343e7b55983d84b876dba : SUCCESS in 3m 16s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3fffaa6b9034771a78d43866f9ec0ee : SUCCESS in 2h 21m 49s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc5c9ea1f5274209bf503c3ca5ec8d20 : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56f9ab4e8cb4be29b246059aa20e8a5 : SUCCESS in 2h 05m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbdcb3c936f141e1b437ef6eff455741 : SUCCESS in 1h 12m 31s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9680b0de9f854370ad0e7a606eafbf3c : SUCCESS in 2h 03m 49s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/32e6e74dd7934c6fb09f9de6473b3034 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/359bc3606fbe4ff3bd4d5bc4c217ebe1 : SUCCESS in 2h 16m 14s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff44e4afd39e458b9dc45cda1c820d5c : SUCCESS in 1h 40m 05s","accounts_in_message":[],"_revision_number":23},{"id":"814d628ea44be3b6ab2f49fffca8073218b82000","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:23:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"3ea0f3b62b7a9056159ffb18aa06e986f5eb4336","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:25:02.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f47f98672640338b81c78f310a2dc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b70d251ca86485894126f25cf748366 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"556c5c67338adac519312dc401cfd0febfde32fd","revisions":{"4a1f93cc89fb4a8ffc79babc4d3ad4a59f3929c8":{"kind":"REWORK","_number":1,"created":"2022-10-21 19:50:49.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/1"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"root","email":"root@aio1.openstack.local","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:50:49.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4a1f93cc89fb4a8ffc79babc4d3ad4a59f3929c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4a1f93cc89fb4a8ffc79babc4d3ad4a59f3929c8"}]},"branch":"refs/heads/master"},"abf1ee7219945eca6377edb0770d1bb420c9cabb":{"kind":"REWORK","_number":2,"created":"2022-10-28 15:28:12.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/2"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-28 15:28:04.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/abf1ee7219945eca6377edb0770d1bb420c9cabb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/abf1ee7219945eca6377edb0770d1bb420c9cabb"}]},"branch":"refs/heads/master"},"6c28b77f300c34e1b480559bdc4a132f0b7ad87e":{"kind":"REWORK","_number":3,"created":"2022-10-28 17:25:48.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/3"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-28 17:25:43.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6c28b77f300c34e1b480559bdc4a132f0b7ad87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6c28b77f300c34e1b480559bdc4a132f0b7ad87e"}]},"branch":"refs/heads/master"},"21aab4334bec7b7bdf1307ae214313af03bfcaa5":{"kind":"REWORK","_number":4,"created":"2022-10-28 20:20:52.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/4"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-28 20:20:43.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/21aab4334bec7b7bdf1307ae214313af03bfcaa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/21aab4334bec7b7bdf1307ae214313af03bfcaa5"}]},"branch":"refs/heads/master"},"4b28d9049f362d923ba3882718476ddccb929dbb":{"kind":"REWORK","_number":5,"created":"2022-10-28 20:25:42.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/5"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-28 20:25:37.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4b28d9049f362d923ba3882718476ddccb929dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4b28d9049f362d923ba3882718476ddccb929dbb"}]},"branch":"refs/heads/master"},"4a3238080cfbfc4de24d05c48fa7014154cccd7a":{"kind":"REWORK","_number":6,"created":"2022-10-28 20:26:30.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/6"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-28 20:26:25.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4a3238080cfbfc4de24d05c48fa7014154cccd7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4a3238080cfbfc4de24d05c48fa7014154cccd7a"}]},"branch":"refs/heads/master"},"106ef46c37308d38994836500858f9a441374eeb":{"kind":"REWORK","_number":7,"created":"2022-10-31 14:22:54.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/7"}}},"commit":{"parents":[{"commit":"5baba403377d4fa44f19d273c993651ca1492798","subject":"Merge \"Avoid ovs restart during package upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5baba403377d4fa44f19d273c993651ca1492798"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-31 14:22:49.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nnot working yet.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/106ef46c37308d38994836500858f9a441374eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/106ef46c37308d38994836500858f9a441374eeb"}]},"branch":"refs/heads/master"},"6cdd7faf917c7b56f63aa877f6cbb8ffc0f0e8bd":{"kind":"REWORK","_number":8,"created":"2022-11-02 18:58:48.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/8"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-02 18:58:28.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6cdd7faf917c7b56f63aa877f6cbb8ffc0f0e8bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6cdd7faf917c7b56f63aa877f6cbb8ffc0f0e8bd"}]},"branch":"refs/heads/master"},"2ab22d32ff30322576be41045667600c5c677ca8":{"kind":"REWORK","_number":9,"created":"2022-11-02 19:03:29.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/9"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-02 19:03:24.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2ab22d32ff30322576be41045667600c5c677ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2ab22d32ff30322576be41045667600c5c677ca8"}]},"branch":"refs/heads/master"},"497c59d6f06597fd158d47b483ba163c5e745fa5":{"kind":"REWORK","_number":10,"created":"2022-11-03 14:19:34.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/10"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-03 14:17:59.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/497c59d6f06597fd158d47b483ba163c5e745fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/497c59d6f06597fd158d47b483ba163c5e745fa5"}]},"branch":"refs/heads/master"},"f352093f8cf7932c02de50e3b09edfcd80477a6d":{"kind":"REWORK","_number":11,"created":"2022-11-03 19:47:01.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/11"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-03 19:46:55.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f352093f8cf7932c02de50e3b09edfcd80477a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f352093f8cf7932c02de50e3b09edfcd80477a6d"}]},"branch":"refs/heads/master"},"200c4673f980f9584d29a5b197499cc8632d85c1":{"kind":"REWORK","_number":12,"created":"2022-11-04 11:21:36.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/12"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-04 11:14:42.000000000","tz":-240},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/200c4673f980f9584d29a5b197499cc8632d85c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/200c4673f980f9584d29a5b197499cc8632d85c1"}]},"branch":"refs/heads/master"},"1fc2d94c6c2cc2ed629d478d18a4d483cad90094":{"kind":"REWORK","_number":13,"created":"2022-11-07 13:53:26.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/13"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-07 13:53:08.000000000","tz":-300},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1fc2d94c6c2cc2ed629d478d18a4d483cad90094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1fc2d94c6c2cc2ed629d478d18a4d483cad90094"}]},"branch":"refs/heads/master"},"d44e5020e5096afd64295ae790922ee7812af825":{"kind":"REWORK","_number":14,"created":"2022-11-07 18:43:52.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/14"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-07 18:43:08.000000000","tz":-300},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d44e5020e5096afd64295ae790922ee7812af825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d44e5020e5096afd64295ae790922ee7812af825"}]},"branch":"refs/heads/master"},"c32c3b371b820dea3b7a100cd5c6ed9e423266d2":{"kind":"REWORK","_number":15,"created":"2022-11-07 20:59:07.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/15"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-07 20:59:01.000000000","tz":-300},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c32c3b371b820dea3b7a100cd5c6ed9e423266d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c32c3b371b820dea3b7a100cd5c6ed9e423266d2"}]},"branch":"refs/heads/master"},"8e6577b8c006dd1b0329fb298375406cf87ca0e2":{"kind":"REWORK","_number":16,"created":"2022-11-11 14:19:09.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/16"}}},"commit":{"parents":[{"commit":"61cd24091858e617ff63b7a37983f75efeed861b","subject":"Disable/stop/mask Open vSwitch on ovn-northd nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/61cd24091858e617ff63b7a37983f75efeed861b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-11 14:18:48.000000000","tz":-300},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8e6577b8c006dd1b0329fb298375406cf87ca0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8e6577b8c006dd1b0329fb298375406cf87ca0e2"}]},"branch":"refs/heads/master"},"9c97f7cb89cb0f83af3c42fe58167a741d6e76d0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-11-11 14:20:34.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/17"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-11-11 14:20:34.000000000","tz":0},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9c97f7cb89cb0f83af3c42fe58167a741d6e76d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9c97f7cb89cb0f83af3c42fe58167a741d6e76d0"}]},"branch":"refs/heads/master","description":"Rebase"},"00ee560f0da539db322d75d9b2a34814b57fad9a":{"kind":"REWORK","_number":18,"created":"2022-11-11 17:53:06.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/18"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-11 17:53:02.000000000","tz":-300},"subject":"[WIP] add ovn ssl config","message":"[WIP] add ovn ssl config\n\nCreate ssl-certs for ovn deployment\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/00ee560f0da539db322d75d9b2a34814b57fad9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/00ee560f0da539db322d75d9b2a34814b57fad9a"}]},"branch":"refs/heads/master"},"88d79ae94801a64e409db409b0bd69382134bdc0":{"kind":"REWORK","_number":19,"created":"2022-11-22 14:29:16.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/19"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-22 14:28:54.000000000","tz":-300},"subject":"add ovn ssl config","message":"add ovn ssl config\n\nCreate ssl-certs for ovn deployment\nssl encryption is now enabled between neutron and ovn componants.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/88d79ae94801a64e409db409b0bd69382134bdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/88d79ae94801a64e409db409b0bd69382134bdc0"}]},"branch":"refs/heads/master"},"c559ed7ca2900bf1157e6fd02853292aeaedef05":{"kind":"REWORK","_number":20,"created":"2022-11-29 21:19:32.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/03/862403/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/20"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-11-29 21:19:26.000000000","tz":-300},"subject":"add ovn ssl config","message":"add ovn ssl config\n\nCreate ssl-certs for ovn deployment\nssl encryption is now enabled between neutron and ovn componants.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c559ed7ca2900bf1157e6fd02853292aeaedef05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c559ed7ca2900bf1157e6fd02853292aeaedef05"}]},"branch":"refs/heads/master"},"49f2c21ac274e16e04ae39ef1ff0e7f5152b12f9":{"kind":"REWORK","_number":21,"created":"2022-11-30 09:25:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/862403/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/21"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-30 09:25:35.000000000","tz":0},"subject":"add ovn ssl config","message":"add ovn ssl config\n\nCreate ssl-certs for ovn deployment\nssl encryption is now enabled between neutron and ovn componants.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49f2c21ac274e16e04ae39ef1ff0e7f5152b12f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/49f2c21ac274e16e04ae39ef1ff0e7f5152b12f9"}]},"branch":"refs/heads/master"},"1a340f5b67964ffbf7dd8dc60a2d2f554e898c45":{"kind":"REWORK","_number":22,"created":"2022-11-30 13:30:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/862403/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/22"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-30 13:30:39.000000000","tz":0},"subject":"add ovn ssl config","message":"add ovn ssl config\n\nCreate ssl-certs for ovn deployment\nssl encryption is now enabled between neutron and ovn componants.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1a340f5b67964ffbf7dd8dc60a2d2f554e898c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/1a340f5b67964ffbf7dd8dc60a2d2f554e898c45"}]},"branch":"refs/heads/master"},"556c5c67338adac519312dc401cfd0febfde32fd":{"kind":"REWORK","_number":23,"created":"2022-11-30 16:04:17.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/03/862403/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/862403/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/862403/23"}}},"commit":{"parents":[{"commit":"9bbb133267525d6e3e473b4310673fc2f208929b","subject":"Merge \"Enable experimental execution of LXB if required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bbb133267525d6e3e473b4310673fc2f208929b"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2022-10-21 19:49:39.000000000","tz":0},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-11-30 16:03:14.000000000","tz":0},"subject":"add ovn ssl config","message":"add ovn ssl config\n\nCreate ssl-certs for ovn deployment\nssl encryption is now enabled between neutron and ovn componants.\n\nChange-Id: If8ca3f2035ada97cff248ad49771eefab95c6c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/556c5c67338adac519312dc401cfd0febfde32fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/556c5c67338adac519312dc401cfd0febfde32fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
