)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~866176","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ie725c8e026c84e424c6664d18ace3b2d26b29612","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"ha_vrrp_health_check_interval","attention_set":{},"removed_from_attention_set":{"11227":{"account":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"last_update":"2022-12-07 13:34:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie725c8e026c84e424c6664d18ace3b2d26b29612","subject":"feat: add neutron_ha_vrrp_health_check_interval variable","status":"ABANDONED","created":"2022-11-30 14:55:54.000000000","updated":"2022-12-07 13:34:01.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8274fc568852145bf1f1cf9a304f4ca04fa54d60","_number":866176,"virtual_id_number":866176,"owner":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-02 16:05:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-2,"date":"2022-12-02 15:34:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 17:36:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-02 15:33:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-12-02 15:34:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"a883e34fd1a32e1df6ab4ec84bbd2eca4ae22204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"date":"2022-11-30 14:55:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44a23c0d655375c5f80233a10d1a655714be7125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 17:36:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a812113435f04dd5822a43d2cde9a183\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef8e10b678dc4839aabb87539258ffba : SUCCESS in 13m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ade961b0e9774f47a13867668f45880b : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/eac1f2a3f75f447aa108e0151bbcb828 : FAILURE in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3deaa47a2eda43b2bd409c97b7266348 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/412f843eef55422d856b945e355b5c21 : FAILURE in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcf4054162f649f69f1f1f8758d512b4 : SUCCESS in 1h 21m 07s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5853f95d54634378afd0b17a39743c95 : SUCCESS in 1h 19m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5b917068b134901bfb93b8722707793 : SUCCESS in 1h 40m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb07b9b879134241b70d94ba7eefc68f : FAILURE in 43m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/916a3a9ed3194742bd07d1d3f3ca87c1 : FAILURE in 36m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b74eb7727748aaa2850e162cc2f87a : FAILURE in 36m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c038e1784e4710a937580ed77bf953 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b7651c44de4814bcbdc2fce97c59c9 : SUCCESS in 3m 16s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54af671e9c2f42e5bf15f696f6b5b8fa : FAILURE in 1h 24m 19s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6373eea50d0047f697bc841631811473 : SUCCESS in 2h 38m 40s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4302f882d6d841e98b137986692b65ae : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d409706e31c344f49913edc4259833da : SUCCESS in 1h 59m 54s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf0acd11dc6641aa95dfa0249e842ef4 : SUCCESS in 1h 37m 09s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/520d8034c3364914941df8b60178aa32 : SUCCESS in 1h 58m 24s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e960b64e6c934454baac1b6f54098bab : SUCCESS in 1h 42m 27s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3da1761bb574d91a5ef92f7dfc967aa : SUCCESS in 2h 25m 21s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54a2dd8b68c949388fb8e750afd78a1a : SUCCESS in 1h 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e76c0c3144ec85adb6d2029be1c9241cbc5861f7","author":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"date":"2022-12-01 10:24:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7170b0e3556ed8301d88b2fe70a0217c348c3bf7","author":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"date":"2022-12-01 12:57:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfd1fe1450da0134263300486e15548d6e6d5203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 15:26:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fc17ac015894fa8ab5a6778ddaa80d4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c65636c82834180954ad1f4b3a5ebe3 : SUCCESS in 17m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8906e6ee28c943ddad2b9640a9f04ac0 : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc8a0f2fbfcc4d5b920e1b7d7507639b : FAILURE in 1h 30m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e678bbc137aa44c6ad530e037ea6907d : FAILURE in 1h 12m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e919af55e7e44fc387d883cb4869c074 : SUCCESS in 1h 38m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c465ca7a67e24e938d0321eb416aa2e5 : SUCCESS in 1h 29m 22s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c984c24180954c66b4d991b870921128 : FAILURE in 1h 52m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d48e549897647729720dd94c06734de : SUCCESS in 1h 58m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6668479ffb8a46199f58ec4df48f8c9d : FAILURE in 48m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/948eb474838843a891b285c161fe23da : FAILURE in 37m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/576ded9c62cc4e33911158acbbbb08bf : FAILURE in 40m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48ab6a23d104c9bb956caca7a2f5fdd : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d0b50743ec04135906f3e6d9d6886f8 : SUCCESS in 3m 34s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac384c9d1f9a48af9066dfc901d83d36 : FAILURE in 1h 33m 00s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1c4f24c4c9d4a3d9d714dde18e06032 : SUCCESS in 2h 20m 08s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f29a5a323284330bc794eba93f603a6 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b98e60b5d1434dc386e78dfedf4c1c16 : SUCCESS in 1h 55m 40s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1270a051554ebb93af2b5fcd24bad3 : SUCCESS in 1h 31m 45s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/30b49660b5154674a284407949bd9b9a : SUCCESS in 2h 02m 12s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bd6f05ceade4412a8f73f7d72a00776 : SUCCESS in 1h 31m 08s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa53afc693a34291966b293ae3813aa5 : SUCCESS in 2h 12m 11s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a00a6675c244a8ad2d815dbb27391a : SUCCESS in 1h 35m 35s","accounts_in_message":[],"_revision_number":1},{"id":"af55064925b308ed96033282bb809067f3c07fa5","author":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"date":"2022-12-02 13:36:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6380b2cf57610dfd24b2e831b63c2dce2a19044","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-02 15:33:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a53b70fdd6c97ba8fa728e5390b9a7319e7c6be9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-02 15:34:35.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"09db88c51406774692d84d4cce0730efb4a8d4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 16:05:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8980971369414812a6579de66eb0231e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e239f282734c4ea58c5d643553bf985e : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ecb5483e1943cea7d24b2e04aa5dcb : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce3312946e7f4f5ea877efe57b36e3f6 : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27d411659fde47c6b525209e80616d02 : FAILURE in 1h 33m 41s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bb5dac91cdd4573a2411ef92f304973 : SUCCESS in 2h 18m 52s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa51990d192d4cdeb35e0d89408f8061 : SUCCESS in 1h 40m 16s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7886bcb38964143ac57d44f27c20369 : SUCCESS in 2h 02m 30s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/f90543baa3bf4bbeb085d89b8ac179f6 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f4fd02042e84dbab1b442b26ee8b22f : SUCCESS in 2h 11m 05s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6291f2a9974d454ea6ff1b0a0f07730c : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1156b6e75e2f46d4bbd04a5467308e30 : SUCCESS in 2h 25m 22s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/248b64e936d8404bab3c5498fc92d4b6 : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8274fc568852145bf1f1cf9a304f4ca04fa54d60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"date":"2022-12-07 13:34:01.000000000","message":"Abandoned\n\nThe more generic neutron_l3_agent_ini_overrides mechanism exists.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a3cdf34be3346be10c9ab7b98ed3b5bbcbde1f8","revisions":{"9a3cdf34be3346be10c9ab7b98ed3b5bbcbde1f8":{"kind":"REWORK","_number":1,"created":"2022-11-30 14:55:54.000000000","uploader":{"_account_id":11227,"name":"George Shuklin","email":"george.shuklin@gmail.com","username":"amarao"},"ref":"refs/changes/76/866176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/76/866176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/76/866176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/76/866176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/76/866176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/76/866176/1"}}},"commit":{"parents":[{"commit":"192b552cf160b91249851f72c8e22e5025ef49b6","subject":"Merge \"Allow to set dnsmasq configuration options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/192b552cf160b91249851f72c8e22e5025ef49b6"}]}],"author":{"name":"George Shuklin","email":"george.shuklin@gmail.com","date":"2022-11-30 14:50:52.000000000","tz":120},"committer":{"name":"George Shuklin","email":"george.shuklin@gmail.com","date":"2022-11-30 14:55:34.000000000","tz":120},"subject":"feat: add neutron_ha_vrrp_health_check_interval variable","message":"feat: add neutron_ha_vrrp_health_check_interval variable\n\nDefault value is equal to default value in neutron/conf/agent/l3/ha.py\n\nChange-Id: Ie725c8e026c84e424c6664d18ace3b2d26b29612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9a3cdf34be3346be10c9ab7b98ed3b5bbcbde1f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9a3cdf34be3346be10c9ab7b98ed3b5bbcbde1f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
