)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~866249","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I7a08ed81a15c0678832bcdd192bdc4e10064bd6d","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/ovn","attention_set":{},"removed_from_attention_set":{"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2022-12-02 21:23:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a08ed81a15c0678832bcdd192bdc4e10064bd6d","subject":"Install openvswitch packages in Neutron server container","status":"MERGED","created":"2022-12-01 04:24:31.000000000","updated":"2022-12-02 21:24:25.000000000","submitted":"2022-12-02 21:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"866249-osa/ovn","meta_rev_id":"ff07e72aae1c3c2783d77db6c837537cc3236bbc","_number":866249,"virtual_id_number":866249,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-02 21:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-12-02 16:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2022-12-02 16:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2022-12-02 16:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 06:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 18:18:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-12-02 10:31:39.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"78f33bb14e0d29d05fb841d2b1df952133d3d0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-01 04:24:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35eac81397dec3bde75a5221b75488f9139787be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 06:36:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b894dc89fa564b65a958c13155d6be70\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54d3247ecba84741a2daa41425cde7ad : SUCCESS in 17m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/40812c2b7be04d8dab80d47ddf9b51a8 : SUCCESS in 1h 26m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/82a3a5b1f7b54519b24e1943bca37fb1 : FAILURE in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba259fe5dc2d4da3802e4b7892a844c7 : SUCCESS in 1h 29m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41206bcc05264d6c9abc8f03ab9cdaab : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07114cf7247a4f3aaedc7d80e6c51d02 : SUCCESS in 1h 35m 06s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3a23097bcb54e578b6fd217e18797ba : SUCCESS in 1h 47m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eb0ebab2fee4356ad694a15695bccfe : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ad60c5d4bb42cca379cf6cb6c33a9f : FAILURE in 47m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79bce29696744397a3f772966b79a235 : FAILURE in 37m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3df6f1d7264d40c6a1e2463201fe6aa8 : FAILURE in 38m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f6ef3daccf4e5485dfba42f89bce0e : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1bf74a6077d422b8f38ca700a1c53bf : FAILURE in 1h 29m 01s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ac62c584ec447bcae0785d8ef03990b : SUCCESS in 2h 06m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21cb7f58015f4d129aacfd98f3d27eb2 : SUCCESS in 1h 30m 31s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b69b656046904e6f9ed6aba461fae2ec : SUCCESS in 1h 54m 51s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/f158732f2ac04c6ca23589d4043bf281 : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/560cca517707461c83d1dba1e252404d : FAILURE in 1h 36m 01s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/15057188eacd4c1bb62cc79c47686e93 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a320aa35e7314c2abd54c151af1e500c : FAILURE in 1h 51m 51s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67a9561ddd644790b4b011bf99e1d122 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2be1cad58ddd0439ffde885980232c96bd9dd4be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-01 11:30:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"56ebdfc09908a670ad4b05e8bbad3c55fb21a9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-01 13:36:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b4c81aeb77948784ab5a10055f571ffae79c5bd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-01 15:43:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1ea2925ca76079836624e60ac6747a2146f96d5d","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-01 15:44:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd1f1064586006f3c90b88b023faf0ba9e5c6c92","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-01 16:12:21.000000000","message":"Topic set to osa/ovn","accounts_in_message":[],"_revision_number":4},{"id":"a2cdc5eeaef6b8b20e079504bca711cfdb680009","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-01 18:18:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3f88315cf4ab5b4a0c4702157d4fcbec6d6b976d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 18:27:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfecd198a87a40788f5a2cdcfc9d5bcd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/300fe901411e4c1abc5a9740145d3fb5 : SUCCESS in 11m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4d4a6f0b3f374227a64dbf3fd739ccb6 : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad0ab003e0cc4dc4ad3b2953fa15d87e : SUCCESS in 1h 34m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39449e71ec7b47e6a047268b77ee1504 : SUCCESS in 1h 29m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dc4da8096cc44dc9ae7bfd36db6a6cc : SUCCESS in 1h 41m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f24db891a8da46899d42b4d026245251 : SUCCESS in 1h 35m 34s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ee14f4f92bf41428d24fdb624226ffe : SUCCESS in 2h 03m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/377ea64d8eec408b92e7cde1ee71e6a7 : SUCCESS in 2h 09m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db10e09e3a0485eae5bb90360980512 : FAILURE in 51m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e733d040cf60451ca43da8ae18b49c3d : FAILURE in 46m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595442b8eb4647399b8a44a67cb0f8cf : FAILURE in 43m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabd508504274e30bf2f7a98f624c81d : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1a9fe5a4b2740f687d88c4a17c2c82d : FAILURE in 1h 33m 46s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b78dd5e6797e42e1949aaaef1bad3ec9 : SUCCESS in 2h 31m 15s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9dfd58576b54a9db03bd8363295f248 : SUCCESS in 1h 48m 24s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27def717060e4f928c008b9d3d67063d : SUCCESS in 2h 11m 51s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b10dbbc9add4b54a0263333b0d3df4c : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/593880e2ca89490b9f234efae8f5cd79 : SUCCESS in 2h 15m 54s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/191edaee8dd84003a010bd36ac302e19 : SUCCESS in 1h 36m 17s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1b588e1c83f4594b0e18af2c6885c44 : SUCCESS in 2h 36m 44s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fbc736e02744de9abd078ff90285364 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":4},{"id":"7e5a2f6439a74252d51a334427aeffbb164457f4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-12-02 10:31:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cb490d8aed2a5cbd1278b0290b640f0a64865efe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 10:33:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c6bd727234af1b4d626acda0ff048240d1353376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 12:53:03.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86cfecdeb20d47fc8f881ce9c542fe45\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a38e32fe64461f93112c550c096cc9 : SUCCESS in 14m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09922d6af6b743ee8ad313683a230195 : SUCCESS in 5m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7d75ce6b72044ce82320581f40791c6 : SUCCESS in 2h 18m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e531bbb66bf74aebb62396dae8a9551f : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f11c27549914cdfbf168f7e8441bb15 : SUCCESS in 2h 06m 10s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3ced822c6124811a5a35dd79fb5b11e : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/add552e1f09f47778d3625e7610bf9d1 : SUCCESS in 2h 08m 35s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9fe1ac8a2694526ba7072c5a08e57c3 : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/618eb648fc2c4a31aee05e56ebb99bf0 : FAILURE in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8cbbff17cfc41119d7bc9ce8ef43f25 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":4},{"id":"6a3c230f42aa89b8ef083e75abf45db7e618b58f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-02 16:43:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"39f1a0ea434214e535349f715856a4759d06ce8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-02 16:43:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b0f2ebf5a492b820bad584cf2a3e1e530bf546f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-02 16:43:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8a809bd55eeebd7d1e7271e48bc9caa1aaf054fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 19:10:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5848be81f4f44790ba2287d33f9266ce\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e045a762756420b8536d9948900a862 : SUCCESS in 20m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6986da8694044195aa6a867840740b63 : SUCCESS in 6m 47s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96a22e43e93a420a92d42379cc582cd9 : FAILURE in 1h 30m 58s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38e04e4e37cb41f5b011e6f0dbe7fdb7 : SUCCESS in 2h 14m 30s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad48b3e284fd4d98b5e8ed84e9828062 : SUCCESS in 1h 34m 53s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0fe072f2d924e7db07a7c91c57fd8cb : SUCCESS in 2h 01m 28s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/79b894b12d4c42fc894fac4fcc5df013 : SUCCESS in 1h 38m 38s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e3cabc5c61480ab40a2f9aef78a86c : SUCCESS in 1h 58m 52s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f0eae02a0f1408e8e6f2621f69c4ce2 : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/986528fe0b5c4222bde0b9c64e2b253a : SUCCESS in 2h 19m 47s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbb6f8e902d943a590a2699999aaff29 : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":5},{"id":"160d3a796bfe643c97a905949f833cf5488f97ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 19:12:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8ec355c0e4bebd439af98c10f6b7e5feca395369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 21:23:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21428348bac7434eab12c314e8105e78\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74ac48f866144cd5b6435b4d65b79fdc : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/370f4d29dce145529bc1174f532cf80d : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f909a6b1ecf64f418174859913ecc8ed : SUCCESS in 2h 10m 38s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/892603327f084f6b8dd633ea8969df3f : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b8b9405463848c689bcc27b0163d772 : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/44c34bb30731447ba2bb257804722805 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/25729fbeb97a44b999aa5a42dac890b5 : SUCCESS in 2h 01m 27s\n- openstack-ansible-deploy-aio_ovn_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/154ea50fc5cd4fb1a8a0f075c59238f8 : SUCCESS in 1h 15m 06s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0a36966ca0246169c9b070129ccdae1 : SUCCESS in 2h 03m 11s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65fde84395fb4628bcd60c3ecd4f3c7a : SUCCESS in 1h 26m 03s","accounts_in_message":[],"_revision_number":5},{"id":"dada9acbedb7706b677a3585701b770b7a7fe577","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 21:23:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ff07e72aae1c3c2783d77db6c837537cc3236bbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 21:24:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d53309d61c64ce09c69954ec47c2151\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b07623cb66d4a9681eb82d00ded553b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c79d3c2e4c8abe865a4923ebe164e20d185c8f34","revisions":{"94c1806e731d04887b093a9e3967c29ff7afa58b":{"kind":"REWORK","_number":1,"created":"2022-12-01 04:24:31.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/49/866249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/866249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/1"}}},"commit":{"parents":[{"commit":"192b552cf160b91249851f72c8e22e5025ef49b6","subject":"Merge \"Allow to set dnsmasq configuration options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/192b552cf160b91249851f72c8e22e5025ef49b6"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"subject":"Install openvswitch packages in Neutron server container","message":"Install openvswitch packages in Neutron server container\n\nneutron-server appears to require ovsdb-client to communicate with ovsdb\nin an OVN install. On metal, this isn\u0027t an issue since ovsdb-client is\ninstalled with openvswitch-common but with LXC, OVS packages aren\u0027t installed\nin the neutron server lxc container.\n\nUbuntu/Debian splits out openvswitch-common utilities from the actual\nopenvswitch service (openvswitch-server), while CentOS/RHEL-based distros\npackage them all up. This method was chosen as lowest common denominator\nbetween supported releases.\n\nChange-Id: I7a08ed81a15c0678832bcdd192bdc4e10064bd6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/94c1806e731d04887b093a9e3967c29ff7afa58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/94c1806e731d04887b093a9e3967c29ff7afa58b"}]},"branch":"refs/heads/master"},"86a9711fd4ce2e4287c8b5aea0e457b24bb067f8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-01 11:30:26.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/49/866249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/866249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/2"}}},"commit":{"parents":[{"commit":"b0db979c900ce85f5dc3a30925c711445bc4dbeb","subject":"Merge \"add ovn ssl config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b0db979c900ce85f5dc3a30925c711445bc4dbeb"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 11:30:26.000000000","tz":0},"subject":"Install openvswitch packages in Neutron server container","message":"Install openvswitch packages in Neutron server container\n\nneutron-server appears to require ovsdb-client to communicate with ovsdb\nin an OVN install. On metal, this isn\u0027t an issue since ovsdb-client is\ninstalled with openvswitch-common but with LXC, OVS packages aren\u0027t installed\nin the neutron server lxc container.\n\nUbuntu/Debian splits out openvswitch-common utilities from the actual\nopenvswitch service (openvswitch-server), while CentOS/RHEL-based distros\npackage them all up. This method was chosen as lowest common denominator\nbetween supported releases.\n\nChange-Id: I7a08ed81a15c0678832bcdd192bdc4e10064bd6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/86a9711fd4ce2e4287c8b5aea0e457b24bb067f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/86a9711fd4ce2e4287c8b5aea0e457b24bb067f8"}]},"branch":"refs/heads/master","description":"Rebase"},"9150ff2de16fd13e27e9ddf82bc230953ea78b51":{"kind":"REWORK","_number":3,"created":"2022-12-01 13:36:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/866249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/866249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/3"}}},"commit":{"parents":[{"commit":"b0db979c900ce85f5dc3a30925c711445bc4dbeb","subject":"Merge \"add ovn ssl config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b0db979c900ce85f5dc3a30925c711445bc4dbeb"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-01 13:25:20.000000000","tz":0},"subject":"Install openvswitch packages in Neutron server container","message":"Install openvswitch packages in Neutron server container\n\nneutron-server appears to require ovsdb-client to communicate with ovsdb\nin an OVN install. On metal, this isn\u0027t an issue since ovsdb-client is\ninstalled with openvswitch-common but with LXC, OVS packages aren\u0027t installed\nin the neutron server lxc container.\n\nUbuntu/Debian splits out openvswitch-common utilities from the actual\nopenvswitch service (openvswitch-server), while CentOS/RHEL-based distros\npackage them all up. This method was chosen as lowest common denominator\nbetween supported releases.\n\nChange-Id: I7a08ed81a15c0678832bcdd192bdc4e10064bd6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9150ff2de16fd13e27e9ddf82bc230953ea78b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9150ff2de16fd13e27e9ddf82bc230953ea78b51"}]},"branch":"refs/heads/master"},"6d225369a7f44a00588010161363bf2448c423c9":{"kind":"REWORK","_number":4,"created":"2022-12-01 15:43:23.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/49/866249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/866249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/4"}}},"commit":{"parents":[{"commit":"b0db979c900ce85f5dc3a30925c711445bc4dbeb","subject":"Merge \"add ovn ssl config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b0db979c900ce85f5dc3a30925c711445bc4dbeb"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 15:43:23.000000000","tz":0},"subject":"Install openvswitch packages in Neutron server container","message":"Install openvswitch packages in Neutron server container\n\nneutron-server appears to require ovsdb-client to communicate with ovsdb\nin an OVN install. On metal, this isn\u0027t an issue since ovsdb-client is\ninstalled with openvswitch-common but with LXC, OVS packages aren\u0027t installed\nin the neutron server lxc container.\n\nUbuntu/Debian splits out openvswitch-common utilities from the actual\nopenvswitch service (openvswitch-server), while CentOS/RHEL-based distros\npackage them all up. This method was chosen as lowest common denominator\nbetween supported releases.\n\nChange-Id: I7a08ed81a15c0678832bcdd192bdc4e10064bd6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6d225369a7f44a00588010161363bf2448c423c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6d225369a7f44a00588010161363bf2448c423c9"}]},"branch":"refs/heads/master"},"c79d3c2e4c8abe865a4923ebe164e20d185c8f34":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-12-02 16:43:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/866249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/866249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/866249/5"}}},"commit":{"parents":[{"commit":"05948f688dd96676b728ecd4864070b66ce9fd4e","subject":"Merge \"Set default plugin type to OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/05948f688dd96676b728ecd4864070b66ce9fd4e"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-01 04:21:27.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-02 16:43:47.000000000","tz":0},"subject":"Install openvswitch packages in Neutron server container","message":"Install openvswitch packages in Neutron server container\n\nneutron-server appears to require ovsdb-client to communicate with ovsdb\nin an OVN install. On metal, this isn\u0027t an issue since ovsdb-client is\ninstalled with openvswitch-common but with LXC, OVS packages aren\u0027t installed\nin the neutron server lxc container.\n\nUbuntu/Debian splits out openvswitch-common utilities from the actual\nopenvswitch service (openvswitch-server), while CentOS/RHEL-based distros\npackage them all up. This method was chosen as lowest common denominator\nbetween supported releases.\n\nChange-Id: I7a08ed81a15c0678832bcdd192bdc4e10064bd6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c79d3c2e4c8abe865a4923ebe164e20d185c8f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/c79d3c2e4c8abe865a4923ebe164e20d185c8f34"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
