)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~868415","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ib6d69bb2ce340b50140216e2abf236a1da93e46b","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2022-12-28 17:36:34.000000000","reason":"\u003cGERRIT_ACCOUNT_16011\u003e replied on the change","reason_account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-28 10:26:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ib6d69bb2ce340b50140216e2abf236a1da93e46b","subject":"Create separate lock path for neutron-ovn-metadata-agent","status":"MERGED","created":"2022-12-22 12:53:51.000000000","updated":"2022-12-28 22:05:58.000000000","submitted":"2022-12-28 22:04:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"868415","meta_rev_id":"0ab8a379e5df618030e5e7e3c144f7cac8fd7369","_number":868415,"virtual_id_number":868415,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-28 22:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-12-22 14:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-28 13:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-28 13:38:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-28 13:38:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 14:10:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-12-22 14:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-28 13:38:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a713081dc5538314ad52ed981c7a73ad4b393989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-22 12:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3087a8f28945d62249cecb84a29fae53245c95a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-22 14:10:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc8f4590556c832c0eb045133b4d08ff1f57dea2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-22 14:12:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8042fff07e77a96bd4c53f52e588d314726decc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 14:59:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/832988551672454d8f2e8b8b2099e174\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de0264f12168434aadb713af1418a76e : SUCCESS in 10m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/969eca629c7c4ad294fbf7667870ec8f : SUCCESS in 1h 03m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/964263536bb6435ea7a264de596d84b1 : RETRY_LIMIT in 8m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68b342cbd5a04ff1a557d0638c5b82d4 : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebef4df3a2a54faea5e42e67e96d438a : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b35d765da1b14f41907f88eac8a526f1 : SUCCESS in 1h 09m 08s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9f20e9f5d774ae9ab73ae3112af5eaf : FAILURE in 9m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdb6a8bc7a5d45979fa3dc0452a0e068 : SUCCESS in 1h 52m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c39de5797364ababf7e9e6b3ce39071 : RETRY_LIMIT in 4m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a80b9b0a5a646b9a73fc37a596d0efb : SUCCESS in 53m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a270b9272e43e0a020abea485bdee1 : SUCCESS in 4m 08s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1c2591017d24daeb4833738097361f3 : SUCCESS in 1h 57m 28s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ba3704499fd4ab59dc235079002deb7 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce454af6ff0a49bd9949abdd4dbda4cc : RETRY_LIMIT in 9m 28s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/40eca1667c2d4d12a9040274883328e9 : RETRY_LIMIT in 8m 05s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8e29f25ae4a49f79026a86a56260d7c : RETRY_LIMIT in 9m 16s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/65d892df82364a8a90ebf652cabf4664 : RETRY_LIMIT in 8m 25s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc83180e657e497cbf2feeedc3cb3935 : SUCCESS in 1h 52m 14s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd637ec8f7a24eb5bb14b80bbbb6774d : SUCCESS in 1h 19m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0e3758dd8117712f34866625bd634b37b17bf894","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-22 15:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48ee5ebb6984fa119b761cb6b5a0a14a61464c1a","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-22 16:10:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3356f9d4ff25a4dfa81c4d31969d5208f193fbd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 18:15:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc31e59d107e4993beda0bc7a678d897\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b311c3601b4ff08ba8b4b203bb7a9d : SUCCESS in 14m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7dac7320c4bc471cb5516b28309288af : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b89b301fb4b4d33a5664a8f6c0aa08c : RETRY_LIMIT in 8m 51s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/293c25cf5a744df9b6537bcb3639696d : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf49987c2ea4c108ae9e54ad9971ca5 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c6f0dbba6534c7ca82744799a9fa58a : SUCCESS in 1h 13m 33s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/88838933699743319c5b67bb9ce37857 : FAILURE in 9m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f07e45e15b2464fa2e00e8a31b51a5f : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea5adee3f79946569ad1a6735ee75286 : RETRY_LIMIT in 8m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a1f7b4aba0419a8a77459b9009622f : SUCCESS in 51m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fcc3d67ec4419b895658f182f5b074 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/805d2e9471a649d0830f2ab84f40eb4d : SUCCESS in 1h 58m 44s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a50f133c3df48a6be3280366ec2e8ae : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c30893cb69cb4f3e816e1913c0d949d0 : RETRY_LIMIT in 8m 46s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad8445ade84c442996cb503ef4a1f600 : RETRY_LIMIT in 13m 15s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e88b11316e834c2d9c17e87347e95d64 : RETRY_LIMIT in 8m 34s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/668a75cd14ff4f6c822e1439b85182ab : RETRY_LIMIT in 15m 04s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e073446dd549f3aa77338dfcb28b23 : SUCCESS in 1h 54m 22s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a993aebe5b2c494eb3ed54f7c062fdd0 : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"43194f2c3a48f53eedbbaa1cd73b174443af2f88","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-22 23:52:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdb5292e2cabe02dfdc9ac5e11a232169eb48878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 02:01:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb83449010ed410cadd490ee20810da9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7085b1d746a4d1996bb25ebb5cc82e8 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cafb94afad694020b0198024de886e0f : SUCCESS in 51m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa0c9860bcd7434fa7685fd7c1d743bb : RETRY_LIMIT in 8m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c52e75951ce3406fa154f32bb65b92ad : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2ecb82694b499bb26241e70e90093b : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9bbb78d30764c0aa489ca366d8fd90b : SUCCESS in 1h 06m 41s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d27c02894cb0410c9c81aebfe3a78002 : FAILURE in 8m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db70a499a8014ca5bf287f58732e56a7 : SUCCESS in 2h 00m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fea6f04a8f6542e79941bf5ba7ad5863 : RETRY_LIMIT in 6m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fbcac1af971417b813b621ee57eab87 : SUCCESS in 59m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e944cbf1fd4d3aa55014fbe55445f5 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3abc230215524c9a8ffb7aec6a24bceb : SUCCESS in 1h 40m 48s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd5a9d0b95d04d3d9b268588ceaa6e06 : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe6b51a788f541259d10c1c9c84218b0 : RETRY_LIMIT in 9m 26s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/0105d0c83e5041d88c5628d64159e985 : RETRY_LIMIT in 8m 25s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b7e861d830849d4933af30a824430e7 : RETRY_LIMIT in 7m 59s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c24e171e1dc949038236a4d847d956f2 : RETRY_LIMIT in 9m 22s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/822ced9dd1974d2f949705c89e7c0fab : FAILURE in 1h 31m 19s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a8ca466dbc455d96418ae72558f77d : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5929565bb2d8cbccde13230e1326604d884dc60b","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-24 05:02:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c51e992688ef8a0410cccbf691132e7c59a7ac10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-24 07:23:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85160d1a342b46e88cfa3dd92c68ea90\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e13fc18752645279b7fd62dbc37b1eb : SUCCESS in 9m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ecde13ffae9f40a0b3e0568bc93ad633 : SUCCESS in 1h 02m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/12a6a6395e2d466b8d1f01efd091731f : RETRY_LIMIT in 8m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a4aa39de97841adbd8060315dd083b2 : SUCCESS in 1h 06m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cd96ea9f4e9449f8db1a27f3fdce3af : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ff4af56de0e42c5a5aaf638b9542d9c : SUCCESS in 1h 00m 29s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/95a1bbeacce8427ba745f11e7df4370e : FAILURE in 9m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ff91a9d3a04d0fab7b041531f0f2de : SUCCESS in 1h 34m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2d38f2a404f4f9189e71e3e4f9c7fd8 : RETRY_LIMIT in 5m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34abf8148a274b6a9d198f2aa365adc7 : SUCCESS in 54m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3b03de88b34090a6aadb9fad2e8fdc : SUCCESS in 4m 52s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ee3bc8005c44dc88c3fe7a0c4fe376b : SUCCESS in 1h 43m 21s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b874d67a84e41fd94b924752d595a4e : SUCCESS in 1h 06m 40s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/70b9d7af0b164828aebd4b270d788e07 : RETRY_LIMIT in 9m 12s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb24cd7da7d043ddb11ce3ee4bf80c6c : RETRY_LIMIT in 8m 50s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2faa15900dc04fbb8a76c59fdbc059d4 : RETRY_LIMIT in 8m 41s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4821776b297c4363b08a69dc9f88e0fd : RETRY_LIMIT in 7m 57s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4120d0fa9c844d59e6288f73e754bbf : SUCCESS in 1h 57m 31s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c27b54d2ef4428e8aa2061725f84ef7 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5f08d55bb7db6d2883b9f6b8abc63f17ed336d59","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-26 14:29:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62f4b4b9c9f95be0506052c1d655b8dfb593863b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 16:41:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d2b2f98d0db4dd4993caeda5f1979bd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b276660185bd489881eba4288e138d2d : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b8694661d00f445ba424f11dad46c238 : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/900c336085ac48f9bd52b3960f4b07be : RETRY_LIMIT in 8m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a22e53c228840bb8c8850e73b70cc70 : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62dccda5236242229bf8dde652fa43ba : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e83f2b9d6bf84a68b0ca3eaa7ebd9373 : SUCCESS in 1h 18m 03s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1b9a7300fa44864974279430e8f8248 : FAILURE in 9m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ae920e83ce744178ec22f21a81f5641 : SUCCESS in 2h 07m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/150bb6de92d84dcb84e1313090ff19bd : RETRY_LIMIT in 5m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/122be042b636431da96adfd717567fa2 : SUCCESS in 58m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63583d912e8645af85570f7662993a63 : SUCCESS in 4m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/073e330e7c6a4f29bd69e189f4943d88 : SUCCESS in 2h 06m 59s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c593d32fb5194eeab84fe147175e71af : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b85d3dba7094fca86a682055c272ee3 : RETRY_LIMIT in 8m 34s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a0a519a440c4fc08c6f41d963659e99 : RETRY_LIMIT in 8m 04s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/944ee0ad3a8a4985a5411cb9adcead63 : RETRY_LIMIT in 8m 12s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d6cc17dc2984d8998c6f69e7d940b8c : RETRY_LIMIT in 8m 39s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f97930aa20e4704a500d9caf6d83541 : SUCCESS in 2h 09m 57s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e05a62e15d7241f2a8a76cbb208a0a24 : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6348214d84fb6de7e4abb96b50f84e56cf008fa8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-28 10:26:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa0b8a5ce4603579d36049767e66b57b94fe6684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 12:34:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fc2c2fa2a7b4f3286d97d3d54910bad\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/387af87f7f4f4de88f3e4238c096a606 : SUCCESS in 12m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c05f31972d241b3be7396e241fb62bc : SUCCESS in 1h 07m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7d5db85588d4b309be4760ebbf2f22f : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f1f1469ccf942338f696f304662d39d : SUCCESS in 1h 27m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/657e3ad6e67a4ff09d99c10a10fac3d0 : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fe65fb8d06f4dc2aacc8b36db3e02e4 : SUCCESS in 1h 20m 59s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/27bd3039b9f64297a002502bc2e53e48 : SUCCESS in 1h 50m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0cd96d87d274b968556100ff71534d2 : SUCCESS in 1h 54m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/010d53ca634c464896602386d724af5d : FAILURE in 38m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7159c6dc8592421d8cd15494caa8fae1 : SUCCESS in 1h 05m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d2abb3ce9346b690eb54abc0c40487 : SUCCESS in 5m 28s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fa7712c02d34bbaa40999b0280f77a4 : SUCCESS in 2h 03m 42s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29e91fe02b5447f0a1a248b58143ac78 : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/261e826f91db44fd86029d4a07292bea : SUCCESS in 1h 51m 14s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/947a3589638446849dedd78e2f57c965 : SUCCESS in 1h 18m 53s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b306d64efaeb48fab761a34f097c5cca : SUCCESS in 1h 48m 17s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e5fe1283d264b13bac66fd365ea54f8 : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d506f516769d45c9b0b9c2a96db7c5ed : SUCCESS in 2h 05m 50s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2822b2d20ef04e0c971b492fa7888cb6 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b18a26eeb252cb526a90953fe61a5c205bb0d3ad","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-28 13:38:30.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5417cfd8104868a958617f6f22243c7d3f749be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 13:40:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b77408d14e24dabf91cb43c1aaf8f98daffaba72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 16:44:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74587586071143a98a4727e2ddb5a250\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c019cebad2d4302b4c923ba1086b5f4 : SUCCESS in 10m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/209be64bb63a4e5a8ec6d29fd207606c : SUCCESS in 1h 03m 14s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a84c3a3169c04d209e09872fce8be60d : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b066dbe00bd04247b8c6aa2c29e9cd35 : SUCCESS in 1h 15m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f208ea35ffe643fabc9ff428c78da83d : SUCCESS in 1h 04m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa35708d10a46268b84994b65b3a9bc : SUCCESS in 4m 25s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe7afe56fe42421baaa71dbe4e96fbcf : FAILURE in 1h 05m 46s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eb9426eeb324270ae9a684aebed29e7 : SUCCESS in 1h 10m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8ea4201df42458b871451f489708b58 : SUCCESS in 1h 46m 49s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f0eb2ef13a84b5ba0f0ffad967e0803 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/064f9c7e7a164885a36e7515784c31db : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee7caa41046d4fdeb2a44029f5b20a35 : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dd3ad478e7f45bab83fb21418603ea0 : FAILURE in 2h 29m 33s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c220eadcf14bcb924bd1814aa4d143 : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":1},{"id":"57944d259aaedfbcbdea8dc1e2416dcd756cb912","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-12-28 17:36:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42217febed679084287c30057e66ea77a1a5b759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 19:35:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa3b23582fa4de1b7f11132485e888c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ecbb4e41b734ecc9f133c4a537b4fac : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f527a97800da4ea38aad15ff4870b5de : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ce8eb735d6647cda9fe70cc785d62cf : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0484800bc20b42babc3cb469fa47903b : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c563873eef3042ba94ca64c3564be861 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/449b19c8800c496d8ae29090449b1f27 : SUCCESS in 1h 07m 20s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6c9fa0b321f49708ccf954e9b2bd6a1 : SUCCESS in 1h 45m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e31e35624d44ddb85b376668801724f : SUCCESS in 1h 33m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/af031bfcecd94e3d88fdd00d0a9d55ec : FAILURE in 36m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27836f9790294a189d440f3c03257b0a : SUCCESS in 59m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0cd18b9c3d40be86c82fde6fa3585b : SUCCESS in 3m 56s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf7092561a6442d1b0d6eccd1586ffad : SUCCESS in 1h 55m 56s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943a98b4ce4c4defb2a96500d5805dde : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b42663ca2f4244fbbfacfbc3100ac3c8 : SUCCESS in 1h 37m 35s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff79a607b4c24d4bb45c93ca26ccf1ba : SUCCESS in 57m 58s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1fbca692c2444cdbd1050a2b342d4e0 : SUCCESS in 1h 31m 59s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa636b18293346a581d931f912cef764 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fce275d2c46641dca37e6c92faa7c105 : SUCCESS in 1h 55m 30s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/829098e25a4f46918d65029d8c314da8 : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2e843e07b1d209b77eca9932070b9044e524b553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 19:37:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ac03cf85ac78529b64ca14bb98b4890bbf91820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 22:04:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ab09eeb4204cff89c39cb076262fe0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81450e96fd854a6882f7e8336ffa6825 : SUCCESS in 9m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d0cee449b765473892b22316071d6d28 : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a288c3dbf7a47df806a8ea868cbb5ca : SUCCESS in 1h 17m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c6c71ede9274f25b3d4cd931896e1a2 : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/061e65489995479c9063dddc845204c9 : SUCCESS in 1h 23m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bba2d1dd64c4fc18ab99e681039402a : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e975da046060482eb96bc9bfb811ca51 : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea5e6843322f4e2187b8fe2b5d41eaa0 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/56146dba2c4f4c0da621ba5c9f03ab36 : SUCCESS in 1h 33m 54s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2dbf93cf9c34bf6b561b429a76c8b67 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/45cba97a09d04250a6c6c3d2219f01c4 : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1783b45da76d4b48946d25c65c93871f : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf55aae694e4dc8902fe719fb8e1b04 : SUCCESS in 2h 21m 18s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c55faac8cf244ffda446c05eaa073a72 : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3c78cdfe30672e85f3d31a5581f239eb7f006b48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 22:04:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ab8a379e5df618030e5e7e3c144f7cac8fd7369","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 22:05:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779b376ced124fceb11e311e494dd371\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303fe98354ea4b5d8440eebabdc7b74f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4cbd2d7adcf4bba95a885c87d2c6e4dc9c1b012","revisions":{"d4cbd2d7adcf4bba95a885c87d2c6e4dc9c1b012":{"kind":"REWORK","_number":1,"created":"2022-12-22 12:53:51.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/15/868415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/15/868415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/868415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/868415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/868415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/15/868415/1"}}},"commit":{"parents":[{"commit":"92fa6a5295d7a7cccb102387fbd92c8c70568a72","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/92fa6a5295d7a7cccb102387fbd92c8c70568a72"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-22 11:55:48.000000000","tz":-360},"committer":{"name":"James Denton","email":"james.denton@outlook.com","date":"2022-12-22 12:53:49.000000000","tz":-360},"subject":"Create separate lock path for neutron-ovn-metadata-agent","message":"Create separate lock path for neutron-ovn-metadata-agent\n\nroot user/group ownership of the neutron-ovn-metadata service caused\nthe neutron lock dir to be owned by root:root, which caused issues\nwith neutron-server\u0027s ability to write the OVN hash ring lock file\nto /run/lock/neutron and prevented the creation of networks.\n\nIt appears neutron-ovn-metadata-agent needs access to the OVS DB\nschema via unix:/var/run/openvswitch/db.sock, which is owned by root,\nso a separate lock path has been created for the metadata agent to\nworkaround this.\n\nFWIW, this issue manifested with upstream Neutron commit\n536498a29a4e7662a4d0b1bb923e2521509ad77a.\n\nChange-Id: Ib6d69bb2ce340b50140216e2abf236a1da93e46b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d4cbd2d7adcf4bba95a885c87d2c6e4dc9c1b012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d4cbd2d7adcf4bba95a885c87d2c6e4dc9c1b012"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[]}
