)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~874949","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I50b99306a52f1a2379e55f390653b274afd5885f","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-19 11:49:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50b99306a52f1a2379e55f390653b274afd5885f","subject":"Use include instead of import for conditional tasks","status":"MERGED","created":"2023-02-23 17:48:35.000000000","updated":"2023-04-19 11:50:56.000000000","submitted":"2023-04-19 11:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874949","meta_rev_id":"fd4c706001385564c558ffb5f0c192c923d030f1","_number":874949,"virtual_id_number":874949,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 11:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-04-19 09:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-14 13:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2023-04-19 09:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 20:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 13:21:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-19 09:42:43.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"2e8615eb36266957a58c6a0475122655c5e6d8bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-23 17:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1be672cb6f9432984684e9c42feb06926899a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:06:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7858e9bdb06492399558990d9030f2a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/368d5e23c09042ba99a1892cbf60d4c6 : SUCCESS in 11m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3f6eb37d0bfc473bb8d9794225d792b0 : FAILURE in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/de78a488ac464576b29d08f9a28c2175 : FAILURE in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc73d19909154e7786be4d0579f0011e : FAILURE in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce4497572537460ea8a942698c1316f0 : FAILURE in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c500ec53368467bb3705a69a9223844 : FAILURE in 1h 22m 01s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f92f65e78754e65b9fb1e3ecf5cbe9d : FAILURE in 2h 08m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba77bc8318cd44ea9742ee74b3369170 : FAILURE in 1h 46m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f68807d8ab24518a644f15b5b9c6999 : FAILURE in 46m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dcbb4119f69457baa3f0577f013dfbf : FAILURE in 46m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e0bffa77164cb684831148be95efac : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebd443acef84a4e9e00f343d9623678 : FAILURE in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05826edeef074c88b41b8ad35827f835 : FAILURE in 1h 21m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/34c18d7b3b06428cbb64150aa4420420 : FAILURE in 1h 54m 51s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a8d7a8804df4cdfa67b4f3000290dad : FAILURE in 1h 31m 33s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d674270345724a4fae11210af0f62f66 : FAILURE in 1h 58m 18s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0f72be25501438f863e69597fba2c31 : FAILURE in 1h 03m 24s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df8297403f1942c7bf67375ece95f340 : FAILURE in 1h 32m 53s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e1c418a41ad42f49c06ddbe328472fb : FAILURE in 1h 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7ee676f8a81bad7b3ed0d3f3bcf6f71f5b36fdd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-04 10:05:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"72ca13c56e76c5889790021a5308e514b46d5580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 12:07:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/527065c50d8e403abeabe775ade2a016\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0af0fc148bde482787d66d569b8659a1 : SUCCESS in 12m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5aa3145dd7bc4b11b18c97c56197c6e0 : FAILURE in 1h 04m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e8e1479aee04572a2943fec32827bdd : FAILURE in 1h 11m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0a06cc571bd49588c46b23591c9f1e7 : FAILURE in 1h 09m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/032a5f37da0d44d8a04cbb3f1bba5c0e : FAILURE in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc18c2e8b1a4fd1988127b16eecfdf9 : FAILURE in 1h 05m 57s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ee37f33806d4d568b2c8dd3d5ee6b6b : FAILURE in 1h 35m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9065c3885804495aed50a7b0b6a0b26 : FAILURE in 1h 55m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2101f6628de845069b84d8da5c0cd8bd : FAILURE in 28m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2267c1d5f03d4c89a3aa69d60c361a8f : FAILURE in 22m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e501af34213c467fa4a87bad54fc619b : SUCCESS in 5m 49s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/370a63cc1bee43c3bea7f8b3b78d70dd : FAILURE in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29e96e2d48047978032301ab3b868df : FAILURE in 1h 11m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/19aaebb82912497b844a23d2cdf7eeac : FAILURE in 32m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/2aa82525ca244ceca853a7425ccdb5df : FAILURE in 1h 13m 04s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfaba998fe3b462981cea6aefdac6801 : FAILURE in 34m 12s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c47db8819f6c48e288e18ecd6e7281f2 : FAILURE in 1h 16m 39s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f9933b1284e4155beabecdedf38a535 : FAILURE in 1h 20m 29s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ac3662fe2474a248e313a95b1900b8e : FAILURE in 1h 35m 27s","accounts_in_message":[],"_revision_number":2},{"id":"531d245bf74cc6e88f5831b8f4d9622373cdfb01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-04 14:36:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f3dfff8f282a58dc565fa7f68138827eb49a0d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:13:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7723f7874d144c9ae18270d822c598d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a0f556e544b4efba78e1ebc09dc6b13 : SUCCESS in 12m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/851034f3b3b74a1c8531f82315635afd : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bff0981a57b487590d349e082beb3b2 : SUCCESS in 1h 29m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f41411116b4848a3b921f56861eb40c9 : SUCCESS in 1h 27m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a975360e43a24b0b8f31b69202546302 : SUCCESS in 1h 32m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4928d8e51ad644d48110011686acbd38 : SUCCESS in 1h 24m 25s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cd0de57e8704d13b4767aa408c3b5cb : SUCCESS in 2h 21m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b322067054074311a0cec32afc9ab9da : SUCCESS in 2h 25m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc0140ffd6e0471bb2ea0628e96fd564 : FAILURE in 24m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bff7910b75924e41942a0abfa2bdb379 : FAILURE in 23m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe65e61fc1747acac9de03f9ee256dc : SUCCESS in 4m 50s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad42661636dd4749b7b9c0ef90583ff6 : FAILURE in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c1105bfe2f448c897a76d38cc6d1f86 : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9ff5e6732e44b71a601c40ad1edfb9d : FAILURE in 40m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d79764070ab47b0b13290a4b0a4e18c : SUCCESS in 1h 32m 00s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27bf120acc9046cb9651d86e23e13764 : FAILURE in 33m 16s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/62b90ffb69614236a26dcb320defc8b4 : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29272b760d2d4fd6b840493698ad68bd : FAILURE in 1h 39m 15s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eb86b9719394c3a8152c8a5d34de425 : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"d6549bd57efe8bca30c237c507e686e5e3ab1d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-13 17:14:25.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d681580c041717b0a6eed550f8df47540198b9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-13 17:14:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"957d9f0c9144cdc71f1fe671e96ca41d92c604eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-13 17:15:26.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f210772242395afb2c32d65dcbd7a600f302de1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-13 17:16:07.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9a9f3df47979d3007994e9f5055602f308ac7092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:19:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c63fb7802af493cb7d677f412cd0097\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fdb47c4f29b45e1b8982d20562b6a16 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e46a9b8b9e0f4d54a8f83350e8e5d785 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbf7e5a6cd5644119e3ef8145ef07417 : SUCCESS in 1h 19m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9fc416e2e4341d7a32a2ea5a3e382b3 : SUCCESS in 1h 51m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb70fbcad0c74f2b8c56d3df833e3321 : SUCCESS in 1h 30m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57e1f8336a7844cdaff08a7035cb368d : SUCCESS in 1h 33m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7639dcace0134b12a4ef1e711a15f5cb : SUCCESS in 2h 15m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24baffe645c0403aa508d16548f10e43 : SUCCESS in 2h 38m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ed255fc842545e9a7c62174822d6a55 : FAILURE in 31m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d00dce3fcf804753bdd5f2bd4b5466cc : FAILURE in 32m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c82350b0be484c9d10b4fe0d349408 : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73cdc2f7175b4f8786a2f14c9ff11643 : SUCCESS in 1h 48m 45s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42de14edc9094ccda1160f89cb95b15b : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e66b8556424494086c2682fdff3e207 : FAILURE in 31m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d876582bba433c9b2f9f5d898867a4 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/18e96ab600af4f2e9031b82a95f937b4 : FAILURE in 31m 47s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/aace9bf1999c46a8b2fea7c8f15f093c : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03fbdcfef6514c2a95b719f6186c5dc7 : SUCCESS in 1h 42m 30s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371d083ce47a44668bc4e19da00aa1a4 : SUCCESS in 1h 33m 17s","accounts_in_message":[],"_revision_number":7},{"id":"5fee9ab87ab514c4b7c6a0fb8a4b6b5f89f9bec7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-14 13:21:29.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"f7ddbd625f7fb712f459f845c31d6b116caa0591","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-19 09:42:43.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"078135b4612fb0f138cabb0f35a50bbac5146f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 09:44:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7c764817a491add08e600020d2c3127038b5ebe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:49:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f5bb6574bd45e194d07f5899b7f069\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e56535387d544cfb9c9d37f23e552623 : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/94c4374a468449fc9fbb8f71649861af : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b53c9b77f8d240fbb6f624cfb99302c6 : SUCCESS in 1h 41m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/545ee1314e60470dbcdf6bf3f5eb60e5 : SUCCESS in 1h 18m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd09d47f36b24ba697b006d16bbdf6ec : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2eb2fe3ce6a54ec39972f0f783131d62 : SUCCESS in 1h 46m 40s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae872d10ec4641bb9e57cc6b4f860ec7 : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/832492fdf8b341b1848a32ed048be5eb : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e5b575023fd446cb815641ad4b4a7ff : SUCCESS in 1h 34m 04s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a38efac9c9d4f179429784e9911deb7 : SUCCESS in 1h 59m 59s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ed1ab18b9a402eb9444de92a38c83e : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":7},{"id":"b6ad7381500a53156700b15e19d5456188a0887a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"fd4c706001385564c558ffb5f0c192c923d030f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:50:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56e33fa07a74a2886000ef501866551\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f25b6970394482bdea79fe06a040be : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"956896e8a75616c1b0f79dd5d10e59c7a497803c","revisions":{"a4848415cd981e7faef4c60582a087159725f404":{"kind":"REWORK","_number":1,"created":"2023-02-23 17:48:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/1"}}},"commit":{"parents":[{"commit":"2d7f8f464357fafbf7a2056fdffc15e1e3912963","subject":"Merge \"Generate OVN certs only for OVN scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2d7f8f464357fafbf7a2056fdffc15e1e3912963"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4848415cd981e7faef4c60582a087159725f404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4848415cd981e7faef4c60582a087159725f404"}]},"branch":"refs/heads/master"},"5ed0164bf49f12d34d2affe8753a95bf437a9320":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-04 10:05:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/2"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-04 10:05:55.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5ed0164bf49f12d34d2affe8753a95bf437a9320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5ed0164bf49f12d34d2affe8753a95bf437a9320"}]},"branch":"refs/heads/master","description":"Rebase"},"a4dc7657ef1176325c8e3112c87e579a43379456":{"kind":"REWORK","_number":3,"created":"2023-04-04 14:36:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/3"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-04 14:36:40.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4dc7657ef1176325c8e3112c87e579a43379456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a4dc7657ef1176325c8e3112c87e579a43379456"}]},"branch":"refs/heads/master"},"8ef3243ba0e314d0870ce52dad860cfa23e12bee":{"kind":"REWORK","_number":4,"created":"2023-04-13 17:14:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/4"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-13 17:14:25.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8ef3243ba0e314d0870ce52dad860cfa23e12bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/8ef3243ba0e314d0870ce52dad860cfa23e12bee"}]},"branch":"refs/heads/master"},"de0c8de0dd33245c41d8218f619d83d5555c63ae":{"kind":"REWORK","_number":5,"created":"2023-04-13 17:14:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/5"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-13 17:14:56.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/de0c8de0dd33245c41d8218f619d83d5555c63ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/de0c8de0dd33245c41d8218f619d83d5555c63ae"}]},"branch":"refs/heads/master"},"5ee750b331c626f351f10ac1a7eed3d18baf0ed6":{"kind":"REWORK","_number":6,"created":"2023-04-13 17:15:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/6"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-13 17:15:26.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5ee750b331c626f351f10ac1a7eed3d18baf0ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5ee750b331c626f351f10ac1a7eed3d18baf0ed6"}]},"branch":"refs/heads/master"},"956896e8a75616c1b0f79dd5d10e59c7a497803c":{"kind":"REWORK","_number":7,"created":"2023-04-13 17:16:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/874949/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/49/874949/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/49/874949/7"}}},"commit":{"parents":[{"commit":"f1a8c358531bdf86d8aeda725bc0b1c347d325c1","subject":"Workaround ovs bug that resets hostname with add command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1a8c358531bdf86d8aeda725bc0b1c347d325c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-23 17:48:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-04-13 17:16:07.000000000","tz":0},"subject":"Use include instead of import for conditional tasks","message":"Use include instead of import for conditional tasks\n\nWhen import is used ansible loads imported role or tasks which\nresults in plenty of skipped tasks which also consume time. With\nincludes ansible does not try to load play so time not wasted on\nskipping things.\n\nChange-Id: I50b99306a52f1a2379e55f390653b274afd5885f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/956896e8a75616c1b0f79dd5d10e59c7a497803c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/956896e8a75616c1b0f79dd5d10e59c7a497803c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
