)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~879103","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fzed~I8bee1850e3a120f7b76f586909e6d74361696e32","project":"openstack/openstack-ansible-os_neutron","branch":"stable/zed","topic":"osa/rdo_ovs_3.1-stable/zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-07 15:23:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I8bee1850e3a120f7b76f586909e6d74361696e32","subject":"Workaround ovs bug that resets hostname with add command","status":"MERGED","created":"2023-03-31 12:34:38.000000000","updated":"2023-04-07 20:05:41.000000000","submitted":"2023-04-07 20:04:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879103-osa/rdo_ovs_3.1-stable/zed","cherry_pick_of_change":878911,"cherry_pick_of_patch_set":6,"meta_rev_id":"9cfc65cc93fa42ee033c5ec7d536760f7f86a576","_number":879103,"virtual_id_number":879103,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-07 20:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-04-04 09:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-05 06:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-05 06:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 14:57:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 21:08:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-04-05 06:18:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"4c17dc48ac9dd310512ffe8800eda811e6ee7466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 12:34:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ef482c1f2ad109a6b7e70aadb2ccfd230e145fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-31 12:34:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e2c978e9cfbceaf541a044004e277e7fb925897e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 14:57:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffefadc8b95d49dab6e1b92d7f152ab7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba18ba43495f458d9bf47cec9f213e43 : SUCCESS in 9m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b4b5f8693d44cd7ad00eb28c6b70ee3 : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b99c015cb78349c98d7b660e6549138f : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bd97d764bdff47d2aa96d864aa3bf059 : FAILURE in 1h 04m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4393f771c12340aba55de5aa3e5c797b : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de4cd66c40f1485485058cf38951c9f5 : SUCCESS in 1h 13m 34s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a12ed2b7fdea4b90a32b220c6577b6fe : SUCCESS in 1h 45m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a6e2bdb2bf74ab4aa0523f408110c28 : SUCCESS in 2h 04m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cae09e708984b42975b12aaa360c164 : FAILURE in 44m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/938b49c878864b4caf631e5d1488f345 : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff71ecc9ca6400e8862d720300ae6c6 : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce629c9cc0c4ef8ade2162d9197d80a : SUCCESS in 2h 04m 57s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15f14a0976174da0835931abef9d28d9 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9224e540453840369680456d0154a712 : FAILURE in 37m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbedfb6cd3674c058524546f2333c5fd : SUCCESS in 1h 15m 54s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b9e1fb21dad41a8a0f19d5210596856 : FAILURE in 35m 01s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3754781ce553474eb9790ebaed0207d1 : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3acc77438ea41ae8ddab56c65d45b9d : SUCCESS in 2h 02m 13s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/338b19a135b6476eb243ec7a8db42749 : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4507458a030452885b8a0b592c162a85b6fa2267","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-03-31 21:08:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c9d5825afcb59229cbdf8425ae525d4b0a98570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 23:42:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59673afa04674c50b5ecef1fe3218d05\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95bd01de1d554b1ab930247a8c2cad5c : SUCCESS in 14m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/43793eb112e747069de9f53709d7db3f : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aa6fececd4541708c27d8838f07f871 : SUCCESS in 1h 02m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/632225b85e044afc8c6dc04c2e69c4e9 : SUCCESS in 1h 01m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d3491cb96a44ce79cc8aff55832011e : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f065c0fde6464c9480ab149f3111bd1d : SUCCESS in 1h 15m 16s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/725acd572fdb4438aa12224587ea6f48 : SUCCESS in 1h 52m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81dc195357234b209c338c22eef2e7ec : SUCCESS in 1h 58m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcf319f052c84392b0d9e63fb08e841e : FAILURE in 32m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/543dfa71a84b4092a62257d3d13e7fce : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40045ee54fb4404eab4545d02d2a2dea : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd7ebef795d24942bf8a8ad170db1ceb : SUCCESS in 2h 01m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75c274f3c144404ca04fc584f9b82023 : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/696dcd6c2b814882885c3439e9a034de : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc7bc3c929674a1696e55d0758218ec5 : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c19e16f15cad42deb0f45477171fc61d : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1e45936c4744dfa9959c684e5aaab0d : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7472beef1f6242498083abc42d478032 : SUCCESS in 2h 06m 13s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c51bc9b013f46e380843b7849528884 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c40e9f31cd468d56af126c0aa92b7e5720cf6c06","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-04 09:03:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8eb80af43b353ae962f60e2676b3952eb291f3e7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-05 06:18:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae2cf211a15d849aa49fb052859f2e030059a14b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 06:19:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c510bac556c56fa5df82dd8a48fbdf18ce2a417f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 08:15:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2417f9436f407f90a3a839b899483d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b4b11202a904dd88cde170ec0d9dccc : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/60383680554542c0868b41780190f707 : SUCCESS in 1h 14m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0230df8f7b2e47b5b6c481e6bc086b60 : FAILURE in 57m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e079265c1b8f4dccba28b2fdda0cfda7 : SUCCESS in 1h 13m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d19fc55cbbf49aeac6710a908a49f17 : SUCCESS in 1h 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888fc103f92c44debe1b36f7c17a4053 : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740e10c9e6964cfbae6d33c648b62212 : SUCCESS in 1h 47m 48s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67812bce6c5b4e26b1a457d3ef89f1f8 : SUCCESS in 1h 06m 02s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/9477ca2afa75407ea4570b87fcf6f7ae : FAILURE in 1h 09m 40s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7989a86cfa0344c58e2f44de3bd3d21f : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd39895ac90140388b4c9396d1deb58a : SUCCESS in 1h 53m 53s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d8032cb4fa54bc2806669a794f54211 : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c037d3abdc3bde320faabde35f8b053e73f86354","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-05 19:22:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09fe5391d7814fd3687d8d7a894240c43f75cd12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 21:30:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba4518b62b084837951370cab0622850\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8562391428184f67b5fdfcb047c87970 : SUCCESS in 10m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8ceb3d0863d64fbeb3e10aa79203fc1b : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a3fc7819bfe49e489769dca9f0c274c : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd64e4a2da1940d1979d55d80b782395 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9bf10c263ec4e17a4263b60429b5608 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26288a87e8a64d55b7b496eed5a6f9c7 : SUCCESS in 1h 13m 04s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c9d88aa73d74ba2b50490f51734f457 : SUCCESS in 1h 38m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31e155d3b99646fca67d4deae8eee11c : SUCCESS in 1h 43m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e081a1b84d2c4a25bb8884cda279f73a : FAILURE in 41m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f320552a98e4ec3bc2d3ee2780f83c1 : SUCCESS in 57m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb91bf468f9347efb2db64054ceaecee : SUCCESS in 4m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0603162a101c48e6ad564ffd242633e1 : SUCCESS in 2h 01m 05s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/693fffecd49f4ded966c1c61c8bfeb5f : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/039399be049045009d37380c98363355 : FAILURE in 34m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9d73799087d46318ea6a9b3685243e4 : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/54b7399e666d436985aaa18eba84435e : FAILURE in 34m 20s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c10406cbc5e54abea4a918cabb16b5dc : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b73f12d9f77d4151aff463bb8ab0eb2a : SUCCESS in 1h 54m 17s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb3a59cc0f87450db881ae7d8b27030f : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ff31ec12622373568017a850c5ebd3b72720cf85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 21:32:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abaa731f0790afa5885fb3f29aff08d2669bda07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 23:35:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24fe5013f96c448883c1f9b8848f3d73\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8965bdd7c64c4fe790121c011eebaa47 : SUCCESS in 10m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/76b2edf83b5b4a0aae2274c6f992042a : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/32884393e0ef41d6a56b604d8ffaa9bb : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee51861d84f14250aecfe9dcc54efc43 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54f2721a20d84eb08d1ca51457ea949a : SUCCESS in 1h 08m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c60761e07d4f5182d3bceed2dab1bd : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57ce86b1dbef425896c41640a44b7fc0 : SUCCESS in 2h 00m 41s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b881d91a56c34dd58f2ec74492fe4201 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/91a88716b214440484ab476f8d7d2912 : SUCCESS in 1h 05m 40s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3c0c71ffbba4e5fb844d439b9bfe49f : SUCCESS in 1h 16m 25s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d79c4939a21646309de4a1aca5b529ba : FAILURE in 45m 34s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7527aa9a3427475285c025f96adcd224 : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e8f79fde90a5934daef1cf2ef16a8df1ba6d4c65","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-06 08:09:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d608fe120831b7377d82323c6f1ddfd0886b4abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 10:18:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52057bf728df4b6d845478644fd5abba\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70b8cfcb0b634842adbd5cabf1e24f0e : SUCCESS in 14m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/36cc5190148642fb8d2b4d02e9878301 : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/36037e017b0749c4848dc2b3830590e0 : FAILURE in 25m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0704bc2589741fcb35be164c07cb5a9 : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7d4b23f46a407d9df641f08aed0396 : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b50f0d5bd1242b4b2ef2a7da8e8ab4f : SUCCESS in 1h 14m 05s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3782534e54f44289bd4c5b80aeb0a7cf : FAILURE in 23m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/173a263c31194513b01d61d32626106b : SUCCESS in 1h 38m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/235f020d788b4981a180e9551b0c570a : FAILURE in 35m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/920af21196ac4a88807466e0eb401aaf : SUCCESS in 50m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e8e29423c754cadabac35bfd60dee9f : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14a5b4533cae4b13bd25dda3952b8c09 : SUCCESS in 1h 54m 13s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb4cd0d2e9ba4ddd9de408abbde155bf : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d2ad4429d744abb8044c87ebec589b : FAILURE in 24m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6406744485de4010b0e2f509fed85f3a : FAILURE in 24m 28s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e50ad9654f4242c7a1fa600698af821f : FAILURE in 20m 23s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc95c0ea6d204309a2e43e422b0c0d48 : FAILURE in 24m 51s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7cf639caa734184bdfb2276bd3ce0a9 : SUCCESS in 1h 56m 29s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa12fbd7267429fab790dd1c91a6033 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":2},{"id":"56db51212afec2a69529c9d47ab9d4e9d8de0c42","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-07 15:23:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16b7dd7e4b801eb04de2d490bb2fdc4ba6b80899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:47:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d74c91592434f9199fecaf800d21703\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e01d3f5fd38c40f6963e14574b6954f7 : SUCCESS in 11m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/af977e7fc56f40bf890c1032322a60bc : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/84b16db189ee41a0aa95eca996065887 : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/522243011b484da3a6cd20a2646254e8 : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369d8098c1ff4ec0bb1aff40b75c8bce : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97054a0d8aaf408aab855da9152ee773 : SUCCESS in 1h 07m 39s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/58988269cc8f4c1897ec9055535e1817 : SUCCESS in 1h 45m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/739f52be72da487aa48d419f339f5d37 : SUCCESS in 2h 07m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4242bf9b32fa43ca883772fe0ccffa6f : FAILURE in 34m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05d8d339a0a041abbd507e4d31026a2b : SUCCESS in 58m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b5223a7a0d4f02ab915d0bc3b8e594 : SUCCESS in 4m 48s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a586eb04cac41cbbee12be9569f3c75 : SUCCESS in 2h 04m 13s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99888e0e030b45eea385d1945887c043 : SUCCESS in 1h 15m 35s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e76396c9ba1482babb29be789bfcb60 : FAILURE in 20m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/cea5896180fa4291bc914c197bdc5ffe : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3522fd424dfa4267b449a6394f3f345a : FAILURE in 20m 12s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d02a6f75262a4b1ab6273e0e23fedb95 : SUCCESS in 1h 10m 29s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fd6764b547943c69d85cc50b1a9a586 : SUCCESS in 1h 58m 28s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2cc7f05087e4c69a03d3fe21c7aeb10 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0b9a1a7b5352c57ddde3dfc51d3255407b61ef3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:48:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7100c4fbfa89353da52fa46234773c08861ca38c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 20:04:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da29a3384e342efbbbf5603a50c22cf\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1b2a2a58b854c4c8941a90e126f1f67 : SUCCESS in 14m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/45b6f530421347e2ab3a3bf26f65696e : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f16bb1b2c47d4e39948133690598d9e8 : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4cffb1c2bc94fe092d8c0893b171f33 : SUCCESS in 1h 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c146d23ea5244f58be29ea397f229775 : SUCCESS in 4m 23s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67df67b313da43189d34a4bb52154a4a : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf4cf8b262eb4aa68822c379c61ebada : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a157cc8e3404507baa46943992bf4fc : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ca35ec173084e1aa9ef5c794dcc1368 : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaacedd3b511417ca4c56c80297427b5 : SUCCESS in 2h 08m 06s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad11e6ee5b9448058455f4226468343a : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d55473613a92fdc40fd45b44db0c6992cc01befb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 20:04:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9cfc65cc93fa42ee033c5ec7d536760f7f86a576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 20:05:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a63cad4936e456da4719415931f709f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7d97eb61de426aa8e72a37cde43fc9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9bd23d7d39b71d27da68059457249a1f6eb3a538","revisions":{"553e8b31bbedec85c45fdf6d3312e76d91740f33":{"kind":"REWORK","_number":1,"created":"2023-03-31 12:34:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/879103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/879103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/1"}}},"commit":{"parents":[{"commit":"a48884675bf54f2b63bc9fae8aec35ecfcada4dd","subject":"Merge \"Generate OVN certs only for OVN scenario\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a48884675bf54f2b63bc9fae8aec35ecfcada4dd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-29 14:24:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-03-31 12:34:38.000000000","tz":0},"subject":"Workaround ovs bug that resets hostname with add command","message":"Workaround ovs bug that resets hostname with add command\n\nAfter RDO bumped OVS version to 3.1 from 2.17 CentOS/Rocky fails\ntempest testing due to systemd unit calling adding hostname [1]\nwhile ovs-vsctl add in 3.1 actually behaves exactly as `set` which\nsimply resets defined hostname on each service restart. To avoid that\nwe\u0027re adding `--no-record-hostname` flag that will prevent such\nbehaviour.\n\n[1] https://github.com/openvswitch/ovs/blob/branch-3.1/utilities/ovs-ctl.in#L51\n\nChange-Id: I8bee1850e3a120f7b76f586909e6d74361696e32\nRelated-Bug: #2013189\n(cherry picked from commit f1a8c358531bdf86d8aeda725bc0b1c347d325c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/553e8b31bbedec85c45fdf6d3312e76d91740f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/553e8b31bbedec85c45fdf6d3312e76d91740f33"}]},"branch":"refs/heads/stable/zed"},"9bd23d7d39b71d27da68059457249a1f6eb3a538":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-31 12:34:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/879103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/03/879103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/03/879103/2"}}},"commit":{"parents":[{"commit":"16c7d9d943c5fc725aebfe7acd2a188f70214515","subject":"Disable CentOS LXC jobs due to the bug in systemd packaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/16c7d9d943c5fc725aebfe7acd2a188f70214515"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-03-29 14:24:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-03-31 12:34:44.000000000","tz":0},"subject":"Workaround ovs bug that resets hostname with add command","message":"Workaround ovs bug that resets hostname with add command\n\nAfter RDO bumped OVS version to 3.1 from 2.17 CentOS/Rocky fails\ntempest testing due to systemd unit calling adding hostname [1]\nwhile ovs-vsctl add in 3.1 actually behaves exactly as `set` which\nsimply resets defined hostname on each service restart. To avoid that\nwe\u0027re adding `--no-record-hostname` flag that will prevent such\nbehaviour.\n\n[1] https://github.com/openvswitch/ovs/blob/branch-3.1/utilities/ovs-ctl.in#L51\n\nChange-Id: I8bee1850e3a120f7b76f586909e6d74361696e32\nRelated-Bug: #2013189\n(cherry picked from commit f1a8c358531bdf86d8aeda725bc0b1c347d325c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bd23d7d39b71d27da68059457249a1f6eb3a538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9bd23d7d39b71d27da68059457249a1f6eb3a538"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
