)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~882004","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fyoga~I831f6d62f0d31384258571e01a4e7cdd75b73e2c","project":"openstack/openstack-ansible-os_neutron","branch":"stable/yoga","topic":"cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-28 12:57:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I831f6d62f0d31384258571e01a4e7cdd75b73e2c","subject":"Ensure service is restarted on unit file changes","status":"MERGED","created":"2023-05-02 11:02:00.000000000","updated":"2023-09-28 17:24:04.000000000","submitted":"2023-09-28 17:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882004-cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","cherry_pick_of_change":879977,"cherry_pick_of_patch_set":1,"meta_rev_id":"8001b18d0a529e1c07031bfabf9db02bb2525762","_number":882004,"virtual_id_number":882004,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 17:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-08-29 18:53:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2023-08-28 11:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-08-29 18:53:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 13:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 18:53:31.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"ed33b2170f900972853d2d8742aed0c0739e263d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:02:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3bf5110f2fe146da87138e1555ea1ee240a9c001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:40:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c389044c62944aed8c815864744476b6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/902ec9cee9994ce5a413e31ec21c84c6 : SUCCESS in 10m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/faab4882a0b74247840e491659eec781 : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/60a4efb5411344f799e42ed3588c0df4 : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b98f319c0154eb7ae192d7105d0585b : SUCCESS in 1h 24m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d7f460a41c7d443b860412ac72bb5ef5 : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3e3029b34c7419a925654d4168c973b : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfa4b3cee1b047af9953aa684767cff2 : SUCCESS in 1h 23m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ad87b4caa8d546e3a93559bc9141eb21 : SUCCESS in 1h 35m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cca620a2b5844d4883c94a79579b0b1e : SUCCESS in 2h 27m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18e481ba3d224cda964fc065625c7146 : SUCCESS in 2h 01m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f529d08f13ed4fd495324569029a294f : FAILURE in 1h 16m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/075321561ba5448ea107bbbb8fb8efd6 : FAILURE in 39m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba49732afcf54605bec55e7b69f6352a : SUCCESS in 1h 00m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524b8019eaf74fecbb384fa478058cd2 : SUCCESS in 5m 21s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f1060b8a5d419cb776f8a9cab9119c : FAILURE in 1h 19m 51s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c41ce5cb9794198953a6fab6609b571 : SUCCESS in 2h 07m 01s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/505e77e9bf1d4c949dfc6d29e1df1cdf : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e219729b90141c9bed24e7afbe0d482 : SUCCESS in 1h 56m 23s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/c00f531de07f48b08e45cfad523318bd : SUCCESS in 1h 28m 37s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9310ca13ab0e444da4411949046bd825 : FAILURE in 2h 09m 15s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2727a0bd9fd64cdab7563d9427ec1c0d : FAILURE in 1h 28m 17s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f0b3ce18487f4737a7d72621554c9b87 : FAILURE in 1h 56m 19s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c094af1443f24433b954486238d7110b : FAILURE in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d65b42451b57b2c6338b95b53fcda22f611f996d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 10:31:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a1831a658d2347dd931debdca6a9c8581c7bf67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:49:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8627c2b1c6f74c8ca8544d34806d490b\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c83ab8dd8aa4cf3a2ced1b39db6eedc : SUCCESS in 10m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e738e4ae5433493dba2bae5f2cd1e042 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd97e456e7f94fa781f5ec4b424e83c0 : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/244fa9c00e49470db6edfb6a09f90138 : FAILURE in 1h 00m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fafdd9b3d6fa432bb9138d12930cb93d : SUCCESS in 1h 12m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e9a841f9aaf4a09a88603883da6515e : SUCCESS in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bce019df0d8d4c3f9aeabde9b4cfa728 : SUCCESS in 1h 23m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7a0441ce340941ab8e6c40ff6ad6805d : SUCCESS in 1h 26m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3533260200d43438ac9e57bc1257264 : FAILURE in 1h 02m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e135fc1db954f5095031efba179596c : SUCCESS in 2h 02m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0605f8b2f5a4908902a55fc3855fe3d : FAILURE in 1h 12m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8e12bb705cc4321a4f1ec635cb8fe82 : FAILURE in 43m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b03ec2189914c1ea0b0df70350ec36b : SUCCESS in 58m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06a031f20a44f9c98f0960179fda9cf : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d281b7c299d43279ff54eda71cc206e : FAILURE in 1h 12m 08s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17291b4bfa3d41e0abfbd5fdc62de544 : SUCCESS in 2h 10m 29s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d1130d139e452884c46557b8c10a21 : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1543b2a8642f47059726a13a69929be4 : SUCCESS in 1h 57m 30s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9ee53eceb2f44a692a1ef1b5496d0ea : SUCCESS in 1h 31m 14s\n- openstack-ansible-deploy-aio_ovn_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a4ef7fe073d4b2380ac1319a4974ed9 : FAILURE in 1h 54m 00s\n- openstack-ansible-deploy-aio_ovn_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/92043fc571f04ec6b9c275957c5600ff : FAILURE in 1h 21m 08s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/01d2ae8ed3404117851d5343f4ff5f3a : FAILURE in 1h 55m 53s\n- openstack-ansible-deploy-aio_metal_ovn-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/b5f11349e10e4a268069f3255b838c76 : FAILURE in 1h 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"803f6608eee2cb63c2411a32e7c1135957695f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 14:43:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"362e061c04657bc50f5240f68e72fd77c6f3e934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 17:08:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e278bb08114a2f8c99d3486bc693f2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a419b99076d4b368e9ae6bcbefceaa5 : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d45010f87644983ae401ad07a054564 : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb91595b0c644966b54d9f8e0f31024a : SUCCESS in 1h 35m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1bf428929474a2c84f89ecde05fc6b2 : FAILURE in 57m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2945ea66a14d4614a98cf3fa16f25fe4 : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a47d6f9509284bb7a6c8f104d594fd85 : SUCCESS in 1h 20m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/661200e1bde84f4e9b98ae533bc4045f : SUCCESS in 1h 13m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/53e3fd4be0aa4b33b94d568bc20bda63 : SUCCESS in 1h 26m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/24e59caf2dc04a82b9eceec9eee66307 : FAILURE in 1h 11m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bcb708cc2454ed8ae42ed77f454848a : SUCCESS in 1h 46m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc4187898e8d4dcea3d2b7c8c49e0c68 : FAILURE in 1h 11m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2d1beb590374f83a0b35f49781fc507 : FAILURE in 42m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a23ce36b4c50479b853c6ac471cfc8ee : SUCCESS in 55m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a6917f467949e78f121226645a2381 : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a04698f1eb74cd184fb5561a5a74f9f : FAILURE in 1h 21m 21s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab5cec0f67f04447b27f36b674f3128d : SUCCESS in 2h 12m 07s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da8e9dfe6e4c442a82a117a5bc8660be : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/344ddcd590834fb18de0e0bb29472218 : SUCCESS in 2h 15m 33s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef1e33ab5d5c409a883894befba1e9b3 : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"1f2e4b342e6127d56cdba2ad95af6ecee9e2afbf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-28 11:03:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"74e264437364dd6129e3b21a6d461f5727e12b4f","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-08-29 18:53:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1109508c493835828352ba1391cd5f4ad5752a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 06:22:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a8c57acb49f871355b5bc18f071c37c559de1280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:27:03.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/892364 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"15b80dbde340e2adc49bbe46a52ca84bf856bf11","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-12 07:49:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5b01c4da5ce858e7cc66a263f9126e65658ac38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:00:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ed1875f0964bae99a338a5cbf70c63\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a707f4015f304fbebf812dfcd606a08b : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b00ac585432d45029ae277efb50000af : SUCCESS in 1h 06m 57s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/51e638e3061e41718f4ee1a1e291c1a2 : FAILURE in 1h 43m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/81b0577b8ff2472d86d6c5f8b46e995a : FAILURE in 1h 05m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ceb1e038aa114e65b995f99d1e5758ed : SUCCESS in 1h 33m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2baacbbab06248a295ee60be5664ca4d : SUCCESS in 1h 40m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbff579f27ec42c2a21bcb983c55714d : SUCCESS in 1h 28m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/228c313af6b14846bdc769f4e1923465 : SUCCESS in 1h 18m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bc8c880a7104b42a7fd8ed87333d1d6 : FAILURE in 1h 04m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbc19f9fd75748479a93e8d0ee364786 : SUCCESS in 2h 07m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1a20fc9deb44fc883fb2621e87a8037 : FAILURE in 1h 00m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0cf43b24ccd427e9dd46d7ea903da6d : FAILURE in 42m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8aa7769667f419fb6c337fa9723eea8 : SUCCESS in 58m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e62b326d5740fc88d49b0c5202b91d : SUCCESS in 5m 00s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964559fdaaaa421b9c85285615bb780d : FAILURE in 1h 16m 04s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad5356d8a2ad46fd98c329f38e516111 : SUCCESS in 1h 49m 36s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad3e72d1a4ec431991a3c348ebe2b62e : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/22bc65d2599840c2b85859470530f20f : FAILURE in 1h 58m 27s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/02b5134b8ed144249cfd990232fcb9c7 : FAILURE in 1h 32m 11s","accounts_in_message":[],"_revision_number":2},{"id":"13f0e57d414d1a80f90774f948212578e78aea05","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 12:57:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5683e99349d9222f0ed346cad722eae664293e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:06:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21265316fb064e2b9422a180ee73d304\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/435233640a6745d0adabec3960d2d7c3 : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/97c88b2dc7084ffda0279652aad8a308 : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf584fef6cf14f549bfaabd3ea3713f6 : SUCCESS in 1h 16m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/de54440fb5a74ac3b0e16e9eb32466ec : FAILURE in 1h 00m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19684023872d44d9b382757e4b5f95f3 : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af48f4a315b04c489fac486fc767d772 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4972f9076c9f4932bc0f68d96e0edf48 : SUCCESS in 1h 07m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6b0fd89b093649bab5b113d8afa73bb3 : SUCCESS in 1h 40m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/afde3c9a148840b69251b43cde3c9eef : FAILURE in 1h 03m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af0e782835674dfdb8987a5064a00d6b : SUCCESS in 2h 04m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bddb8eeaa454dd1a1fa8c40486ca639 : FAILURE in 1h 05m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/781aaff9761a4c0189239c8a0312421d : FAILURE in 35m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f384f704f3d7476ebac27e8a66b49b80 : SUCCESS in 55m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e178bd246c247a4b6b0f11abf58011f : SUCCESS in 4m 43s\n- openstack-ansible-deploy-aio_metal_calico-ubuntu-focal https://zuul.opendev.org/t/openstack/build/123e89f7d97943b3b7f48b9e20539395 : FAILURE in 1h 07m 56s (non-voting)\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a51d37437484893af5ac06942399235 : SUCCESS in 2h 01m 36s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cbefa72d4994aa195acc8f10cbba1d5 : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b31bd60dee0f4c1e8edfcf07401ec632 : SUCCESS in 2h 07m 18s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/90b0de8a55e2462e892e4cd66ac64dd7 : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":2},{"id":"41f997573c400a549ec4ec562653607810d39f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:08:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c83f1f889cb28780b7597f3f25c0bdec08142eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:22:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79ec53fa2d2488bbccfa244ede74f55\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6913cc5df254eb18b25e16a0b11576c : SUCCESS in 12m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8ff73ade3d1044549125a114754f27bd : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f14ed67954b24934b034c8ec1a6ae675 : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea5a335c8b05497aafd4ac2640355548 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17584c11350a44a0987efdadea9357db : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/59991afff3464451a9c0b19b47a870e9 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a53912fd86b4cba8d8f7a17dd4c1477 : SUCCESS in 59m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a578f8726b143bd863a04760e729571 : FAILURE in 46m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2f6ceca5af42d09ce86a3970dba5ab : SUCCESS in 4m 28s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61a0aab300ed47c08c26260bd0952073 : SUCCESS in 2h 09m 31s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dba2313655e04ad09e30b489738ba85b : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f8ad82eacac4275a00ef0626df8a313 : SUCCESS in 2h 02m 52s\n- openstack-ansible-deploy-aio_metal_ovs-centos-8-stream https://zuul.opendev.org/t/openstack/build/f48ffdf71a154c2f8957e3f97fc95740 : SUCCESS in 1h 28m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1c250afce815f83ce8ff1b5a2d31d0df1f957393","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:22:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8001b18d0a529e1c07031bfabf9db02bb2525762","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:24:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0c0b44efe3493d8713da8f663665ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f8d0e4f46c454ba9a292de1a90178a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"106575241c81087bc6c7f57c5e75a3fe72032c91","revisions":{"6dc5240460b9438df90b55b71011395c9823b49a":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:02:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/882004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/04/882004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/1"}}},"commit":{"parents":[{"commit":"d37795a5b857b0d13bf9faf7881108b14cfd2474","subject":"Allow to provide custom configuration for VPNaaS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d37795a5b857b0d13bf9faf7881108b14cfd2474"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 14:57:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:02:00.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I831f6d62f0d31384258571e01a4e7cdd75b73e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6dc5240460b9438df90b55b71011395c9823b49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6dc5240460b9438df90b55b71011395c9823b49a"}]},"branch":"refs/heads/stable/yoga"},"106575241c81087bc6c7f57c5e75a3fe72032c91":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-22 14:43:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/882004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/04/882004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/04/882004/2"}}},"commit":{"parents":[{"commit":"7df3ee54aac881f1fda91556dcebe20218c195c1","subject":"Drop OVN CI tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7df3ee54aac881f1fda91556dcebe20218c195c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 14:57:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-22 14:43:26.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: I831f6d62f0d31384258571e01a4e7cdd75b73e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/106575241c81087bc6c7f57c5e75a3fe72032c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/106575241c81087bc6c7f57c5e75a3fe72032c91"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
