)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~890089","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I74735ad2f127a4c62d4e5c4d24dd1af76e5b76a3","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/db_port","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-14 10:53:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74735ad2f127a4c62d4e5c4d24dd1af76e5b76a3","subject":"Use proper galera port in configuration","status":"MERGED","created":"2023-07-31 12:25:14.000000000","updated":"2023-08-14 10:54:51.000000000","submitted":"2023-08-14 10:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890089-osa/db_port","meta_rev_id":"f6244681b93ca66bc0fa9d5be90f58524d7d3adc","_number":890089,"virtual_id_number":890089,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-14 10:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-08 13:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-14 08:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-14 08:46:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 15:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 13:38:01.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-08-14 08:46:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0062e179b1809d5958657d8c3dd25da6c30719d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-31 12:25:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a61c70a53038c1667ebd3d9b108b71fb687d0d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:15:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f0dcfaa1e4b41638436e6d826599cb8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6baa187b69054f88ae840a6feb16549c : SUCCESS in 17m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88cd34a49477454dab2a1a1a69f99d21 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e4df809510240f38810c3d9e946ab9a : SUCCESS in 1h 24m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/58d24e12570e40eea6b11988d7401632 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e20a1abcc4e46b7bb0c86d30d852cf1 : SUCCESS in 1h 44m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e45f3ce312994dc1a6fd1bcf66ae91f8 : SUCCESS in 1h 40m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3b2aff1a07470d97628a277b5d1a88 : SUCCESS in 1h 23m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01fda7e5e56c41f9809fbecf1a55b110 : SUCCESS in 2h 43m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a27e24014d194803963814add1021f63 : FAILURE in 1h 30m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ce9362e76954203bfb65033ff859ebc : SUCCESS in 1h 02m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22f5b713e553445380883ac874ceffd7 : SUCCESS in 59m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a197c6ddd0194ba2a931deb2e8a05be8 : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41447cea87bb4ad4be28882280a8b16b : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9907f696e224125901d033278dad8ff : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/94c6934e23a64e3f935e8203e1e1bf61 : SUCCESS in 2h 18m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/4467ecef9bf941f9a47ad840f67ed887 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e2efebe08f04795adc2b9b31c142e1f : SUCCESS in 2h 16m 40s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e4ddbc0844f4ac38430ca7cfcd6ba5b : SUCCESS in 2h 07m 26s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b89f8e102324748a6f260a5e48b147e : SUCCESS in 2h 01m 29s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f97ed834e3249debff444e64180c4ba : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e3d27701f875dec184d9addd617a95e553791d03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-07 07:05:15.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4b170d1d0403a30687dd172b76aeabcb75d924e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 09:48:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7c01cb2ced4242b5c1f0013a77ed39\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/586356ed864c4e009028679f00327364 : SUCCESS in 13m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57acfebafb084bb5b45055e43cb3d0b3 : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/22013fdd15de445080a3376703f3f2df : SUCCESS in 1h 35m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/700033ea2fd64b25b906772f65ece8ad : SUCCESS in 1h 36m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3feb92570074e048260903ac60aec73 : SUCCESS in 1h 38m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd12bf682cf34a838ee7950b20425cf7 : SUCCESS in 1h 44m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3dcfe0a423f49098551df733d3a0b1e : SUCCESS in 1h 44m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8fb25fcd1d64709be1e2b43fc60b37b : SUCCESS in 2h 33m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78dcf29deeb448fcba69c84b9ad305a1 : SUCCESS in 2h 28m 45s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cbd9a45b1b674ede9f53acb308d2a8e3 : SUCCESS in 58m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecdad7934ee04309a301161e8c109d76 : SUCCESS in 57m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289ee566dd024131b10fef0aa447408f : SUCCESS in 5m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59ece951e3c1400489a9d0f6b299e15a : SUCCESS in 1h 54m 07s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff305d6105d4324ac83d2f8520bec32 : SUCCESS in 1h 31m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea8d3ba5aa414478b56bda6fbce53626 : SUCCESS in 2h 05m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e93bda7fb18416ba0a1bc5c2319b4d8 : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf2193829c614f4fbacc52eb779b113d : SUCCESS in 2h 11m 33s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/068273d0eb4d471a8e7fc410bcb775d5 : SUCCESS in 1h 35m 00s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/122f10af6c0344239b6d8c7d9641ec2a : SUCCESS in 2h 00m 27s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56b1250a77bf4a898250a577f8ee5b25 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":2},{"id":"24a45f098fdc50a2d3156ed637ef3ba78153edbc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-08 13:38:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d61a9dc7ebec8b52d28993837dd809d3851299dc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-14 08:46:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c00139d2b7a10e1e4cad69c173a8ca4e1588900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 08:48:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a63b481bb82bba68ab101545b2f17458d778cbcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:53:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c2f9d5a10145e2b154a2c4ccee4523\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37327fde4dfe4d5bb32d07fb92aa17c5 : SUCCESS in 12m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bab3a9c0721f4a0ab3fd8642263b0d6c : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae1de4a9db1244099f9630422968f328 : SUCCESS in 1h 34m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d7475b450434229a9d9febe4b6f0883 : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eee4a3b817a849258cddf6d924986e81 : SUCCESS in 1h 26m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/879b734e8ec442b799f47c72cf3a662c : SUCCESS in 1h 49m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1a4cfa909d434cb1103e410fd6fe96 : SUCCESS in 4m 22s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69c9ae0ff8d74de0874791198541054d : SUCCESS in 1h 42m 22s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37fbd2c44f2e4cc6a5d143b7b90b72ec : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/77456694ec8e4ff4a1e0a188297722ac : SUCCESS in 1h 42m 18s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/60991be694ee4999ab5d6dd749634db1 : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17ecef53dac34073bad7b4def750d74c : SUCCESS in 1h 55m 10s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df7fbe93bb6745629171a39fdb66cce0 : SUCCESS in 1h 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"57422627813837f87a0fab1a50245a025112471b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:53:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f6244681b93ca66bc0fa9d5be90f58524d7d3adc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 10:54:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d34d2017e2f4e02bf4497315ba920ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c53ec8b42049e0818f7875eacdcd58 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b398f5f435cd96c6f08d01863cc3bfeb40748ec","revisions":{"ece8b6925897a47a8a308fc3bd66bdc3e0a0730a":{"kind":"REWORK","_number":1,"created":"2023-07-31 12:25:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/890089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/89/890089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/1"}}},"commit":{"parents":[{"commit":"45f823de0c11e595aee6ff54edb46c6bad451c08","subject":"Merge \"Configure OVN NB and SB DB Connection probes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/45f823de0c11e595aee6ff54edb46c6bad451c08"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 12:25:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 12:25:05.000000000","tz":120},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservuce\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nChange-Id: I74735ad2f127a4c62d4e5c4d24dd1af76e5b76a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ece8b6925897a47a8a308fc3bd66bdc3e0a0730a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ece8b6925897a47a8a308fc3bd66bdc3e0a0730a"}]},"branch":"refs/heads/master"},"2b398f5f435cd96c6f08d01863cc3bfeb40748ec":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-07 07:05:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/89/890089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/89/890089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/89/890089/2"}}},"commit":{"parents":[{"commit":"45f823de0c11e595aee6ff54edb46c6bad451c08","subject":"Merge \"Configure OVN NB and SB DB Connection probes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/45f823de0c11e595aee6ff54edb46c6bad451c08"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 12:25:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-07 07:05:15.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nChange-Id: I74735ad2f127a4c62d4e5c4d24dd1af76e5b76a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b398f5f435cd96c6f08d01863cc3bfeb40748ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/2b398f5f435cd96c6f08d01863cc3bfeb40748ec"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
