)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~893938","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ieb8d1640c2150c6d7ff58881f534ce407e205b8b","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-07 11:35:14.000000000","reason":"Change was abandoned"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2023-09-07 11:34:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16011\u003e replied on the change","reason_account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}}},"hashtags":[],"change_id":"Ieb8d1640c2150c6d7ff58881f534ce407e205b8b","subject":"Fix l3 agent group determination for vpnaas","status":"ABANDONED","created":"2023-09-06 15:44:17.000000000","updated":"2023-09-07 11:35:14.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d26be0d81ddc55050d08999bd6926905b232d49","_number":893938,"virtual_id_number":893938,"owner":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-06 18:15:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2023-09-07 06:18:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 18:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 06:16:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-09-07 06:18:10.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"76b1a6cf00315d02707fb9f48723170a1d1e11c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2023-09-06 15:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17ce8a8b7b26e8ef2381b03081de4402dc709b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 18:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2768633077174d9b82ea26bfa13f7772\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31322299790d4489b401bb4bc0bdce0a : SUCCESS in 12m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ca0a6b9091ba4c1aaff45964cd7e166b : SUCCESS in 1h 13m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d138aee5ade146e5bb310f60546cd826 : SUCCESS in 1h 24m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddf77bb2fc594210a2bdf7e649144f62 : SUCCESS in 1h 42m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9956f5e80d4b4dee80dfa8e91a3c1b61 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cb4c53d9447430dbf04e8bb692d3e59 : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1503df2b5acf44f98c8558f492af7861 : SUCCESS in 1h 36m 58s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7546f7c7f225430d983d9d2829d8fa40 : SUCCESS in 2h 23m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed43a1b82dde48b7b3644da487bf6a94 : SUCCESS in 2h 17m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aae3c92f71bd4816ba008bb4445755e4 : SUCCESS in 1h 03m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e044ad3338e4eb0909d5d3555a5d6e5 : SUCCESS in 58m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08946c61de04a38867ca541044c2571 : SUCCESS in 4m 26s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39b5db493a554ff28a548885318aa76a : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46e9d372efec4f8f863f1282492c47d3 : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d8a1876858340d0bef215e8e6d2b769 : SUCCESS in 1h 50m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd12f681e18942a7a9608ed1e7e0d69e : SUCCESS in 1h 39m 35s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/448175a547e049349b26cea98c010354 : SUCCESS in 1h 41m 30s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/26925be6b48f42beaf99098d47fdf2b4 : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70b40b8a60c644ecb5631896a37495b6 : SUCCESS in 1h 44m 42s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/967193d639d6468286d94dd990ded0a7 : SUCCESS in 1h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"83bce1d3efae8908b8a3e601a16bc948056b55e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-07 06:16:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"817a7bba5ef99c3d55ae5c2828e66066fa083e4f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-07 06:18:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dec25e78b59e30b855c01b78fe9d3d954795d53","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2023-09-07 11:34:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d26be0d81ddc55050d08999bd6926905b232d49","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2023-09-07 11:35:14.000000000","message":"Abandoned\n\nduplicate, but worse","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"145a748ea421b704bfa107628338eb1f02e6e650","revisions":{"145a748ea421b704bfa107628338eb1f02e6e650":{"kind":"REWORK","_number":1,"created":"2023-09-06 15:44:17.000000000","uploader":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"ref":"refs/changes/38/893938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/38/893938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/38/893938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/38/893938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/38/893938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/38/893938/1"}}},"commit":{"parents":[{"commit":"4abf3e9383b26350a2e389162d9c1e14a7c2f33c","subject":"Merge \"Retry applying OVN connection settings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/4abf3e9383b26350a2e389162d9c1e14a7c2f33c"}]}],"author":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2023-09-06 15:38:09.000000000","tz":-300},"committer":{"name":"James Denton","email":"james.denton@rackspace.com","date":"2023-09-06 15:44:08.000000000","tz":-300},"subject":"Fix l3 agent group determination for vpnaas","message":"Fix l3 agent group determination for vpnaas\n\nThis patch changes the logic to determine l3 agent group\nmembership when deploying vpnaas where the neutron-l3-agent\nmight reside on network nodes not also the controller/deploy\nnode.\n\nChange-Id: Ieb8d1640c2150c6d7ff58881f534ce407e205b8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/145a748ea421b704bfa107628338eb1f02e6e650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/145a748ea421b704bfa107628338eb1f02e6e650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
