)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~893952","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2Fzed~I34256e4ad22169ae6907a3c40270cb714cf33466","project":"openstack/openstack-ansible-os_neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-12 21:33:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34256e4ad22169ae6907a3c40270cb714cf33466","subject":"Check length of network_mappings","status":"MERGED","created":"2023-09-07 11:25:45.000000000","updated":"2023-10-12 21:34:44.000000000","submitted":"2023-10-12 21:33:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893952","cherry_pick_of_change":893924,"cherry_pick_of_patch_set":1,"meta_rev_id":"e9f0186e02ea08b1f5ab319d3eb47d0892200636","_number":893952,"virtual_id_number":893952,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 21:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-10-12 07:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 19:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-12 19:22:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 11:25:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2023-09-07 13:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 19:22:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"95aaa4d710b5be824b6971b1ec2df667189a0e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-07 11:25:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aeff55bbb38621babb1b25282b563b0ebf505fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:52:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f8c3fdd23704f8c94b34ead49600cc3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ec51a088ed432586144e8e44e080a9 : SUCCESS in 15m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/603a6ff8af8540c49ef72f05dde24b22 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d93e005ce77444589cadac3f4d28563 : FAILURE in 1h 09m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53c096e373e74b88a42e43f7ee82bbd8 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c0b318131e84609b7101bf94dd943d1 : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0f5b5a967934cbf97bc0bdb2bbf1466 : SUCCESS in 1h 14m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1964f68949724dbf8b92345a49da1dea : SUCCESS in 1h 58m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83ef892e34b420ba9d41efb8d7d38a1 : SUCCESS in 2h 00m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d650945a14cc4ae6a1ac46c8189f9a0f : FAILURE in 39m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/023bc83961874c36b8154f7dfbcfb972 : SUCCESS in 58m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3c090f41bb4c3f8648357ab81b09ba : SUCCESS in 4m 33s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f182b1a6dd244f992c435c57a905961 : SUCCESS in 2h 19m 44s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9f176b4f29b4da9af4bf72454af8f61 : SUCCESS in 1h 29m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba1503d7acba44abb1707ac2a140e02d : FAILURE in 1h 33m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a3a364e4faa4cfb829795286f5688ac : FAILURE in 52m 51s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc374887bf844d59e3300c93631c461 : FAILURE in 1h 39m 47s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bef33b0cdac04135968f548d2b4eb204 : FAILURE in 58m 30s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d6d0cb98d64149ae349d60e1fa8f80 : SUCCESS in 1h 54m 16s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/847dd6c44ee4487587ee157acaa63fd0 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"57b16bfc402283e0f32e235b086ba79fc8151814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-12 07:04:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d8037fb414186e0fa620c72bd95dcd67ee695b73","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-12 07:04:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eb98529f77619c82c66d5a9c81adb3f655f41530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 09:46:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2529dcfc15b4545a21d94fdd2d1734f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c4e808fd5294c66a72144d807c1be47 : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/12e5457f1e734380b6f273b8a43c1241 : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/228bfa880e8f45ffa88c8ea88b425732 : SUCCESS in 1h 21m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6ac142640ab447a9808aec67629812fc : SUCCESS in 1h 22m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/390ff26382514132a360e590a05e150f : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2c6f0f2debf4944a66c769cd73fae52 : SUCCESS in 1h 14m 35s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24e98e04f7084d28ac903a2892dd414e : FAILURE in 9m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf5c3bae1c8d4dd0aa6b5deb7a1f411f : SUCCESS in 1h 46m 22s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/717428b6c0244fdea818dcf8f575fe8d : FAILURE in 34m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63e15279070b42479904b2e79520e3db : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd656f27d8a4c82a08ee47d76649b56 : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ce248a245e24881a922dab815e13c18 : SUCCESS in 2h 29m 36s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6103a2abd1024df3905e7f5046550a76 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fd86babc3894810ac3f805eda31bdc0 : SUCCESS in 1h 57m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/39a76b72555b4e2e8696f09a5b4513a9 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0d1f81ab7fa4c28b8ddb190ec60c761 : SUCCESS in 1h 57m 20s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b93ef4e702194c8d9f211f0d56e89a9b : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ee59d8b2564de386ef1f2e4568bed2 : SUCCESS in 2h 01m 26s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4d5824a3e84560b7919c9684f40c62 : SUCCESS in 1h 21m 10s","accounts_in_message":[],"_revision_number":2},{"id":"09a22313b0a3a67cdd147d647f0efaff316ec7b0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-12 09:56:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2b298e93c66c238a70433349f955cca583318a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 12:08:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfea5794110a42f186e500e7e4297093\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8acd475edeab410abb5b20e6445d9a9a : SUCCESS in 10m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6d100a439c584ffbb012cdb9ca1459b4 : SUCCESS in 1h 04m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3cae97f9bf449dda00b0914cbcca91f : SUCCESS in 1h 22m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/76e0b1284b474610866d0119d4f5a083 : SUCCESS in 1h 11m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87e3026e6bde437ba7a11d2eccc84580 : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a38c6a1c8a3048a4853885ecf6679b2b : SUCCESS in 1h 18m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0201d663d4a14243b9c0afb47ecffa03 : SUCCESS in 2h 05m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e17eca44fb401d8af2405c542d2c6a : SUCCESS in 1h 42m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4dbea2a031604387a3c28f1a5b1e5e71 : FAILURE in 41m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf05890446744aab80f470ac99984d88 : SUCCESS in 53m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1203caa118c42b7ae4c4f8bdfb0f9c1 : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81dd8fafae5b4efe87bebad9903ad155 : SUCCESS in 1h 49m 48s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62ca091e88784e6f875d72246f39ade1 : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/962a93145e7b4f8a9ce8cbc005b95868 : SUCCESS in 1h 50m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cf8e85874b745fdb5f2e0c96e50b052 : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd8b422270f44b4eaf52012357844f9b : SUCCESS in 1h 50m 46s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c526648e2efa4a9580410459a8cd1864 : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cc27ec9413649c98f37a3a6c5460a65 : SUCCESS in 1h 55m 38s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fabd1802b4db481bb8867ff707c0eb74 : SUCCESS in 1h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2b6d2f89cf8ff7e1216913d1cf30fea41a3c3cd6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-12 19:22:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68ff8aa1aa958c7750732dda9ce81c61fc23a246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 19:24:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b175e62cbb17c8b602eb9b7a531be4ccc16dab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 21:33:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be484d7a54244a8b1a5ca54fecf23bd\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/982eeb22c2cd4de4a68e80f12bcf7eea : SUCCESS in 12m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1427fa68413d445b9fc9d2bcc59502e8 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad76b471b051488da69d555b1511950d : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98f95aabaa414d0991756fb21a9814b2 : SUCCESS in 1h 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7720c91496f4d3fa1ad01f884dab3d9 : SUCCESS in 4m 57s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d3896ab94b549e488947660d353d5ef : SUCCESS in 1h 54m 07s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4ac2d4debb849159c9559d01660f768 : SUCCESS in 1h 20m 39s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bef6fcc6c524dcaa024d17357b88d3b : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/acdae5cd28e14c40a95d12155c5a8469 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19ab8dfb632b482a9b9ca95312f01135 : SUCCESS in 2h 03m 19s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ea349fc0cfa4322a5e7038902697f49 : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b80c404d2a76c2547add2b146bfcf9756e2a4d61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 21:33:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9f0186e02ea08b1f5ab319d3eb47d0892200636","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 21:34:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ec50551a2954f4287a8a8c95168e71c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec7af1a11fb442d88500a64e0dd793a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d6fd0887d76ce8188ace2f276123c6893e10e7c","revisions":{"155c598afae32ce65f14a14fb18a13cc3e10cdd9":{"kind":"REWORK","_number":1,"created":"2023-09-07 11:25:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/893952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/52/893952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/1"}}},"commit":{"parents":[{"commit":"cc9e79a206f223a3f5ce477b69349b7f85e1142d","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/cc9e79a206f223a3f5ce477b69349b7f85e1142d"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2023-09-06 13:53:08.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-09-07 11:25:45.000000000","tz":0},"subject":"Check length of network_mappings","message":"Check length of network_mappings\n\nThe OVS bridge creation logic for OVN deployments may fail\nwhen the provider bridge has not been defined. This patch uses\nlogic that exists in the OVS deployment scenario to check the\nlength of neutron_provider_networks.network_mappings to ensure\na value has been set before attempting to create the bridge.\n\nChange-Id: I34256e4ad22169ae6907a3c40270cb714cf33466\n(cherry picked from commit 7c0ac931b0accd02f4c03b994dcaaa65a2b901ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/155c598afae32ce65f14a14fb18a13cc3e10cdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/155c598afae32ce65f14a14fb18a13cc3e10cdd9"}]},"branch":"refs/heads/stable/zed"},"7d6fd0887d76ce8188ace2f276123c6893e10e7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-12 07:04:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/893952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/52/893952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/52/893952/2"}}},"commit":{"parents":[{"commit":"36ae7cc97c097d7dd9d6bc0b5b0fa444af4b35e8","subject":"Fix typo for  vpnaas_custom_config distribution","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/36ae7cc97c097d7dd9d6bc0b5b0fa444af4b35e8"}]}],"author":{"name":"James Denton","email":"james.denton@outlook.com","date":"2023-09-06 13:53:08.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-12 07:04:30.000000000","tz":0},"subject":"Check length of network_mappings","message":"Check length of network_mappings\n\nThe OVS bridge creation logic for OVN deployments may fail\nwhen the provider bridge has not been defined. This patch uses\nlogic that exists in the OVS deployment scenario to check the\nlength of neutron_provider_networks.network_mappings to ensure\na value has been set before attempting to create the bridge.\n\nChange-Id: I34256e4ad22169ae6907a3c40270cb714cf33466\n(cherry picked from commit 7c0ac931b0accd02f4c03b994dcaaa65a2b901ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7d6fd0887d76ce8188ace2f276123c6893e10e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7d6fd0887d76ce8188ace2f276123c6893e10e7c"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
