)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~894768","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I8d61ad4cca3ce7d4ecc09c69e04396ac2154833b","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"bugfix_condition_task_copy_vpnaas_custom_config_files","attention_set":{},"removed_from_attention_set":{"34933":{"account":{"_account_id":34933,"name":"shahab taee","display_name":"shahab taee","email":"shahab.taee67@gmail.com","username":"shahabtaee"},"last_update":"2023-09-29 07:36:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8d61ad4cca3ce7d4ecc09c69e04396ac2154833b","subject":"replace group_name to group_names","status":"ABANDONED","created":"2023-09-12 14:22:49.000000000","updated":"2023-09-29 07:36:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"460e214c16e85c32db7a39a9cbbe342b432c811c","_number":894768,"virtual_id_number":894768,"owner":{"_account_id":34933,"name":"shahab taee","display_name":"shahab taee","email":"shahab.taee67@gmail.com","username":"shahabtaee"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 14:38:51.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-09-12 16:56:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd6a07c03034c90ebf07357d2384b39d6572173c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34933,"name":"shahab taee","display_name":"shahab taee","email":"shahab.taee67@gmail.com","username":"shahabtaee"},"date":"2023-09-12 14:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d37873e065a7704a20cc796e3c3d962030fd807","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-12 14:38:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62ff71b636dd37bac40d3d8ee4eb0627fd9a5306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80774db3a9a043baaab778108ea46979\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/056f22ac5668463da97d07b14ec991fd : SUCCESS in 13m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff369f605a004b108d4ab60826041c5a : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a15325e78ebf481b86942493b9ae44f4 : SUCCESS in 1h 45m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf1986d0ae384ad885639f08848c13a3 : SUCCESS in 1h 48m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dc88d73e450409aa9f66abb7ba34b6f : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/61c2dffa7e9d4eb18afba91e1944e4e6 : SUCCESS in 1h 49m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cbaa29ff0574f0cbd4267e9fd507deb : SUCCESS in 1h 45m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91580a1518744ee2b873837ee83e0c9d : SUCCESS in 2h 26m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57931fb2ab8144b3a319c8835f16c491 : SUCCESS in 1h 50m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71a98298604a4e19910aa70376198100 : SUCCESS in 55m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9c1a2f9996487092df8ad7a9a74a4b : SUCCESS in 57m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86aaca0a1cf7416f8d9a5f31f919175e : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f18c3f917f3c4a0fab0547cb1a494ea9 : SUCCESS in 1h 53m 39s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71c1ee41652146f48322995fec0f5e54 : SUCCESS in 1h 34m 29s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e8b2786fdbf4ce282bdfa34abe475c1 : SUCCESS in 2h 01m 15s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d9d58fbf4e454987c7b0d92f82be08 : SUCCESS in 1h 36m 28s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b803622b7066465a8b002fe28bfc5251 : SUCCESS in 2h 12m 18s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/282ee55cfc984607bfa3714863d754e6 : SUCCESS in 1h 33m 50s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09e2a337dd9448e494a3b7accbd3a37f : SUCCESS in 2h 06m 14s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea5a45efd56f4dc9a6edcd051f9f0af8 : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"83f61d1518d1fc3aca235b21898edf93d35f4f0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 07:36:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"460e214c16e85c32db7a39a9cbbe342b432c811c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 07:36:55.000000000","message":"Abandoned\n\nfix has been already merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"114d31d6c550d8485ac2892f2360717cc28b021d","revisions":{"9ad71c46f3f9b065a99fac7308296eb0bd9d3da9":{"kind":"REWORK","_number":1,"created":"2023-09-12 14:22:49.000000000","uploader":{"_account_id":34933,"name":"shahab taee","display_name":"shahab taee","email":"shahab.taee67@gmail.com","username":"shahabtaee"},"ref":"refs/changes/68/894768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/894768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/1"}}},"commit":{"parents":[{"commit":"7c0ac931b0accd02f4c03b994dcaaa65a2b901ae","subject":"Check length of network_mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7c0ac931b0accd02f4c03b994dcaaa65a2b901ae"}]}],"author":{"name":"shahab taee","email":"shahab.taee67@gmail.com","date":"2023-09-12 14:20:26.000000000","tz":210},"committer":{"name":"shahab taee","email":"shahab.taee67@gmail.com","date":"2023-09-12 14:20:26.000000000","tz":210},"subject":"replace group_name to group_names","message":"replace group_name to group_names\n\nChange-Id: I8d61ad4cca3ce7d4ecc09c69e04396ac2154833b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9ad71c46f3f9b065a99fac7308296eb0bd9d3da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/9ad71c46f3f9b065a99fac7308296eb0bd9d3da9"}]},"branch":"refs/heads/master"},"114d31d6c550d8485ac2892f2360717cc28b021d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-29 07:36:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/68/894768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/68/894768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/68/894768/2"}}},"commit":{"parents":[{"commit":"a190ae4f61cce0da08028f96c959fd3c26e2d700","subject":"Fix conditional for non-OVN deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a190ae4f61cce0da08028f96c959fd3c26e2d700"}]}],"author":{"name":"shahab taee","email":"shahab.taee67@gmail.com","date":"2023-09-12 14:20:26.000000000","tz":210},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-09-29 07:36:42.000000000","tz":0},"subject":"replace group_name to group_names","message":"replace group_name to group_names\n\nChange-Id: I8d61ad4cca3ce7d4ecc09c69e04396ac2154833b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/114d31d6c550d8485ac2892f2360717cc28b021d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/114d31d6c550d8485ac2892f2360717cc28b021d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
