)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~897461","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Iaaf87b6d5400fe88c7edf86995ea9ba891866678","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa/deprecate-odl","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-11 11:38:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaaf87b6d5400fe88c7edf86995ea9ba891866678","subject":"Deprecate OpenDaylight support","status":"MERGED","created":"2023-10-05 14:46:20.000000000","updated":"2023-10-11 11:39:48.000000000","submitted":"2023-10-11 11:38:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897461-osa/deprecate-odl","meta_rev_id":"0778610f6d1fd95d1dca7ff5e6d4d9268cc94854","_number":897461,"virtual_id_number":897461,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 11:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2023-10-10 21:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-09 09:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 07:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 17:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-09 09:37:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-10 21:47:17.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"c6012443682339d9f8062ec8ec7bc2bdbddd9c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-05 14:46:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45d5adf431fb70264a5596f4544b337d6938857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-05 14:48:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"499b3f98b09362b66b0db95e4c5f0a82ff7fc4cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:46:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e968e092424436b59a45390cb06baa\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8e6e5a65f54a75865b5c1b4e1091b0 : SUCCESS in 13m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5283c7a8c9cb4d0b924b257a07924b2a : SUCCESS in 1h 30m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cfe80e567b04189a92cdec249c01cf2 : SUCCESS in 1h 35m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/690eb393b7dd49da930ff3c7bde18171 : SUCCESS in 1h 51m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c34bfe6f47624b8bb3207a809e288a87 : SUCCESS in 2h 06m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/310b6047135c494d88372a046aba4c59 : SUCCESS in 1h 44m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/413d02fccf844304960650a792bcdbec : SUCCESS in 1h 37m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53903b0fe798453eb7c84796acddf382 : SUCCESS in 2h 41m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a3f32501274aaf8b9611e2dd593105 : SUCCESS in 2h 38m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc663653347148bf84ed93c12a4f48f3 : SUCCESS in 2h 50m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67e21df09fca44b4a566032769526f28 : SUCCESS in 2h 34m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6221782015c45629510637c2b90c246 : SUCCESS in 1h 04m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/241fd7f4d3a148929f0d75ced5c50ed5 : SUCCESS in 53m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51dc14de3b70415f94a1b6ebfea8c6cf : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/444aba38a60c4c63ac879457846432fc : SUCCESS in 3m 31s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d810aaf068d4ad4ba23aa82e4032724 : SUCCESS in 2h 07m 48s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf68941a5fcd4f5388196e3bcc7f22d4 : SUCCESS in 1h 44m 27s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0919e6a1d960485aa652808ac29956fe : SUCCESS in 2h 19m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/f177e802717549f89463ad461dd1aa1d : SUCCESS in 1h 49m 31s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5db655ec069146c88d622fa05d7a1605 : SUCCESS in 2h 30m 50s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ab98a647dcd40db806f3d19714d44ed : SUCCESS in 1h 45m 04s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9ec6237911f492e85d1f674af0379dc : SUCCESS in 2h 15m 22s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/505aefa14189439b89e474c3d5923e0d : SUCCESS in 1h 36m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f4953317fe5b8f8ceaacf900cf1ccefc369eb3b5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-09 09:37:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"391e5110061f9cdaabac318d7b25b21e152098d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 18:31:19.000000000","message":"Topic set to osa/deprecate-odl","accounts_in_message":[],"_revision_number":2},{"id":"566069c40574e2ce01b65f6ae56d0f592d5d15f4","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-10-10 21:47:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aaab98770d9ca0732d7921ebd71429d095936412","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-11 07:32:49.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"40089b74e74f976e61d013dba61ec84952c792b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:35:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00582fe7a0aa65d96c3a3bb62c0acd86ce140fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:38:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff38d87f1ae48eb82ae9870dd2a0789\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/922e5864ef3d4548823fa87c67585ab0 : SUCCESS in 10m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f1d2d43d09f04d178c293abe9bfdad4a : SUCCESS in 1h 46m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/42d49fd5d5c44a83b76701d83e1e3266 : SUCCESS in 1h 51m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bb8cef689e544b6afe0f6748cae0fae : SUCCESS in 1h 46m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/557af2e49e734252a3bf1a36c9658370 : SUCCESS in 1h 45m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b1d9f1409cb442eb7c1495319d73fb7 : SUCCESS in 1h 40m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f812d1fba5184510a7a4308d971f8479 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d96ee7834a1d4ebc945f1e4d584b468a : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45cc0c2370140d9b5cb9ce961c17e0f : SUCCESS in 1h 56m 57s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4356f5d18b74fa4a69aa12ac9efb589 : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/d92c2c4045124adbb377837ebf53b87b : SUCCESS in 1h 44m 18s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/955ba1467be24af5828a46548d238d96 : SUCCESS in 1h 39m 52s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c0d39f20fb2473390da6537674265f6 : SUCCESS in 2h 20m 44s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f44072174146467ba331ea6d99b7069a : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3747fb9d6514eb1cff810724cd9d8c1ce6ccb2e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:38:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0778610f6d1fd95d1dca7ff5e6d4d9268cc94854","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe21212fd784811b45fc3f10c859307\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f96200fadb547bcb226aac80d847491 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59697ba1c54f96c409cfbd8a3f4178868eb0f0e9","revisions":{"f1d1187d3736dcb1919bd5a0425fe01f46474428":{"kind":"REWORK","_number":1,"created":"2023-10-05 14:46:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/897461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/61/897461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/1"}}},"commit":{"parents":[{"commit":"a190ae4f61cce0da08028f96c959fd3c26e2d700","subject":"Fix conditional for non-OVN deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a190ae4f61cce0da08028f96c959fd3c26e2d700"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-05 14:46:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-05 14:46:11.000000000","tz":120},"subject":"Deprecate OpenDaylight support","message":"Deprecate OpenDaylight support\n\nChange-Id: Iaaf87b6d5400fe88c7edf86995ea9ba891866678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1d1187d3736dcb1919bd5a0425fe01f46474428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f1d1187d3736dcb1919bd5a0425fe01f46474428"}]},"branch":"refs/heads/master"},"59697ba1c54f96c409cfbd8a3f4178868eb0f0e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-05 14:48:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/61/897461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/61/897461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/61/897461/2"}}},"commit":{"parents":[{"commit":"a190ae4f61cce0da08028f96c959fd3c26e2d700","subject":"Fix conditional for non-OVN deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a190ae4f61cce0da08028f96c959fd3c26e2d700"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-05 14:46:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-05 14:48:53.000000000","tz":0},"subject":"Deprecate OpenDaylight support","message":"Deprecate OpenDaylight support\n\nOpenDaylight support has been deprecated by Neutron team in 2023.2 [1]. We remove support from\nour code to address that decision.\n\n[1] https://opendev.org/openstack/networking-odl/commit/517df91c9ef84b4e462e6cc9d70b7e14ad5a511c\n\nChange-Id: Iaaf87b6d5400fe88c7edf86995ea9ba891866678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/59697ba1c54f96c409cfbd8a3f4178868eb0f0e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/59697ba1c54f96c409cfbd8a3f4178868eb0f0e9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
