)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~897656","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I90454ba50840f7cb900586a7b870161a0f4adc01","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"osa-bookworm","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-11 08:09:09.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I90454ba50840f7cb900586a7b870161a0f4adc01","subject":"Workaround ovs bug that resets hostname with add command","status":"MERGED","created":"2023-10-09 07:59:32.000000000","updated":"2026-09-09 14:29:27.000000000","submitted":"2023-10-11 14:37:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897656-osa-bookworm","meta_rev_id":"b68df1d49c1a0de98155f7d5a0494a52becf4764","_number":897656,"virtual_id_number":897656,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 14:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-10 08:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-10 07:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-10 08:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-09 10:06:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-10-09 10:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 07:52:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-10 08:46:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"c7976d8db2011101a0823865cabfc662073475e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 07:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82b7cb5c24fe8322867233973eb8465acb7e9226","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-09 10:06:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0f5b02bd20254b1ffdbb2e2d133565d6d3cf958","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 10:25:41.000000000","message":"Topic set to osa-bookworm","accounts_in_message":[],"_revision_number":1},{"id":"68ee44e715d5fec40b1c761be3e9f1515368ea9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:56:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7a65be47f04952a76522d1b293cc20\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f2eb167ae5f450c86c34050126ff774 : SUCCESS in 13m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57e821fc48194047b161d2e9668415fa : SUCCESS in 1h 42m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1325fb500c19423991ead542d900ade1 : SUCCESS in 1h 51m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24d515b72d4b4fb4b1668b0def59f431 : SUCCESS in 1h 59m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a25498834fe847c6a45afca4ce85ac6a : SUCCESS in 1h 40m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d86abafbd60404c9247a2a4ad98e6a9 : SUCCESS in 1h 54m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89f1f8fcc45446598696ad4c13cab3f5 : SUCCESS in 1h 51m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2e5db863bc4f4e799dd611663ee8e19d : SUCCESS in 2h 48m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/472584740d2d4b73bec2849415b389cf : SUCCESS in 2h 24m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f019c39102d40a59fce332cddb2874d : SUCCESS in 2h 30m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/caed6dd193ba4e44b7712c3fa64d99ba : SUCCESS in 2h 04m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51997874eda143cf814266cdf73342df : SUCCESS in 1h 04m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a9ff003ccd64fcfbac171ef1ad5d2f9 : SUCCESS in 1h 04m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6f9d3adc48401f83dd19abf29c1173 : SUCCESS in 6m 25s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d4e301391554a0b874e1a82e1ccbac7 : SUCCESS in 2h 18m 12s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12dfdd6ac7c4449f9e4b558e87bfdf36 : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/56cb213a534547d7ab842f4a6a89a686 : SUCCESS in 2h 17m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/997f30076732445b81fd21fd2c3e34bb : SUCCESS in 2h 05m 39s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5095bdbae68e45fd822b57f406325a19 : SUCCESS in 2h 20m 44s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/df164f69316849b6bd83652943091218 : SUCCESS in 1h 51m 07s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3451eccb96944b3a87f28461d3db60dc : SUCCESS in 2h 13m 24s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc774baaa3b4805bf41250fae0dd0b3 : SUCCESS in 1h 45m 34s","accounts_in_message":[],"_revision_number":1},{"id":"91d6df193b938776b077646c9044dd491d0e0b2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-09 15:26:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ae3641d3b2136661149c46574c4fbbb44415e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 18:05:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c402d76afd42f09535e8d10c3f397b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63cf82d7aefa46e4b447fa0551506576 : SUCCESS in 13m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/28d0944ca5764061a69d40ae743421fb : SUCCESS in 1h 41m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f7248de48c0448781608c85f33e1394 : SUCCESS in 1h 46m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be9bbe1784784654a0ad8ff2a082810f : SUCCESS in 2h 03m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2695a86978af4e75b8de9f4286e2020f : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24854bfca48847b0b84c6498f7f2196a : SUCCESS in 1h 46m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/deb27d5ef007442e89513f5d6b0ca0de : SUCCESS in 2h 02m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd1cc0020ad74b47aa1b56287664b4ee : SUCCESS in 2h 35m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b51da955d3aa4e3f932bb33f225cfde5 : SUCCESS in 1h 51m 00s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c66aa5a1751e4dfb8b53fa2dbfaab77a : SUCCESS in 2h 32m 41s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26f36b3d370b4122b1006014ea2a2ac0 : SUCCESS in 2h 22m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84048faec66c4dc893df555f7b43eaab : SUCCESS in 1h 07m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39ad5a1b739a4189afd5683dbad953dc : SUCCESS in 1h 00m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6abf7bb5f624c0a8ddcdfc5a3b03b98 : SUCCESS in 5m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc479a6601b04658bacea6c287033a20 : SUCCESS in 2h 08m 56s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17db4fa5a2f14ab2819cc3d3cbecd4e2 : SUCCESS in 1h 40m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/82115535bf30491b8dbc886df3e0dca7 : SUCCESS in 2h 24m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1bc7acbd0134092b0bd6c3913891117 : SUCCESS in 1h 51m 03s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a710eb0bddb6496e9c88e74b575a3668 : SUCCESS in 2h 31m 37s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9551fb548396407492dab3d12fc70d7c : SUCCESS in 1h 41m 00s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1750c76ccac4bccb1f021833d911b95 : SUCCESS in 1h 45m 14s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/756edc39d878405998c72767e26ceccd : SUCCESS in 1h 34m 19s","accounts_in_message":[],"_revision_number":2},{"id":"111a5310c16d266f9d1953e261df2c67797354f8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-10 07:52:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"339693067c14e304aa1f6685adbd902d0094b491","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-10 08:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"72a7bd39c9231c920fac82dbc0674433e808d97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 08:48:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bc49ad538699662f5c8005f8d4efbeea32b9e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 10:57:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ae7572d049040db9f40ee6daddd8a40\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dafc88c231b747a898e195797c656e6f : SUCCESS in 10m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e42a83e2a5954dcab98269c93562a38c : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4409ce8d44ce4d4494b9d6798d6aa18f : SUCCESS in 1h 47m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25eb0b31df754d92bf4607264127bc6e : SUCCESS in 1h 38m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73444339ed31400595a0c9c03f1518a4 : SUCCESS in 1h 53m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dbbae43dd6e400dbd865843d15129e3 : SUCCESS in 1h 28m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ef33d1bf06452ca884547a5418a14d : SUCCESS in 4m 53s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48f205e4235e4d4ebdefdd6646a1ec96 : SUCCESS in 2h 04m 18s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30ee311806a44120b823fe0911b09550 : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4ddc6d9065f4dcd9fde0e9511bf20cd : RETRY_LIMIT in 16m 24s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2369a5a9970f452ca39c9253e8c5bc98 : RETRY_LIMIT in 11m 50s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24d8dcf0293a4810963f59036c92c396 : SUCCESS in 2h 01m 38s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a2d56cc44446bab143c5c7fd926cc4 : SUCCESS in 1h 40m 42s","accounts_in_message":[],"_revision_number":2},{"id":"79711c35aa8332d4ff645696f6ce72a5838b0ce6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 14:21:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b8bf90f24c89105f7c6308a4443dcdc71e0761d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 14:43:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4373cd7543ce5f4ace149a8e7757d9762b816bc2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 17:30:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22dd2a3540dbb4fe758826f77256b78aab9617ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 17:34:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64f8aa8d404a433abd1ef1238361b0f3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cbe6eafb91a4b74b20f777285e882c6 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6f4c0fe6b29a4745a3a8df461f5276a2 : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6a04f9f48654565a5185017355020cb : SUCCESS in 1h 33m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48a6dd2986ed42fc92a977b2607e3769 : SUCCESS in 1h 27m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c658a978bbe244e4b0dc7cab6293e6bb : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bef7b8bc541c40ffb0b8cfaaf589cc3c : SUCCESS in 1h 47m 54s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d068ba6350543d3a9dab79c5dabcd96 : SUCCESS in 1h 33m 11s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a19b1bc53f84cf3b67a1665e05fffa0 : SUCCESS in 2h 40m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5a79f6837ca4e94bd36e650bed87e74 : SUCCESS in 2h 03m 10s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44e81d19b1c64df687fcbbb2802dc82d : TIMED_OUT in 3h 04m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/386fdc813ae3426aa36ced21bdf6b10a : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a48d7ae9ebc4ecdb85efa91bacc67a8 : SUCCESS in 1h 04m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90af8113976e42bbb2ba90af2dd10ea7 : SUCCESS in 59m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c1c6c1a8d84f40a6cba96d69c6245f : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb6cc79668ac419e9bbaf17c8132ad9e : SUCCESS in 2h 13m 55s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec3250240d9a409ab5c5431145c1c0cc : SUCCESS in 1h 20m 23s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/48cdb9614a2d402db1f100856e7757bb : SUCCESS in 2h 16m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/a46a288b4de24fb284354d8a84cf03c9 : SUCCESS in 1h 52m 12s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/95c463592dd74a7db4742d67178e2d9f : SUCCESS in 2h 20m 15s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e0f3d96e6b14a5bbf97c42a734ecd33 : SUCCESS in 1h 55m 45s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e172753efc74f6ea5aac8c91a934700 : SUCCESS in 2h 09m 08s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df0568334612470d84cfa788aed493b6 : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7d9d75142c74d9f53feeaf4dd9c78d5e16b2b130","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 08:09:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"62790078cdbbc1c358175d5329e4ce5d61c4a67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:26:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b80b6c83d644dda96ab53bfd4dcaa6d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fea32d91b08d4453b39b1bdbd82d35c3 : SUCCESS in 11m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4d8afef726034c19b9a8064e7c68969c : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f64c12f6f8c4b72ab1fb4a2e6994fe3 : SUCCESS in 1h 53m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f62c53d348a64808b7162cea906d4210 : SUCCESS in 1h 48m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12dc0e8d6959438e84dff5b48b00c0eb : SUCCESS in 1h 42m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21cd663704744f2385b633d8fdc4dbc8 : SUCCESS in 2h 03m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c2ef2eb59764e4a88e8c59e8836c98c : SUCCESS in 1h 49m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0bbc7f236b54c76bf14769e2718ab0f : SUCCESS in 2h 51m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff546f1a2add4ac293847e22c73a1281 : SUCCESS in 2h 25m 05s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9dae8a4562724a3d9ecb8d1e18f54a99 : SUCCESS in 2h 54m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1b4ddd52e3f4e029d3c1c60c0c69258 : SUCCESS in 2h 13m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dbaf242b2a12445fae4a19b41e6dca81 : SUCCESS in 1h 09m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63dcc0cc696e4e3a88fc2e76c7f1b739 : SUCCESS in 1h 05m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc0baf760284d96b639ac6a7df7251e : SUCCESS in 4m 43s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00bfa78fecce452796f406c3eb781431 : SUCCESS in 2h 19m 59s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2658f28ef73a4647b6dd7b8f386dea98 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/bba8bc8347db4ed79330c02ad3247a8d : SUCCESS in 2h 28m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/a83deb6681774c238692ebdb4451f0f9 : SUCCESS in 1h 44m 30s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e41e2dce0b364fc7a447d9d0da00ff37 : SUCCESS in 2h 32m 45s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd032603723d445dbef13a8ddb510dd8 : SUCCESS in 1h 57m 31s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f53d728360e4818acedacd42b4788dd : SUCCESS in 2h 03m 38s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3b42fcef2994461a2822964c4954d27 : SUCCESS in 1h 46m 54s","accounts_in_message":[],"_revision_number":2},{"id":"bfad372d3f70fc791e1c509faa56f5937fd460da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:28:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a396a0c38a3695ef61f73796f5ad1193d01c2fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:37:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b57c8a33f3442fba61076661f72191\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/587b7c3a5e3242968605a79177ed3718 : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/91024d4155884dcfb4286b8cc25c51ba : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50313865da074365838fb9f8c4e42661 : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b006469e0c2743fe80e57c6de787960a : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3fb77a4d89734109bf3b0f76a76c0904 : SUCCESS in 1h 49m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d24b6c1c0fa743c5bca16a28003ff8a3 : SUCCESS in 1h 44m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d563c84267bf4b0593fdd4107e7b4f62 : SUCCESS in 5m 37s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a363ed697d7c4e139ffaf785dbd9dcb0 : SUCCESS in 3h 02m 44s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d665b541173447de849d4c19a757b4ee : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3314698f8c9467f986e411a06511816 : SUCCESS in 1h 52m 37s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/03e2715b42f6469f85056f57b291c773 : SUCCESS in 1h 55m 27s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd2be95ecdab4298bdef2b4e1c0ad919 : SUCCESS in 2h 21m 59s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23b286faf9bb464da1e8a8e608f8bd03 : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":2},{"id":"28637913bac114d4de62d3d1ad9906b2316ac308","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:37:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff233595cd56944983e3496dbd38ef883b3583d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49af80a3c36425894318408e4a81d8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd09bb20e28440b688c80ee121d3be20 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"441f144188bafbad996215294b22e038d866faeb","tag":"autogenerated:gerrit:revert","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 14:28:21.000000000","message":"Created a revert of this change as I879a65bcbda57353eba98940a8bd8e066b54f544","accounts_in_message":[],"_revision_number":2},{"id":"b68df1d49c1a0de98155f7d5a0494a52becf4764","tag":"autogenerated:gerrit:revert","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 14:29:27.000000000","message":"Created a revert of this change as I879a65bcbda57353eba98940a8bd8e066b54f544","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be3d0f3a492af173e4adad431393e34d703d95e3","revisions":{"7b539df914fa0905f3e4a105cf714c5af613e8a6":{"kind":"REWORK","_number":1,"created":"2023-10-09 07:59:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/897656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/897656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/1"}}},"commit":{"parents":[{"commit":"a190ae4f61cce0da08028f96c959fd3c26e2d700","subject":"Fix conditional for non-OVN deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a190ae4f61cce0da08028f96c959fd3c26e2d700"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-09 07:59:07.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-09 07:59:07.000000000","tz":120},"subject":"Workaround ovs bug that resets hostname with add command","message":"Workaround ovs bug that resets hostname with add command\n\nIn Debian 12 OVS version to 3.1.0 is used that is affected\nby the bug [1]. Until that is fixed, we\u0027re masking ovs-record-hostname\nservice.\n\nWhile this was fixed be OVS version bump in Ubuntu and RHEL, it\u0027s still\nan issue for Debian 12.\n\n[1] https://bugs.launchpad.net/cloud-archive/+bug/2017757\n\nChange-Id: I90454ba50840f7cb900586a7b870161a0f4adc01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7b539df914fa0905f3e4a105cf714c5af613e8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/7b539df914fa0905f3e4a105cf714c5af613e8a6"}]},"branch":"refs/heads/master"},"be3d0f3a492af173e4adad431393e34d703d95e3":{"kind":"REWORK","_number":2,"created":"2023-10-09 15:26:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/897656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/56/897656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/56/897656/2"}}},"commit":{"parents":[{"commit":"a190ae4f61cce0da08028f96c959fd3c26e2d700","subject":"Fix conditional for non-OVN deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/a190ae4f61cce0da08028f96c959fd3c26e2d700"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-10-09 07:59:07.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-09 15:26:54.000000000","tz":0},"subject":"Workaround ovs bug that resets hostname with add command","message":"Workaround ovs bug that resets hostname with add command\n\nIn Debian 12 OVS version to 3.1.0 is used that is affected\nby the bug [1]. Until that is fixed, we\u0027re masking ovs-record-hostname\nservice.\n\nWhile this was fixed be OVS version bump in Ubuntu and RHEL, it\u0027s still\nan issue for Debian 12.\n\n[1] https://bugs.launchpad.net/cloud-archive/+bug/2017757\n\nChange-Id: I90454ba50840f7cb900586a7b870161a0f4adc01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/be3d0f3a492af173e4adad431393e34d703d95e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/be3d0f3a492af173e4adad431393e34d703d95e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
