)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~904572","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~Ic964329c06765176692f7b0c32f33ec46360a3fb","project":"openstack/openstack-ansible-os_neutron","branch":"master","topic":"bug/2002040","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-05 10:39:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic964329c06765176692f7b0c32f33ec46360a3fb","subject":"Add Availability Zone variables","status":"MERGED","created":"2024-01-03 14:03:11.000000000","updated":"2024-01-05 10:40:16.000000000","submitted":"2024-01-05 10:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904572-bug/2002040","meta_rev_id":"2793bfc9d39f07952f0d52d4f1f243e49e4cbc46","_number":904572,"virtual_id_number":904572,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 10:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-01-05 08:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-04 20:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-01-05 08:29:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-03 16:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 20:28:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-01-05 08:29:04.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d49cf85af7836c21553adb08aba9ec7218b780d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-03 14:03:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fdf3e6340a62ae617f3dc1705e3ea7b63d83021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-03 14:03:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3faf2041815be1a98d5bd5b8a5ac579cdf613c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 16:16:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a83d5b9bd9c45aa8522709791805ae9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/417026e383864c1d95f01b208e8709ee : SUCCESS in 11m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/21ea61f579ee4db2a74cb994b5556e4d : SUCCESS in 1h 13m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e119de092cfd4cb7acc3d0577e3fb984 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c02afa06acd5440c90f000eed829a5f2 : SUCCESS in 1h 10m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ac902f88acb4b0495e408423076aa2e : SUCCESS in 1h 20m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccbbe3180d3446cfb308070ddca8c42b : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91ff1d6560d64b2eaca95c0f16678bed : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b439fa7703a8490bb761549aa03a6351 : SUCCESS in 1h 15m 21s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22626a2065e24d73a77eaa375eb22405 : SUCCESS in 1h 06m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d3c7b10e629d407bbb872d105822fe1f : SUCCESS in 2h 00m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/808b3a1c8e3c4600b6d9597bf3107b8e : SUCCESS in 1h 42m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc1b2750d0cf4204947553b52b12e56b : SUCCESS in 1h 48m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9df94dd7f24414ba0b8cb958d7e469d : SUCCESS in 1h 53m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e02f3e3357144257b396e8e481ffe81c : SUCCESS in 54m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c15acaf79e03486cac30f64a0d93a4bc : SUCCESS in 1h 02m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1912de0c645e4e9d9c9e5e1fcd859aa7 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d0d44ad55e04b0284daf3c08ffa5a11 : SUCCESS in 3m 19s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7145d56b01824c86bd460be67c3afce4 : SUCCESS in 1h 39m 06s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3670c116a664268a87c3fb29092ac08 : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9045db019bf470c9a25d0897f3ea5c3 : SUCCESS in 1h 46m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/a70ec46da2e94263b254c6e3c5b24db5 : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b2c370c1f0845b48e18335a9f0b214a : SUCCESS in 2h 09m 00s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ebec3fe97f44d6e936140a24060cc46 : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceb70b00a2604c4496c02da8e4045f8a : SUCCESS in 1h 39m 34s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/997d78b82fea4165be2199ee124f681a : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"aba625ec190e6d08205568eee9470963a83fc573","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-04 15:23:52.000000000","message":"Topic set to bug/2002040","accounts_in_message":[],"_revision_number":2},{"id":"24b26337b5a96aec7730d97cd84b6a0a39b906da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-01-04 20:28:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"72e302673fd5ddd1941ef5caf03fda67594a6fe3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-01-05 08:29:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50915c569b94937ba359175242e9c519330e5345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 08:31:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1dbda28bd6ade25b8bf08b047ac11487f0e52c8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:39:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d7d2af406043d9a78cd579768799b2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbe7c09f54194c30a83a0413adb2b76d : SUCCESS in 10m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88644a644791404cb719e395590ac5db : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e0fd8dd08b8d43c99e0571e9dfcc98f7 : SUCCESS in 1h 10m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fadbc0646626464a8fd97c4cdf8e9afd : SUCCESS in 1h 30m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d49a83659bca4e83bd9e7bcc6d84a481 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d575be2b7bd74ebbaef8f5289e27a93b : SUCCESS in 59m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d4c6b19022d4bd78bf5c3a09fd2d782 : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77d05be109343bbaab274f9fc1b1197 : SUCCESS in 1h 00m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fae060348b44742bdab04d94edbec41 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e28ad9991c4a40a7ae069e084358ed7c : SUCCESS in 3m 21s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa757d30ef164f789802499c7bca22f6 : SUCCESS in 2h 06m 27s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41666a1351404bc9b188071e206bf098 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bac20b735ea429389e0b58d470901ae : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba749f596ceb46e89a54e45c3777424d : SUCCESS in 1h 26m 03s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d963fc54398464987cefde4633181e1 : SUCCESS in 1h 43m 11s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6316ffa1fd44711911294f96836031c : SUCCESS in 1h 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0468f1a63d2bbd3399e843980e77c6be7e98b67d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:39:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2793bfc9d39f07952f0d52d4f1f243e49e4cbc46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:40:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6558942634e14d59884c04ac3b008bca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bde3c2b2584bbca1e907dcef6d7d1a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70bb847605b3002c58802511eeec99193d0f9220","revisions":{"005079bd49035948b8c01b31ff0450156fd96c56":{"kind":"REWORK","_number":1,"created":"2024-01-03 14:03:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/904572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/72/904572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/1"}}},"commit":{"parents":[{"commit":"ef4d3278556c6e34f79b195cfb3e0d300f7184f7","subject":"Update VPNaaS package for RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ef4d3278556c6e34f79b195cfb3e0d300f7184f7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-03 14:02:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-03 14:02:36.000000000","tz":60},"subject":"Add Availability Zone variables","message":"Add Availability Zone variables\n\nAt the moment the only way to configure multi-AZ support in Neutron were\nconfig overrides, which work quite nicely with LXB/OVS scenarios. However,\nwith OVN changing configuration is not enough, and command that sets\nup OVN Gateway should provide extra CMS option.\n\nIn order to improve AZ support in Neutron role, we add couple of variables\nthat control behaviour and allow to perform required configuration without\nconfig overrides for OVS/LXB/OVN.\n\nCo-Authored-By: Danila Balagansky \u003cdbalagansky@me.com\u003e\nChange-Id: Ic964329c06765176692f7b0c32f33ec46360a3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/005079bd49035948b8c01b31ff0450156fd96c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/005079bd49035948b8c01b31ff0450156fd96c56"}]},"branch":"refs/heads/master"},"70bb847605b3002c58802511eeec99193d0f9220":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-03 14:03:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/904572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/72/904572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/72/904572/2"}}},"commit":{"parents":[{"commit":"ef4d3278556c6e34f79b195cfb3e0d300f7184f7","subject":"Update VPNaaS package for RHEL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/ef4d3278556c6e34f79b195cfb3e0d300f7184f7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-03 14:02:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-03 14:03:27.000000000","tz":60},"subject":"Add Availability Zone variables","message":"Add Availability Zone variables\n\nAt the moment the only way to configure multi-AZ support in Neutron were\nconfig overrides, which work quite nicely with LXB/OVS scenarios. However,\nwith OVN changing configuration is not enough, and command that sets\nup OVN Gateway should provide extra CMS option.\n\nIn order to improve AZ support in Neutron role, we add couple of variables\nthat control behaviour and allow to perform required configuration without\nconfig overrides for OVS/LXB/OVN.\n\nCo-Authored-By: Danila Balagansky \u003cdbalagansky@me.com\u003e\nCloses-Bug: #2002040\nChange-Id: Ic964329c06765176692f7b0c32f33ec46360a3fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/70bb847605b3002c58802511eeec99193d0f9220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/70bb847605b3002c58802511eeec99193d0f9220"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
