)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~927992","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I86655567810c61dbed0415afd2e7ff343f20c736","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-09-10 13:37:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86655567810c61dbed0415afd2e7ff343f20c736","subject":"Do not kill ipsec on L3 cleanup","status":"MERGED","created":"2024-09-04 14:22:30.000000000","updated":"2024-09-10 13:39:00.000000000","submitted":"2024-09-10 13:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927992","meta_rev_id":"07c3e5b50226cd7d0cab550b3a7d1c72d2f85bea","_number":927992,"virtual_id_number":927992,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 13:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-09-10 11:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-09 17:00:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-09-10 11:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2024-09-04 14:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 16:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 17:00:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-09-10 11:50:34.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"646c60e761b2afc8837ea6d2d25c20e3e0e944ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-04 14:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd431c0a4f8cdf733c70004a1826f680185fe6da","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-04 14:26:42.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"512f5337da521f91c931272248df00a4a307a62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 16:46:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86e6e9b047e148daab461199cf5b57c8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9c9b361e2840328ef9ffe221031c95 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/57118d70504643fbafabedb2c6d226cf : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6e276cf5d8245bf84c74d7695752053 : SUCCESS in 1h 15m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78e73fd7c45e49f8b83fb0a70e50afc9 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b5f8db7ae724adbb36f67d46526e81a : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ae4598455d84196a9784175a4519295 : SUCCESS in 1h 33m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7583d2cc8014f7a8ce5c33db633b96f : SUCCESS in 1h 17m 49s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e792251ae364f7c8c78daf1a0e81b4a : SUCCESS in 1h 20m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a81ff3163e54539bb38bd0c9aff8707 : FAILURE in 46m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b512a474fa14b02a77534fff65a18fa : SUCCESS in 1h 51m 55s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7dc38dc638664522a979c7b213917c66 : SUCCESS in 2h 09m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2318ed7dfcbd488aa0e7eeebb56d1117 : SUCCESS in 1h 38m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a8130b5cfa249e29d5f57c50ec0d7be : SUCCESS in 1h 02m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61867f994548468ab0f28a0d5d5d4d42 : SUCCESS in 1h 00m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb97eab193a462999c585518663a905 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/347c20066f9e44e1a1d1643799688b51 : SUCCESS in 1h 50m 16s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f7b1b0911054d0cb547d32bbc146a31 : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1111f95bd8b4f0cb6ed705aeab9b900 : FAILURE in 42m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/86aff28109914a4f8e94e71109b57c41 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/01e9e69b57c54d3f893abff260505868 : FAILURE in 43m 02s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36f5fa0d15624fe6957b36e61e4dba15 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52eb91d28e004cc4a78f489e26091db5 : SUCCESS in 1h 47m 12s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/911382c77b9043be86f5be2cd908c247 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7b3563206b9f09414954d94a717f9283f6dd0286","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-09 12:14:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf78c78677c062b18158e7e3bff35ec547503840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 14:25:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f22540e5728417b8887c52e83576f97\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d86d7a1053164024941f09fdf1b271e6 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e881e09b01d7490e9422d44deef6719f : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/85fd22f8e94c41c1ad300aaa9216d084 : SUCCESS in 1h 17m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28ec4b49d0fc41d3a566000a626c2bea : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc2131a36e5d4568a874b44e17554fcb : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d0be9a34b60468c900030470e84bb5b : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b8b33506cec4800beed59ad799f1823 : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f944550ed0fd48cba5bb562f29fef1e1 : SUCCESS in 1h 09m 16s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfac21de956f4fb1a5c74eeefb70adbd : SUCCESS in 1h 08m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97b68d98b38441a3b0e71ee612bedff7 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c75117d81b3f44258d454c6877674ee4 : SUCCESS in 1h 19m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0582e6331ff414db62cebe895535c19 : SUCCESS in 1h 13m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/050ef73db7bc4ce7b5101dcd50d271a7 : SUCCESS in 1h 44m 28s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c9fa73a325045b8bca2d253ecc36916 : SUCCESS in 1h 59m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d65946db3b24b6784ee661ec6c144b7 : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1e443447bdd7425aafec4c5d8cf0b496 : SUCCESS in 1h 04m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ab69da23114433586e27bb4c2d5c3b5 : SUCCESS in 58m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf18ec9dae0e47b2b6becbd81ae83a5f : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84ee05fc19944c5db28b98d67e3b3cc8 : SUCCESS in 1h 37m 55s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4c0a09a468546ab999733ae36bb46d4 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c56693d3aa54b90ac9c34ce1ab0cefc : FAILURE in 39m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/86e200e646814ce8b355d5a19553c717 : SUCCESS in 1h 22m 38s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e55657b677f4b13b1090c21abe37621 : FAILURE in 39m 59s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8b5532e9aa241a19876e25c35a8d951 : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8df7fe619b344dafba4775fa422022ff : SUCCESS in 2h 02m 31s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92265a826ad8482381afed13670ef3dd : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"72460c81f6fbfdb2f262cffa8ef1091d17dc79a1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-09 17:00:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19296d69c23514bc46877eda985839fd061cdfd8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-10 11:50:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80898517f9db041307301ae3b39fa53cfa820b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:52:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86f2678e431f9a5d54ea8a5d7e9cf2ba4e694435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:37:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50feae1d1a8c4a00969c1507d10b10ed\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3749a183e944651a52686de9cf7a298 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6fd1b89a6e3b4bab8c79e86329c33780 : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4e88a53ea3d4bfb9ed11aaf5a28c649 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f5512671e174d3ca793834220e71f51 : SUCCESS in 1h 14m 40s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3493737540b4bfe9fdb600eabbf369a : SUCCESS in 1h 11m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/981aa45bad0b4ac6bcbd6518da7eb1bb : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5db6ff6e014648cc98eca17d037b6d56 : SUCCESS in 1h 23m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37c27d826ec64df79506c5a164813306 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af0948c3eeac468fb24cc7585374a474 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4694ac74a3042389bbef793dc0fc678 : SUCCESS in 1h 16m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b40496dfca486fa9c985889e1a38c1 : SUCCESS in 3m 36s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b48ad2651f14b02a6b48a314ea63094 : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ce0b12d7afa4e299726966d6336e54b : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal_ovs-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca3db90e52f848db9be39f8c4ed4ae74 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_lxb_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/30913eb120fb4da68be9f074a0ed46d1 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1182ade022ab40ab89eb26e21f5b2249 : SUCCESS in 1h 39m 01s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df3ca5670dd542199e1218e58545596a : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3d0e852083b9db409a7b720c4de83dd66e9971a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"07c3e5b50226cd7d0cab550b3a7d1c72d2f85bea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:39:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013cdbfd60284773b99fde550e881047\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4978627732aa4de1913683e9b07282d3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3912bf3a5f04d90323ea821a7ae3a04a57b3fe33","revisions":{"3912bf3a5f04d90323ea821a7ae3a04a57b3fe33":{"kind":"REWORK","_number":1,"created":"2024-09-04 14:22:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/927992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/92/927992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/927992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/927992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/927992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/92/927992/1"}}},"commit":{"parents":[{"commit":"6e23144f4847cfbe01115985e90afc65181800ce","subject":"Use cgroupsv2 for L3 cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/6e23144f4847cfbe01115985e90afc65181800ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-04 14:22:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-04 14:22:21.000000000","tz":120},"subject":"Do not kill ipsec on L3 cleanup","message":"Do not kill ipsec on L3 cleanup\n\nIn case VPNaaS driver is enabled ipsec process also runs as part of the\nl3 agent service and should not be touched by the handler.\n\nChange-Id: I86655567810c61dbed0415afd2e7ff343f20c736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3912bf3a5f04d90323ea821a7ae3a04a57b3fe33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/3912bf3a5f04d90323ea821a7ae3a04a57b3fe33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
