)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~932413","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~stable%2F2023.1~I28c928362ef009c1b49673005463b653d038faf9","project":"openstack/openstack-ansible-os_neutron","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-16 09:54:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28c928362ef009c1b49673005463b653d038faf9","subject":"Ensure that services that intended to stay disabled are not started","status":"MERGED","created":"2024-10-15 13:43:34.000000000","updated":"2024-10-16 09:55:55.000000000","submitted":"2024-10-16 09:54:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932413","cherry_pick_of_change":932357,"cherry_pick_of_patch_set":2,"meta_rev_id":"f71b897e991a2ae9f3a51affa77f3e0e0105e905","_number":932413,"virtual_id_number":932413,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-16 09:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-10-16 07:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2024-10-15 17:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-10-16 07:17:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-15 15:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-16 07:17:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d99ee821955cc8fb9460d75dcf79ff81c4977097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-15 13:43:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fb9f1458555ecb326aa73d3f84f0694d17211779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:46:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/204dcba1c3b04144b290bac393a0128c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/826bec5a46c7420baf3f2c0e8e7095b4 : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bfeb5693810345f388a720b362f7bcd1 : SUCCESS in 49m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/021f1b57b2ff4007953427f222dd88b1 : SUCCESS in 1h 29m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3523a4cff79f42ea82bff71118785af4 : SUCCESS in 1h 26m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02d687b407104b278a864e58e6b7a9a2 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e09d5e5376b140d0b0ebf684ada7f867 : SUCCESS in 51m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ecab33e7dc41488ca597b3e12a092e48 : SUCCESS in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1789c15337fe4596a66ffa2f7106ed0f : SUCCESS in 1h 31m 47s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/133a6b54b24547169bfe0e4578837f7d : SUCCESS in 56m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2970cb6d9f124cbca2ec5668db62e8b8 : SUCCESS in 52m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d90b13adc44e40a639a1fbccbf4cfb : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c29951eecb4408db05b93ca70b4c8f1 : SUCCESS in 1h 45m 21s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a0b45e8b72b444493487e5ee96c3b28 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_ovs_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d146d909cfb4ec4b430a1c96fb72b36 : SUCCESS in 1h 44m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/32b74006b3b74c80b0261b58c95030d8 : SUCCESS in 1h 19m 38s\n- openstack-ansible-deploy-aio_lxb_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bdf8a9f0ee94ecb8e1602ee18775bf6 : SUCCESS in 1h 37m 47s (non-voting)\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a42e6335b95a478592b26a604ed4601f : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8d3ca6440d49c8a6d8bc184d16a6f8 : SUCCESS in 1h 41m 11s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cc0a5192971425b80cec5fe5f2f6454 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"63247dcadbbea25b2d3f3f1f4868a2b74aa32bf8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-15 17:35:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78184b48bd22f1e29ecd13abd24d278e4b912307","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-16 07:17:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f48e2bb158ceca8a2d8884ef3baea7e74db8f9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 07:19:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11842757d56df2b717735f9a1c4b38f02768120b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 09:54:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff28dca651648a580b7f01d3267beee\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f978673f08d9467c93e6526f1b56d36c : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/52455fc5a2b6498b9c7d116d0d278c96 : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00bc431d52b0424aa5b65c3b6e9b7e45 : SUCCESS in 1h 32m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96ba2047be6a482a92258de373a09f13 : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d0fca18400b4e15b67efc9b7abbbcd8 : SUCCESS in 1h 32m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb44a0f3818f4b19b25cd564676ecea9 : SUCCESS in 1h 30m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79de8be663df49b8ae435cbb17481721 : SUCCESS in 3m 59s\n- openstack-ansible-deploy-aio_ovs_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b79e85ad4e2418e922b72b2d5e54915 : SUCCESS in 1h 58m 42s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4680b96895545139085b392ab99dd64 : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-aio_metal_ovs-centos-9-stream https://zuul.opendev.org/t/openstack/build/96a57a6f635742bfb91f11bb97883336 : SUCCESS in 1h 33m 10s\n- openstack-ansible-deploy-aio_lxb_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f96f61e768c24d15b82f3b81fa281737 : SUCCESS in 1h 24m 48s\n- openstack-ansible-deploy-aio_lxb_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa99fe7e182247bb9c9ebf0317a4f66c : SUCCESS in 1h 53m 20s\n- openstack-ansible-deploy-aio_lxb_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ef76be2b80486392487876e45780b3 : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ac2ccc7cf8757f98f714af6ad6b661a15b729629","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 09:54:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f71b897e991a2ae9f3a51affa77f3e0e0105e905","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 09:55:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e83c81e1e240479b9ef226c964b06d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4dd7afa08c41a18de2804f8225dee3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eaeb251d949764cf84f2fd9bcc652107dfb3df3b","revisions":{"eaeb251d949764cf84f2fd9bcc652107dfb3df3b":{"kind":"REWORK","_number":1,"created":"2024-10-15 13:43:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/932413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/13/932413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/932413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/932413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/932413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/13/932413/1"}}},"commit":{"parents":[{"commit":"d5cf90ef01ed2d627987d63fc7a32e8b99e9da08","subject":"Do not kill ipsec on L3 cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/d5cf90ef01ed2d627987d63fc7a32e8b99e9da08"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-14 18:04:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-15 13:43:34.000000000","tz":0},"subject":"Ensure that services that intended to stay disabled are not started","message":"Ensure that services that intended to stay disabled are not started\n\nThis mainly affects neutron-rpc-server service, which intended to stay\ndisabled/stopped.\n\nWhile we\u0027ve introduced logic in vars, which is respected by systemd role\nit is not respected by the role handlers, which brings service back up\nrunning unconditionally.\n\nThis actually breaks neutron behaviour after merging of [1], which relies\non neutron-rpc-server being disabled.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/927881\n\nChange-Id: I28c928362ef009c1b49673005463b653d038faf9\n(cherry picked from commit 96fb295bbc5c77c8ea139f523d1cc4e8b506f165)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/eaeb251d949764cf84f2fd9bcc652107dfb3df3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/eaeb251d949764cf84f2fd9bcc652107dfb3df3b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
