)]}'
{"id":"openstack%2Fopenstack-ansible-os_neutron~961340","triplet_id":"openstack%2Fopenstack-ansible-os_neutron~master~I25d517d7590a358ee15ddaa2d89b42f54207b518","project":"openstack/openstack-ansible-os_neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-10-27 12:44:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I25d517d7590a358ee15ddaa2d89b42f54207b518","subject":"Ensure neutron_use_uwsgi is treated a bool","status":"MERGED","created":"2025-09-16 06:35:11.000000000","updated":"2025-10-27 17:30:23.000000000","submitted":"2025-10-27 17:29:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961340","meta_rev_id":"9e99fe54df057b2e585662f2dd93a367a478a0fc","_number":961340,"virtual_id_number":961340,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-10-23 09:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2025-10-22 03:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2025-10-23 10:46:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2025-10-23 10:46:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 08:57:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-22 03:53:18.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2025-10-23 09:57:31.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"8934af49ec07a5195f58c910184331f1cf229141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-16 06:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1995394a4a9553762986176f707c80a560717321","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-16 06:35:59.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"87e7941c051b84fcab17c34b01cc42035d85b96f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:57:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab11cdfee4ed434785f7f1c3d36e4c58\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dd5db0f9836404a93a130b71cb8a984 : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ad79f0cbb0ce46a1ba8b3aa459533119 : FAILURE in 1h 07m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/56ecd6fc10af49dea448aeafa19b8c25 : FAILURE in 1h 13m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c60bc7e465e4e7989a103f5f191cbde : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04921ba0fe154410866f613198475b7b : SUCCESS in 1h 22m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb885ae3a50b4a9aa3da4c69f3d7a015 : SUCCESS in 1h 58m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac5b0f387064e77a248291651b649b8 : SUCCESS in 2h 16m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb6da9eab1be49338ee0d548b7087bce : FAILURE in 56m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b7b1893691407dbb69348143710b59 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa436cd42672428b8a8dc60dd39b9ea8 : SUCCESS in 2h 05m 02s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6822dc5049943a6a92f5ec4962614b4 : FAILURE in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e26409cd5aebde53cdfd008e904b51cf27712da7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-21 14:55:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f1e5869b6ef89974f0833346358eea074f43384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:16:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d47cfc0f732494ea716ab2433c2a061\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1860ddd37e45eba82582c0560c8786 : SUCCESS in 6m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b3a71bff52f4411287d22f7be9c400ac : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e74805b126b04b30be543628618f6827 : SUCCESS in 1h 19m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e53086cd751d4788b281cf3af485daf4 : SUCCESS in 1h 04m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2be5463492894789bf9b70fd89dfa3e6 : SUCCESS in 47m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87248ee45d9e4deeb654eddb664ba84b : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d75da53aed294cea88f8107424c63f01 : SUCCESS in 1h 32m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf2c43f9ae604bec942a60befd29fef6 : SUCCESS in 2h 03m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4a8220f11ab4948bd8ad4825ea5a79c : SUCCESS in 2h 14m 34s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/51a9830ab99e4c2e9087486acf07948b : FAILURE in 6m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfb6b25e825b4706a0697d8b7cfca248 : FAILURE in 54m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc0686ed3a24708b6537808cf333a80 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a90afff37e1f474388d722c3894935f0 : SUCCESS in 2h 06m 35s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fe8e93356f4429d8dce8eebdfc123b0 : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c1a44390c2de8f97430a651928568c35fb5b0467","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-10-22 03:53:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6b1fba3441523a057add77c96d88a249f5d480fb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-10-23 09:57:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1024d63623f6c757bb5e2a9b8a05d5389ce63ce3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-23 10:46:08.000000000","message":"Patch Set 2: Backport-Candidate+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"405def7c56fa86fcb279e4fa5c5598724edaed97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 10:48:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d255c150e7a9a0a12f2cb7e53ef7e8796adc23f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:27:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/866f5b814c3d48c7a19500ead78726c4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004569e2230842b6b37e69d07197b5b0 : SUCCESS in 4m 16s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f9f4b5843fa0400a820506a991093fda : SUCCESS in 48m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5f38c47ce218423e953ff8716522e570 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c0025bb3d8454686b9ef55e61228d6a0 : SUCCESS in 1h 13m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a91f903d1827419b9224715cf667fcf9 : SUCCESS in 45m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c15170c43f24c929366c0b9fbc345bf : FAILURE in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06e243011ca24c79a6f551aea13b1797 : SUCCESS in 1h 34m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e4c0b0213344e2bb64f61c4ae675bb5 : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f5040dd407a408ca9a81f11a2b7d042 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c7f748096d64f0288882551e11b8331 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c6245c0ccad11e968340f37f602cf49f65d44264","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-27 12:44:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7051b256555d7358b927d0aaacea4cd96b01e6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:15:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815fb83ba2f8404bbf28897f35277992\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f491da78f1f048448cbf9760a8030d0b : SUCCESS in 8m 56s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/90757d774eac4c89844e8fab03e454ff : SUCCESS in 1h 05m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f34676ae84b42ca99ec0bd59e4a18ba : SUCCESS in 1h 30m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a778cca01f5458ab0b4c5569f6b33fc : SUCCESS in 1h 06m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ea5fcb6cbf64dd9906e5ec917895663 : SUCCESS in 45m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67e62eaedb4a4ae484b3b9baeba789bb : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e68713f423fd47738e8de505636022fb : SUCCESS in 54m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19fc4d2dda604fe7bc5f67a3ebee877c : SUCCESS in 2h 06m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/060cb2e1241e4be8ad41409c2b69c31e : SUCCESS in 2h 24m 47s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f714f04100946cf9137fe6f4a5c3bd5 : FAILURE in 5m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f32d7c8d1479419496a700d1770e725c : FAILURE in 57m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f2df238b6b42d3add4675d620bb7ed : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e7a182590fd45aaa2cc6624638fc426 : SUCCESS in 1h 34m 52s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16f977e084274827a4c145ebfd175447 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1e168b884bede0c4526adad82cc07148e8c7c04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:17:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d746fca7fcc614f8d52d1df86931fbdc37a52ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:29:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/610371ea72bc432f86249dde96304458\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/376b8a3e136d4fd89fc9271dc2ec7fbb : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/796fac6ce05f4773ac2622d07ac56213 : SUCCESS in 42m 38s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/49aea9c3768449afbd3140eb03dae1bd : SUCCESS in 1h 08m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4946082318f34decba4f826441f33289 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ed7feedf22e34f8ca99520f46c0f6d09 : SUCCESS in 46m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23d7b624dec74c388b380cb2bbc4b1d1 : SUCCESS in 46m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/001e74cd55594207b73e63cf52c732f2 : SUCCESS in 1h 31m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bdf71b3eef54da0a5bc3ecbf344258d : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_lxc_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ca1ec6fb4bc40e48cafb125f8885b79 : SUCCESS in 2h 06m 53s\n- openstack-ansible-deploy-aio_metal_ovs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf4f03ae36f145b28e019210b373b4d2 : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":2},{"id":"67259c28f3b86388b3ee930e83c9f51a24356a66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:29:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e99fe54df057b2e585662f2dd93a367a478a0fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:30:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66c72507c6954bd9b7ae785cd0cfb3e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8717aadeb81c4b5b98fd96e547e3ff11 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b01782fc7e171ec08cd8ce3f4c4d9d03734401a9","revisions":{"b1f9e74f420279e8641f4f06dd6306b31e02637a":{"kind":"REWORK","_number":1,"created":"2025-09-16 06:35:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/961340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/40/961340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/1"}}},"commit":{"parents":[{"commit":"f43c14604e52b4bf1e15303c7f94d9650f73c916","subject":"[doc] Re-arrange OVN Scenario in the list as a default driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/f43c14604e52b4bf1e15303c7f94d9650f73c916"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-16 06:35:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-16 06:35:01.000000000","tz":120},"subject":"Ensure neutron_use_uwsgi is treated a bool","message":"Ensure neutron_use_uwsgi is treated a bool\n\nIn case the value is passed as an extra arg, it may be treated as\na string, leading to an unexpected behavior.  For instnace.\n`\"{{ neutron_use_uwsgi | ternary(\u0027started\u0027, \u0027stopped\u0027) }}\"` will result\nin `started` if playbook is run with `-e neutron_use_uwsgi\u003dfalse`\n\nHowever, if value will be defined in a file, it will be respected.\n\nCloses-Bug: #2096937\nChange-Id: I25d517d7590a358ee15ddaa2d89b42f54207b518\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b1f9e74f420279e8641f4f06dd6306b31e02637a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b1f9e74f420279e8641f4f06dd6306b31e02637a"}]},"branch":"refs/heads/master"},"b01782fc7e171ec08cd8ce3f4c4d9d03734401a9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-21 14:55:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/961340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_neutron","ref":"refs/changes/40/961340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_neutron refs/changes/40/961340/2"}}},"commit":{"parents":[{"commit":"5f482bea9c0e3c4e2dbae93744ae7c9082f3d659","subject":"Remove functional test environments from tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/5f482bea9c0e3c4e2dbae93744ae7c9082f3d659"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-09-16 06:35:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-10-21 14:55:36.000000000","tz":0},"subject":"Ensure neutron_use_uwsgi is treated a bool","message":"Ensure neutron_use_uwsgi is treated a bool\n\nIn case the value is passed as an extra arg, it may be treated as\na string, leading to an unexpected behavior.  For instnace.\n`\"{{ neutron_use_uwsgi | ternary(\u0027started\u0027, \u0027stopped\u0027) }}\"` will result\nin `started` if playbook is run with `-e neutron_use_uwsgi\u003dfalse`\n\nHowever, if value will be defined in a file, it will be respected.\n\nCloses-Bug: #2096937\nChange-Id: I25d517d7590a358ee15ddaa2d89b42f54207b518\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b01782fc7e171ec08cd8ce3f4c4d9d03734401a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_neutron/commit/b01782fc7e171ec08cd8ce3f4c4d9d03734401a9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
